)]}'
{"id":"openstack%2Fneutron~896804","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ide3204748274cbab9731310c2e2bd8bd8c9a53ff","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2024160","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-29 14:04:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide3204748274cbab9731310c2e2bd8bd8c9a53ff","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","status":"MERGED","created":"2023-09-28 14:37:59.000000000","updated":"2023-09-29 14:05:41.000000000","submitted":"2023-09-29 14:04:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896804-bug/2024160","meta_rev_id":"1beb5957de5da1547cfcb9b181060980f1b071c4","_number":896804,"virtual_id_number":896804,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 14:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-09-29 11:34:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-29 09:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-09-29 11:34:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 14:37:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2023-09-28 17:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 09:52:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-29 11:34:22.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"241d9ccad47fe5ca7172c2c132adec00a995c205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-28 14:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef482ad8b51ee1b99576d95ae9308187feeabffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:00:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe98f7aea877404a96363db258e44cd6\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/e1f0a48f27b34b1c9555ba64e694f867 : FAILURE in 1h 21m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a343d78b1e3b499fb4ee5e9ae456dc8b : SUCCESS in 36m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd827e9eb0b44525a2820787626ad657 : SUCCESS in 14m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3b7dfa129e24bc4a97dc26c1768bf10 : SUCCESS in 27m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c205d586e9354b6eb2967683dffc2d99 : SUCCESS in 29m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a5a2fed14a4b0cbe5223f0da6651be : SUCCESS in 11m 05s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1a7f2417db244c37b7c8940c5eb8506d : SUCCESS in 22m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2850bb5600c49f1a53a823d4aa06cc2 : SUCCESS in 52m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/477222edda9e495f9d9eb65ca463386e : SUCCESS in 1h 22m 38s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c30a12b602b041ce839fe9d2778877ec : SUCCESS in 1h 16m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d8037271ef29442e94c3f7802a5f9839 : SUCCESS in 2h 12m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a8153c98fec942f3a0d36c526069deea : SUCCESS in 1h 11m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ee1cbeb07183434290c5435e985114d5 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b3701f53019e10c5dcf0105e612df7ccad4b14b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffee7e78f7984df1ac2f63450e055439\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe2be76d4b19436f9982b9572acfa139 : SUCCESS in 42m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2cc2a3e147254e8497e497cac7739125 : SUCCESS in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"751823bc39ee178fd965e98561d4aed476e4531f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 09:01:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80c36c24fbf6a86654302db136d891048ec215a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 09:41:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58783267cea04480af01f5cc6ee22db0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/498ee8a8c2444f0591e855aee738f71d : SUCCESS in 37m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f15f3ed780c4d688df45c30e231dbde : SUCCESS in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13635d9ff223ed1068c14f5fb6c74d47492ac13c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-29 09:52:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8382be444641f47d88e63274935c5658c86553e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92572466525486a8a14dee858401659\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/64493a4748b146a49fa4cb04cb844413 : SUCCESS in 1h 30m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d1caa0742db4b1c8298a4097491f1e7 : SUCCESS in 33m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c432fb5d062042fd93de5bb77d04dc7d : SUCCESS in 14m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d96172965404bd6a3ab71563c02c63e : SUCCESS in 21m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9d3290354724443bd35a60539adaaae : SUCCESS in 16m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a97d9507a814dea9dec2b1d0cc161dd : SUCCESS in 11m 42s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/872578790e0b472dbf79998bca6b6d01 : SUCCESS in 20m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2c45521554e4d879889618d64c953cd : SUCCESS in 53m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/93d0ecb9892a4c3991ea81052a13cb9c : SUCCESS in 1h 11m 07s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/224a37e5916a48a49c3c1f8eee97b11c : SUCCESS in 54m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c50aeebca203421aa6d68642abfe2527 : SUCCESS in 1h 31m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77ab3b42693c4b56923ed8a9ffe8a0d4 : SUCCESS in 1h 00m 42s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2cc5bd23c33e4893a8aeb0837ff4328e : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":1},{"id":"911be511f2dbbbd447e547471111ff6d7e4890c4","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-29 11:34:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd1af86696c082b93c169c11480fcd909df0950b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:34:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d8a0e3bf7566b45737a40bdcf575813f8fe4849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:04:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b3a6354a39498d98e7cd6d3e304ca6\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/b77b532e19b0430fa448e8c406aa76ae : SUCCESS in 2h 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca689c1b5cc4bd0a9215007615684c5 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4940d7a5ad5744b58a0e1c947b418b2a : SUCCESS in 20m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/173ca082b05744339b2d9e132b2937e5 : SUCCESS in 20m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c75c856c368456b88605edeb935063e : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/091a3fe8552d4960bb96fe60808a22b4 : SUCCESS in 56m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ab94e6e4eadb48f69f3d3fd1d00a69b9 : SUCCESS in 1h 03m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0caf38e4fb6460fbb0fa0fa02590ac4 : SUCCESS in 1h 25m 39s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/176f5942d1934ddc8207b9dba8252295 : SUCCESS in 1h 11m 58s","accounts_in_message":[],"_revision_number":1},{"id":"2c0beca254c20b3f6a4b71e9ca80c84eae7fa6ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:04:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1beb5957de5da1547cfcb9b181060980f1b071c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:05:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e83241f3fcf48d48339bb7540897d87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11ba86398564ee892b3083e7f41f5c4 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2319e7c35f79798cba199f52a5a7c4a4ef7bc088","revisions":{"2319e7c35f79798cba199f52a5a7c4a4ef7bc088":{"kind":"REWORK","_number":1,"created":"2023-09-28 14:37:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/04/896804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896804/1"}}},"commit":{"parents":[{"commit":"4de1ddf4e3441f9121bd3cc0dcb249b1bcc60512","subject":"Merge \"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4de1ddf4e3441f9121bd3cc0dcb249b1bcc60512"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-09-26 20:02:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 14:37:51.000000000","tz":0},"subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","message":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent\n\nWith newer versions of OVN, port bindings are removed and added\nfor both virtual ports and container ports (aka trunk ports in\nNeutron world)[1][2].\n\nWith this, Neutron recently landed a change in order to handle\nvirtual ports for the update/delete steps[3], however, this has\nresulted in unexpected behaviour where the binding information\nis wiped for trunk ports, then subsequently removed on the\nnext maintenance run.\n\nWe should make sure that we only have row deletes for\nvirtual ports specifically rather than all ports.\n\n[1]: https://github.com/ovn-org/ovn/commit/ec933537f9b04b35ef4b79fb2b4743f9095da209\n[2]: https://github.com/ovn-org/ovn/commit/28d40c0a024227c9c845117087b7594723ff0ec7\n[3]: https://review.opendev.org/c/openstack/neutron/+/883681\n\nRelated-Bug: #2024160\nChange-Id: Ide3204748274cbab9731310c2e2bd8bd8c9a53ff\n(cherry picked from commit 1160aaa4716904fdba06a8dcfbc1d5a2ce419e72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2319e7c35f79798cba199f52a5a7c4a4ef7bc088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2319e7c35f79798cba199f52a5a7c4a4ef7bc088"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
