)]}'
{"id":"openstack%2Fneutron~896883","triplet_id":"openstack%2Fneutron~master~I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc","project":"openstack/neutron","branch":"master","topic":"bug/2038413","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-15 07:46:55.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-17 16:21:19.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"25468":{"account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"last_update":"2024-06-10 12:13:52.000000000","reason":"\u003cGERRIT_ACCOUNT_25468\u003e replied on the change","reason_account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}}},"hashtags":[],"change_id":"I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc","subject":"[OVN] Update lsp host id when virtual parent moves","status":"MERGED","created":"2023-10-04 08:12:30.000000000","updated":"2024-06-10 12:13:52.000000000","submitted":"2024-01-17 18:08:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896883-bug/2038413","meta_rev_id":"c0464bbffec3ca95c89e247e610a1c3a4101f077","_number":896883,"virtual_id_number":896883,"owner":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-12 09:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 18:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-15 07:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 16:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-17 16:21:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"CC":[{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 08:35:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-04 08:43:54.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 09:04:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-04 10:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-03 16:52:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-01-17 16:21:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-06-07 15:32:36.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"CC"}],"messages":[{"id":"ddf16bc37dcf19ed30abed5e3bf02fd650fd2377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 08:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65883325a9d9217968244bce3130b2fd5af3abf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 08:13:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7ba6c95886432d53314390c3e41dd438b81d561c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 08:35:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f1113d4fdc8eb68c6e5b054729815dda8d6931ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 08:43:54.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b8ae4b6fec211b1e3b6bb0b37f3f305623307f87","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 09:00:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54c455bd192699a52a3b617e062e71d81c202795","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 09:04:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe5e33e3e4754ff39d8c88f105bdf925\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cc41f62011e4eeba16de3c5c0fe1b38 : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77e793245bc043cfb26b0f86becbb041 : SUCCESS in 48m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd67b1bce55843bda66a25ceb46a0e78 : SUCCESS in 48m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2fafe0b67304578ad4b045941a000da : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d454dffe51c1eea14bc192b770c777977a0ee050","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 09:14:43.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d8b5471f3c37859413b9b012cb1154713ded387f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 10:28:12.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b0ea15d16764068bdee8cd26551c82d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3b7c3c39aad4453f8d5bc607b1f6e66f : SUCCESS in 1h 30m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c368a56ff714f96b08a1e8952abdf11 : SUCCESS in 42m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abfd28a1b34646c9ae87cb1c396f0250 : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fefea5291e334da59d1d6fbb4178ce59 : SUCCESS in 30m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c929c1d3921d4a0fae323c35360666a5 : SUCCESS in 27m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5c83e97710400c879b7514ea119d2c : SUCCESS in 24m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5e3a0c3c79f4a27bc9b66992639baf2 : SUCCESS in 24m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/507e0c1044fc471c9db82414a9cb086f : SUCCESS in 10m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1628fc4fc7874bd9bd9e84d739e52407 : SUCCESS in 28m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d6f12edeeb9048598200c846e3e0f741 : FAILURE in 59m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2837a4e423994b70b737463c1ddbfd18 : SUCCESS in 57m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccc48de40c3143cc98380642ff7cc9bc : SUCCESS in 2h 08m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/359fefb79e26455394b8a8810f6efae5 : SUCCESS in 1h 17m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2555d86b1f17af752927abfbcc809c8532a44b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 14:03:33.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7783cbe147c44070a7a0b725f0723dddb380e2eb","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 14:03:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dfdd27009922554ffaf46226a5535b55ca40c263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:16:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df3f927a2a80415e8865d32219ba8dc8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f89b84eb612401a9d2d3391bc52fd67 : SUCCESS in 2h 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13ffc280ac8e490e99a3586671ac1c56 : SUCCESS in 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d2fcc9eeecd4a1ea60d0e4a3e33bf75 : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ec10f0c5fc34a498c4eaa8a8395d1e0 : SUCCESS in 30m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb7f5eeb8ce48eb9502c1f7e828049f : SUCCESS in 31m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3112485a670b40e6a787bdb84d9ae45b : SUCCESS in 25m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0904df751b9e4298a22c4aa35801a9bb : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499d9fac34a34c39a84043033c261689 : SUCCESS in 14m 36s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4bc5ca3d0acd471882d6f492ffa9d32a : SUCCESS in 24m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4961ca2308994d0581af5bfd2436ce07 : FAILURE in 54m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/41af9095fa9f4d07b538cb9595aa9168 : SUCCESS in 53m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/389516e241134d178dbe9584b5a27639 : SUCCESS in 2h 05m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/663ea1359c334033a7f98ea96a40e912 : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":3},{"id":"4de3e309e416a6ed1f400f50715d87829851c75f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 17:19:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b3ffd1dded4445953d654b3f4fd2c4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad855f2614964309b811c4960205064b : SUCCESS in 1h 11m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bb2a56a140a4051b6b710c4b720fd8f : SUCCESS in 1h 11m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/649dd2ad524f4281bc34ac509d0cb565 : SUCCESS in 1h 10m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2798c776ae74486a1bf85c1eea67563 : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e4a510ec4caf05d698bfae3e1a609b65186b479f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 17:34:50.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e62f0c3300ce508451d761aa7654630076267a93","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-04 18:00:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"79c7402b8ee0650142e8d5b6eca4257287cfe389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 20:36:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832a86eb1a0f48a8975a7f0706329df0\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/865d3fc747b44a83ab6f8741638e302e : SUCCESS in 1h 30m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e0e4bba8be14efc95eb3ae79591234b : SUCCESS in 29m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9bd5ca5fb34b93aa6a5d436d43d890 : SUCCESS in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e84822b52cc45cab58f7bbe7d85544f : SUCCESS in 26m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa465172ee9e439bbbf4db50e7a4c42d : SUCCESS in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b4eb1d3733f4833aab4aefd6a5dec54 : SUCCESS in 20m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6dfa3c5ec87a4e1ba4d739b19b59ac4e : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558e2efcacbb4dc4bacc64496ea2913e : SUCCESS in 12m 19s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/079ab03d6abe4d229f42f9ca009b7d0e : SUCCESS in 24m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d984c7e07244be08cb60caef4437583 : SUCCESS in 54m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/830c3fbd71244e32a6c372401c5f4f61 : SUCCESS in 55m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9a995bc809134779aa6084dcec7c679f : SUCCESS in 1h 19m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c85c153b3634c818845ad5c510748fb : SUCCESS in 1h 20m 38s","accounts_in_message":[],"_revision_number":3},{"id":"057d41e0e01f1c6a138ad18d968df1ea2acfb3e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 23:25:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa6848fb396d4d0ba1e75b4e3caa64e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e0d25d76c3547f0ac36de87ca4fd498 : SUCCESS in 39m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c96809f49f814983b545500762760dbe : SUCCESS in 40m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb0058d978af4562adff61484ebd412a : SUCCESS in 39m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57e610475f7344d78c8a7f21f4726b52 : SUCCESS in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c6df4d469aa0c3d6d6686d0422af7d4e18bcd631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-05 12:05:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"27fbe343fa7a01d16bd1a89655055f5d558bf171","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-05 12:06:33.000000000","message":"Topic bug/2038431 removed","accounts_in_message":[],"_revision_number":4},{"id":"eb1559f86ab25dd4d330764626477dc7c42c23c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-05 12:06:42.000000000","message":"Topic set to bug/2038413","accounts_in_message":[],"_revision_number":4},{"id":"0e4942c79b2d0084f87a6cb120fc9eb40a051d07","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 13:29:25.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f2e39b171ee511840e6ed0ff534a06728f1af0ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 14:20:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a83d845f0504c62aaf7aeb2cf47214e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aeb04e2ee5844ba288bde9e93771a73d : SUCCESS in 1h 49m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3713abb28a446558f55d602e9f89a00 : SUCCESS in 40m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d8ce35cb19e4c55aa8ede78dcd43116 : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc0f35ee89cd4caea5b0707c5bcb70d3 : SUCCESS in 22m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28b6b82123514edc8437eae88acf0283 : SUCCESS in 28m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a8fb26c588f4d2f9aeb47f4557f8063 : SUCCESS in 30m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ea098e9c0854323bb1b4f08ce792a3a : SUCCESS in 26m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59276df2e9dc49ef80b6366b3bd0d3db : SUCCESS in 9m 38s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/af207c7e91b84712ab4b7c9e82acbeff : SUCCESS in 28m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09cacb638dde45e1b578702d6b8ed4b1 : TIMED_OUT in 2h 13m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5856b52fd5d74daf9611bfd8a9d9c2bc : SUCCESS in 1h 01m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bc9328efb3c54dab80a28dbc1777800f : SUCCESS in 1h 39m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4937ae6d50c742858e2f2123c3ec7456 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":4},{"id":"7e18480b24d2312ac55f0d823b54ce5f0648309f","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-05 14:51:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5521b69fcb5da71481ed3b3ad0d986e062df15e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 15:00:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f84fa790eae44c08f64f8dea61eb2e1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a129ec610344d15bd6466f24887189a : SUCCESS in 41m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68a1357020d24956bae79083fb5352b7 : SUCCESS in 42m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75d2e20cb3e24f85b3c27b4a63ac6c96 : SUCCESS in 42m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b36ee4c4493f4f4eb74ded7bf2f80731 : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08f8d494ac9f4908f8d9933b7f9063b7c1ffe85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:46:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd0d557a4b7448daa47eedb480f631db\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e778d3197b234715b8af7f0177df0757 : SUCCESS in 1h 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/680870a0eddd4ba79e983bfe704ec8d0 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d838c0ac3c654f278665b5d44dc21857 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be0030217a994ba8a42c7578e11e3d96 : SUCCESS in 25m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4df3fc068e3469191a8e255736bd723 : SUCCESS in 23m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16d8bd9a3b4044918b47c1d7e48da902 : SUCCESS in 25m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e1bee47823e41ac9f5717a29e9a49d9 : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409bfc355f1e444d983974165611ce91 : SUCCESS in 12m 22s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bc310da2c5d348a181c46f477a53de70 : SUCCESS in 26m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8efad615617c48f3bcefa2059228602a : FAILURE in 1h 02m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/26c8b2f1e2364379a9477ae6ae01953f : SUCCESS in 1h 02m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fbca032e24f43958991a7a392aafea8 : SUCCESS in 1h 48m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90763db3d6c94ad6a16ba36d636f0de0 : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":4},{"id":"3a1989e1605196c1e1a242da6a1f65e9f4366549","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 13:45:31.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6fb9c9a9179546eadad2fddb2a5cf3948b011a9e","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-09 14:54:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8dac1244dd843b502368e7f2ee732cb3d70fdca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 13:47:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"b08673e40b57997603d57b1073028356c414c50b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 13:50:38.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a5b7bf477f6fd8b8abbc1d85ab7a0044431cf02c","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 13:55:15.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"fadbf8959d3f2974311c8476ca77b240a1543c18","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-24 14:18:33.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"6a6025696932083406970d83e405ba33f8e61e8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 14:45:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78410dc4bee42a4a130d4bc8974fa0c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/143d1866c6d94650a039c2b32fe94b68 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7030e23542f4f948ddd09c8e7289e40 : SUCCESS in 52m 31s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6f9c362714e4c59b0849f85aa1e11df : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10f078d5714d416ba34c2be159051ae2 : SUCCESS in 45m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9f8823d5272b17927af9d2ad815fe6653716f2e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-24 14:48:17.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"72ff70809b9862de332073ccf7f78582d66054ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 15:37:18.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed15832d8030437ca089789233e56f0a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25a37d9265594bd6a98ba2e7b06492b5 : SUCCESS in 1h 30m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb4c8bc4e4e34a21aa2e54cc75a6e0b5 : SUCCESS in 33m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fe1f7e431e43f28180cc5cc4d7acc5 : FAILURE in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08623b5f08af4867b1c2f208ffdd1a75 : SUCCESS in 25m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61c0b5ebb5c240eba798ba6de941ed48 : SUCCESS in 23m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fd5c40bbe194deca3862158966aee87 : SUCCESS in 28m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f89cc86b204647cbb85b08ebe2faa718 : SUCCESS in 23m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c30faff6334481b000aa2ab7c4111a : SUCCESS in 11m 46s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d4906cd8d82c492fabffe35cd3e30513 : SUCCESS in 27m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d6543cf0471b420c9afdce6955386702 : SUCCESS in 58m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42739b9d91c94eee93329684d386cc06 : SUCCESS in 1h 01m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/00e919c94885428090b7ec066dd077af : SUCCESS in 1h 06m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4129aec9a8a4dc29757d26393ad27a8 : SUCCESS in 1h 44m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fbbeaf51a99c4a79b5935121a53c9741 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":6},{"id":"3676931001cfbf76d7fdb6f814495a5dbb85cc6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 15:49:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"30da32c1ecd939aca92b27a27c433b076f345d92","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-24 16:19:51.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/83/896883/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"e9cb3c33862740c2ba7723cb8f2dba2c71e094e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 17:38:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49c7641f8fef449c9be9549ae4406910\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cddf5fa9632e450f8541bc6634f1712a : SUCCESS in 1h 47m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea379d97c9c94242a60728ca8beaa373 : SUCCESS in 33m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa831cb555c4202a89097cf8633b2d6 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1178c39678734ad0891283048be29af5 : SUCCESS in 24m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c08fd23c2e104d16b43322a48502176e : SUCCESS in 28m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b382a6fec99a44249f77599c3e429777 : SUCCESS in 29m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a7908fad73a43ae902a6d7eac4061d7 : SUCCESS in 21m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed344191d9849c584ce545e45e98596 : SUCCESS in 12m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3f35a01127d24e14bb67fcfb3486887e : SUCCESS in 26m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9cbabe8068e43d08502f57e58605dfc : SUCCESS in 1h 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/68d781a7d9f74fd593fe299e95ebbdb3 : SUCCESS in 1h 08m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9983a8369a524d9392bcf1e51c8fe8b2 : SUCCESS in 1h 08m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17cc6877fe324ac1afffbc9a039b1ddd : SUCCESS in 1h 38m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/02d532e82d0742b2b76433c41e5a580a : SUCCESS in 55m 07s","accounts_in_message":[],"_revision_number":7},{"id":"9a703421f77af59f4c32a18909ab100777437743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 17:45:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cdb283c40604e7daa54917d86c308a6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/066a7d74eaa446608a40d0cf8d7ecfd0 : SUCCESS in 1h 07m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a3e913ce9b3434b8f94be52da9ab744 : SUCCESS in 1h 04m 21s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f35875392524fb89d8f919f15ec6cad : SUCCESS in 1h 05m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50ddb31702df450caa1b29ac444bdebb : SUCCESS in 1h 01m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"33fa38d40c5eacbcc2efffbddac8b04ad8c0b381","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-25 15:12:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c10a4ea87cf23d3165402f184da3ee02f112ac42","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-03 16:52:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0749a457f27c297eaa664043c9bfa37bec9d5adf","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-06 09:31:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8f4c07c7f8750fbfab54953ca0a14c34e1bbb127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 15:37:23.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"e21fcbd5eb6208657c4197b0c6e0fc7428e2071d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 15:46:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"59eb1f061ec0c6e1094481b07851802bc0296cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 15:50:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"71601edfeb69296b1fc9ecd62347160628694037","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 15:56:37.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ff558358d01fc5e3995583c85313fb00a2b5a749","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 16:08:37.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/83/896883/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"18aef43bc67fc5eca75a88b00c139d6110faa4b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:41:57.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c9203fb0b341df8360b56aa34d9c2d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2672aed3540142ad92f6f6f5da713ea9 : SUCCESS in 46m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6729cd7afbb248c5ad2e6847328a0b2a : SUCCESS in 48m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0cf773b863f45fb824b2060ee67586f : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"363c4a306b9be09e58a18215c3b633bb76a6d5a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 17:11:40.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/83/896883/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"ad44925e64cd52f6f0dfed6036cd8cd7f01ac542","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 17:43:04.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/83/896883/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"85f44c52308e36b682737e3d7b91f02e8e5059a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 17:50:27.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75ceb1db4fb348f6b5a042af20ec74c4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49bb092d723c4e05ac73b790d7eec480 : SUCCESS in 1h 57m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b48e35f1e0da46b7825e9a5e4c0be2bf : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/402cde1b95334213a1fd2e2edc2f9cf5 : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f6e6e433934ad4bdc0024f31cebd89 : SUCCESS in 25m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fc55f3d9c174e5f8ccdfffe2d3f36d6 : SUCCESS in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286e404cf39c434faacd5ad45fa36e41 : SUCCESS in 11m 27s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/83318257e3f74f658c1bc575695009a1 : SUCCESS in 31m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3d6cec4c676b41049f4ba2264e680e53 : SUCCESS in 1h 06m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a70ffa305a242dca2308aec2049d9c3 : SUCCESS in 47m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d60af21039de4d3ca347b3aa364c7568 : SUCCESS in 1h 04m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5c967b473fa4520a202ee3b44643883 : SUCCESS in 1h 58m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2fca81a5346c4eed90aff328bdb65673 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":10},{"id":"9407b635daf85d2f5e314d63a1b70fd173606213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 21:35:10.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"cb8e1594c070b40f5ef8763605eb6c3f89e48666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-11-13 22:03:46.000000000","message":"Uploaded patch set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"fd98103c1cc51624ac6784c1ae24c34618153de9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 22:06:42.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/83/896883/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"3f9a37ca907aa70bbe7626a9237215dfbb8a2ff5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 22:38:11.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/83/896883/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"6322f11441a810c56db91e566a334ba8f496b8a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 22:47:56.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51650390689144a6831118bf2244db47\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc101b8335b14af98c326aad77e6c21f : SUCCESS in 42m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dba6ad58b9454dada302d41ca2408c47 : SUCCESS in 35m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/08031652c8e8435d841e07c030da7ff1 : SUCCESS in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8404bbc943313b371b94d447cc99706c971c74aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 23:47:41.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc739835296d4ea786f0d74c58114174\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c1c0d144001047d79e9c5e5ecdc6a248 : SUCCESS in 1h 42m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c875958600940999da3f65cc5bc08b4 : SUCCESS in 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31618a7b236421183cd3e2b872e2b2d : SUCCESS in 13m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1effa1dbc1b4db3a0468e1a4ce60771 : SUCCESS in 25m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f2cd872a47f4ec2876dd1d8d7d941d0 : SUCCESS in 24m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b043901801433094e970fcd681742c : SUCCESS in 11m 47s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d80943af558f49a39d6f9d96544c8db2 : SUCCESS in 23m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/82a6a0793085406b9de247d97c5e0e50 : SUCCESS in 1h 04m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d769bfd5f684448bcaecf9a7f5220a5 : SUCCESS in 1h 00m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/575537ce203d4802af6d1cae2a5395bf : SUCCESS in 43m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d5d80ee8c4b472f9b43d1afb7a196a2 : SUCCESS in 1h 22m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a93510c25084c22a781babfe1a54e75 : SUCCESS in 56m 03s","accounts_in_message":[],"_revision_number":12},{"id":"2ee98b0b41ce48c27dddf9f2c636ab1998a97505","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-07 08:22:48.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"697ae2e216338085da4b4f8461a400a0680550bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-12-07 14:07:16.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"7eed915f7820b44594d81fe9eb4771dd9a782b78","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-12-07 14:12:29.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"800258768bb78dd10d1ef501d241167f73c8ebac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-07 14:43:10.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/83/896883/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"e71cccdd11cd6d7b6b3e1fa774453fbea2316473","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 14:51:21.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b8013a48ea48dda58376d0860fda5d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9239e548e6694c6992d2bfb4a3d0b9bb : SUCCESS in 18m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47405c430344404581aae84192c763ec : SUCCESS in 41m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/32e2b4bf68684fc09cfe32722c1d4660 : SUCCESS in 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0342c664f28e58efd1c4f3a3321b3089855f49f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 16:51:36.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1796af0dc15949789e5cc5c904bfb646\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e50afdc8eef94ed39e8fea4aaf567463 : SUCCESS in 2h 30m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53140dea0b7f4564af9e4ee199fde3dd : SUCCESS in 41m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e4257aa9774f4d8135447e21bb92e7 : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/004dbec573ef41449085c99985840f9e : SUCCESS in 34m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/561ed297738b426a8a66efa5146bc74a : SUCCESS in 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8920ef1afc3414fabfa000e2b071f11 : SUCCESS in 12m 54s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ad66dfcf2b594e268044dc35a80d6ed6 : SUCCESS in 31m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/33803f2ce54348f8a2a822fbd3f83b99 : SUCCESS in 1h 14m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eba68c75940f4131a2aecea67584053f : SUCCESS in 57m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ff455fe50fd4c85b83943e31ce756c2 : SUCCESS in 1h 07m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ed2e0505b63421fb9752aa229fe407f : SUCCESS in 2h 19m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86814da6b449428688652c4e75613fdd : SUCCESS in 1h 24m 11s","accounts_in_message":[],"_revision_number":13},{"id":"e998bff9727af20e13383afd946ee33e01c0c17c","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-09 14:43:43.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5c76cd87004d824861215ed115aa5e20bf38f42f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-09 15:50:37.000000000","message":"Patch Set 13: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ea4e75c85322e06847ee7887da23bc544228f71e","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-09 15:56:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c0e54a47332a06cf9b7e89cb1da8a8760af55706","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-09 16:04:54.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e3d6450c09fd28ad3bfc95dfb9fc6e71fb273c70","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-09 16:14:57.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"0b8034603f1c3207e6feebbdce076ae0ebfa7516","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-11 11:11:44.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"2dfc6d9e9211be08549aea1c96838a1b0ab580f4","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-11 11:23:01.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e6d48c4e14f99d56abcad2ccf87a568d9f0d32be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 11:36:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c3012c4237c41b293d9a9faaf547295\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6583ebedd3a48c69fe54c569d2275f3 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ce2b94a8e0e46d79bd715fb9bd5b663 : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dadc84e37b4e4553affe63c6b4d56e44 : SUCCESS in 23m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"88603d0e51022016d6f3b7308d0487027a9239e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-11 11:42:55.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/83/896883/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"da180fa109047c664d71aec764ffe7cd8c835e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 13:08:36.000000000","message":"Patch Set 14: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b70a462795974927a2943cb582bd796d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/365d2c59d510472292c1f59467613dd7 : SUCCESS in 1h 54m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7e0a79d31d04ee8b8a53d8339b02031 : SUCCESS in 40m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70747222a6d4c828a4b217eb2a4e016 : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8430dd7582b541da8c70c3678c75685c : SUCCESS in 23m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/119e37e6fcae465ab209f88bf54de89f : SUCCESS in 23m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b836be8c87f448b9cdd27a981f5e797 : SUCCESS in 12m 21s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fc936baa2d274f5a8e0b14da2a684450 : SUCCESS in 26m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/187ca315024f4a7ea02ff18061566217 : SUCCESS in 1h 04m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/660d807ab9e74056bca83a8fc8311186 : SUCCESS in 53m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/20b2a91f481545ac8b6237fb26232c08 : SUCCESS in 1h 01m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d075fb6519dc4d59a0f1566af4cf35cf : SUCCESS in 1h 20m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ba9483fed934a03951e9d03952bce8d : SUCCESS in 1h 20m 29s","accounts_in_message":[],"_revision_number":14},{"id":"0c490528d9a8b2a50e84ac91ecc737b481001ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-01-12 08:50:46.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"ebb6d5668edeab8600ed91a907e70d2a707514ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:18:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2af3195ec718465682d3e51fc76c49a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccefb341c112444aacbfd21316465e37 : SUCCESS in 26m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92db7b8358c74b8e8ba43c00e7dcce4e : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e163d5cda0b44d8da06e6f04057de9d6 : SUCCESS in 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"acbe0a041e4735bd7ee842dfad2146bcde5019b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-12 09:41:42.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/83/896883/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"c8b03d6f4855c4c8ee65d2b5e9f4968e7d8c898b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 11:22:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05acfaa109f0487292ede11bb6d58e28\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b4a615f6e1ff4dc398220565f698f932 : SUCCESS in 2h 01m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b89e196bfa844903a19e880a906261a3 : SUCCESS in 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eae1500866764529bbdf09007ee2ce6b : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78bef4bdfdea4015800bb8054125e662 : SUCCESS in 29m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2152cb8bfed342f89b3b04d686fc36b0 : SUCCESS in 25m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beafe7256bb34e448a3003d3ee24a908 : SUCCESS in 11m 46s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/55d9fcb183d249719d1e4348a1074c5f : SUCCESS in 33m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f15bd382afb54cb3ba923d8d7501abef : SUCCESS in 55m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6eb20af51a7142839420a8d2a5d4667a : SUCCESS in 1h 03m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/770c5bceadc247a8b7c11d7c7e5f7de5 : SUCCESS in 56m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f00910278a24680a3bb53c4ccc6d0fe : SUCCESS in 2h 10m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5ad456a46d6449f4b5cfb1ef64bcad76 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":15},{"id":"987060026b9f5f794933f57fddb9294755dfa4a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-15 07:46:55.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"395aef69fef8b16892c9c9f7d5e889af374fd2b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-17 16:21:19.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"236e015acee7d31a75c22c10f6c22dd7170689ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:21:51.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"534bce67bdf3e8a2e8adfc28d402467de0c70c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:08:16.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d1edeb71d24097a60626826aa6ecf6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f32197e669d44a84b0d1091f54a08510 : SUCCESS in 1h 45m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dff4a38de1224dc99e6061c5444c24e7 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e60558939b68483d99dedbb2cfab9db9 : SUCCESS in 27m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4d98920001c4297997cd8c1317d5373 : SUCCESS in 25m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e284bc4d8a0489b99c79241167e3089 : SUCCESS in 13m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea8bc69818b24d8788826d1f2bacdd34 : SUCCESS in 57m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd73696af2eb4028b4eeff246fd0fca3 : SUCCESS in 1h 01m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2dcf79b706a04c629c67434971a4cd97 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":15},{"id":"f9e6266243cee24a9244b2318b8100a929bf532a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:08:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"d87f11f496c20a73829f35f73c26fe1044e7cca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:09:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/febd87c4ff73427ab01da9140da93698\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb6be4b23be449daf796c8dd7d690e2 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":15},{"id":"e34e35eaa42dd6964facfbb43bb88e2ec555c0c4","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-07 15:32:36.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c4f4f099fbe76fbf3a5c79bfe9abd6e535da3b45","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-06-10 09:42:11.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bb8bee9a62e3a0963a42c5adaf88710e64451213","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-10 10:02:05.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c0464bbffec3ca95c89e247e610a1c3a4101f077","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-06-10 12:13:52.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"e68a920c114010a36a667d90866bfb6243148b6d","revisions":{"00cccff9ceea1158e4f96bfeed2dc3210de5b9f2":{"kind":"REWORK","_number":1,"created":"2023-10-04 08:12:30.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/1"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 08:11:54.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent changes","message":"[OVN] Update lsp host id when virtual parent changes\n\nWhen a virtual port is moved from one port to another port this event\nwill update the binding host id in the neutron database.\nIt then relied on the PortBindingChassisUpdateEvent to update the\nlsp port with the new information.\n\nHowever, the PortBindingChassisUpdateEvent is triggered simultaniously\nwith the PortBindingUpdateVirtualPortsEvent, causing the host information\nwithin OVN to stay one step behind, rendering the information within the\nOVN DB inconsistent.\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00cccff9ceea1158e4f96bfeed2dc3210de5b9f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/00cccff9ceea1158e4f96bfeed2dc3210de5b9f2"}]},"branch":"refs/heads/master"},"b09752c46a61acba31ba874e5f33b50bc312f818":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-04 08:13:37.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/2"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 08:13:23.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent changes","message":"[OVN] Update lsp host id when virtual parent changes\n\nWhen a virtual port is moved from one port to another port this event\nwill update the binding host id in the neutron database.\nIt then relied on the PortBindingChassisUpdateEvent to update the\nlsp port with the new information.\n\nHowever, the PortBindingChassisUpdateEvent is triggered simultaniously\nwith the PortBindingUpdateVirtualPortsEvent, causing the host information\nwithin OVN to stay one step behind, rendering the information within the\nOVN DB inconsistent.\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b09752c46a61acba31ba874e5f33b50bc312f818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b09752c46a61acba31ba874e5f33b50bc312f818"}]},"branch":"refs/heads/master"},"bcc6553329e6ab376cde4a8707871da7df726099":{"kind":"REWORK","_number":3,"created":"2023-10-04 14:03:33.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/3"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"michel@nederlof.info","date":"2023-10-04 14:02:20.000000000","tz":0},"subject":"[OVN] Update lsp host id when virtual parent changes","message":"[OVN] Update lsp host id when virtual parent changes\n\nWhen a virtual port is moved from one port to another port this event\nwill update the binding host id in the neutron database.\nIt then relied on the PortBindingChassisUpdateEvent to update the\nlsp port with the new information.\n\nHowever, the PortBindingChassisUpdateEvent is triggered simultaniously\nwith the PortBindingUpdateVirtualPortsEvent, causing the host information\nwithin OVN to stay one step behind, rendering the information within the\nOVN DB inconsistent.\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcc6553329e6ab376cde4a8707871da7df726099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bcc6553329e6ab376cde4a8707871da7df726099"}]},"branch":"refs/heads/master"},"61a2fae7a1164c5fac2fde883b4c4ea59f965e44":{"kind":"REWORK","_number":4,"created":"2023-10-05 12:05:09.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/4"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-05 12:04:47.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent changes","message":"[OVN] Update lsp host id when virtual parent changes\n\nWhen a virtual port is moved from one port to another port this event\nwill update the binding host id in the neutron database.\nIt then relied on the PortBindingChassisUpdateEvent to update the\nlsp port with the new information.\n\nHowever, the PortBindingChassisUpdateEvent is triggered simultaniously\nwith the PortBindingUpdateVirtualPortsEvent, causing the host information\nwithin OVN to stay one step behind, rendering the information within the\nOVN DB inconsistent.\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a2fae7a1164c5fac2fde883b4c4ea59f965e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61a2fae7a1164c5fac2fde883b4c4ea59f965e44"}]},"branch":"refs/heads/master"},"0a1c563ea5ccd47af2cd2ed637117e29f9077196":{"kind":"REWORK","_number":5,"created":"2023-10-24 13:47:47.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/5"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-24 13:47:13.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a1c563ea5ccd47af2cd2ed637117e29f9077196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a1c563ea5ccd47af2cd2ed637117e29f9077196"}]},"branch":"refs/heads/master"},"aae7db0f2a8d931a85f5520ac0af14e284e3ed53":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-24 13:50:38.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/6"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"michel@nederlof.info","date":"2023-10-24 13:50:38.000000000","tz":0},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aae7db0f2a8d931a85f5520ac0af14e284e3ed53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aae7db0f2a8d931a85f5520ac0af14e284e3ed53"}]},"branch":"refs/heads/master","description":"Rebase"},"078876aae0c47a0c90a35b58cbe96aa3d9898217":{"kind":"REWORK","_number":7,"created":"2023-10-24 15:49:22.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/7"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-24 15:49:08.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/078876aae0c47a0c90a35b58cbe96aa3d9898217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/078876aae0c47a0c90a35b58cbe96aa3d9898217"}]},"branch":"refs/heads/master"},"529439523053685db8308f8ffd03edd461c48043":{"kind":"REWORK","_number":8,"created":"2023-11-13 15:37:23.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/8"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-11-13 15:36:41.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/529439523053685db8308f8ffd03edd461c48043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/529439523053685db8308f8ffd03edd461c48043"}]},"branch":"refs/heads/master"},"75d9cbe001288dd72311135bff9dbd9ca4038da2":{"kind":"REWORK","_number":9,"created":"2023-11-13 15:46:34.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/9"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-11-13 15:46:20.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75d9cbe001288dd72311135bff9dbd9ca4038da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/75d9cbe001288dd72311135bff9dbd9ca4038da2"}]},"branch":"refs/heads/master"},"137744edd362e6c4a7eee568422002e36e24dbe2":{"kind":"REWORK","_number":10,"created":"2023-11-13 15:50:54.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/10"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-11-13 15:50:44.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137744edd362e6c4a7eee568422002e36e24dbe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137744edd362e6c4a7eee568422002e36e24dbe2"}]},"branch":"refs/heads/master"},"06b3735cd630eaa60ed34b9640a42c1b417dca72":{"kind":"REWORK","_number":11,"created":"2023-11-13 21:35:10.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/11"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-11-13 21:34:53.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06b3735cd630eaa60ed34b9640a42c1b417dca72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06b3735cd630eaa60ed34b9640a42c1b417dca72"}]},"branch":"refs/heads/master"},"b373fbd3fb7f2dbe898946ff97860cb747b22fd2":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-11-13 22:03:46.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/12"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-11-13 22:03:30.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nCloses-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b373fbd3fb7f2dbe898946ff97860cb747b22fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b373fbd3fb7f2dbe898946ff97860cb747b22fd2"}]},"branch":"refs/heads/master"},"e2f8894ce86ecb06d80c1905b9bafea5d00966fe":{"kind":"REWORK","_number":13,"created":"2023-12-07 14:07:16.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/13"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-12-07 14:06:58.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nCloses-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f8894ce86ecb06d80c1905b9bafea5d00966fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e2f8894ce86ecb06d80c1905b9bafea5d00966fe"}]},"branch":"refs/heads/master"},"cddd4e08611184512b8190988697455bf4f3a300":{"kind":"REWORK","_number":14,"created":"2024-01-11 11:11:44.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/14"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-01-11 11:11:28.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nCloses-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cddd4e08611184512b8190988697455bf4f3a300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cddd4e08611184512b8190988697455bf4f3a300"}]},"branch":"refs/heads/master"},"e68a920c114010a36a667d90866bfb6243148b6d":{"kind":"REWORK","_number":15,"created":"2024-01-12 08:50:46.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/83/896883/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/896883/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/896883/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/896883/15"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-01-12 08:50:30.000000000","tz":60},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nCloses-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e68a920c114010a36a667d90866bfb6243148b6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e68a920c114010a36a667d90866bfb6243148b6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
