)]}'
{"id":"openstack%2Fneutron~896939","triplet_id":"openstack%2Fneutron~master~I5b7b55b0b365df7246a571cea97a392cdf89bdb6","project":"openstack/neutron","branch":"master","topic":"bug/2036705","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-19 21:01:39.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-10-19 15:48:25.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-10-19 21:01:39.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-19 21:01:39.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-10-19 17:20:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"hashtags":[],"change_id":"I5b7b55b0b365df7246a571cea97a392cdf89bdb6","subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","status":"MERGED","created":"2023-09-29 13:38:05.000000000","updated":"2023-10-19 22:02:27.000000000","submitted":"2023-10-19 21:01:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896939-bug/2036705","meta_rev_id":"65da6519e4eb14238097bcf4be4e9c4eb019e733","_number":896939,"virtual_id_number":896939,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-10-19 18:39:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-19 21:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-19 15:48:25.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-19 17:20:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2023-10-19 18:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2023-10-19 18:20:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 14:09:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 15:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 17:22:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-06 15:35:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-10-06 15:35:38.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-06 15:35:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-10-09 21:31:49.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-19 18:20:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"5da044acad03669cb146a4fb233cb8cf3905dbe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 13:38:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc17df0ac27042154d80ddaf53b3f440b88865a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 14:09:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/39/896939/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0b40a75f7ff241359ba41225db9a3cb2fc383f79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:47:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca45d5a276ac4a5c8ad57721cb47584b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a0400e7d3efb4498bade4ea7c5bc978b : SUCCESS in 1h 58m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b698660f12b2416c8f1ad625a7b4f4d4 : FAILURE in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c74ca7cd264efdb60a3f41d0281d82 : FAILURE in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60a12380d16a43dc9adebcdbc788d460 : FAILURE in 21m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f68f7ba64464b5a8cfc329be100d8fc : FAILURE in 27m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2436f97e79884bea8faff3a907e1f3ca : FAILURE in 27m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/089b7f7d13944d738e44ff255b848cfb : FAILURE in 22m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7e73f6d5bf494c83c11c510270a5a6 : SUCCESS in 9m 51s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/928c996117a944a4ad8b394a0facadf9 : FAILURE in 28m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca2ab293378a4ac3bcfe6162cf4ddadf : FAILURE in 56m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c543606502a54bcdbdd4841a1da5c640 : SUCCESS in 46m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7c730e37846744bbbb2f07cf2997b725 : SUCCESS in 1h 26m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/761be1c9c662413ebc20d5642de50ca8 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bbd55fdbbc1c60e1b806e4c774b16b2724b57385","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 15:55:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ef136ca29d100e455fcd15579613f6d39bb13ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:11:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a421bcb263734703a93a4b92f9658696\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e95087a4f6c247539f4030d7ef04988a : FAILURE in 1h 06m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/922a53fa062048d3b3b888b87448ab14 : FAILURE in 43m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c221422a0c584e34bf8e87c74f57253d : FAILURE in 1h 03m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2bcf2f6e7de24284b1f8cfe8834a534b : FAILURE in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f111350cd699268d0d627a1f4f1ab58c00b58a67","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-02 16:17:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4f391472d605023aa1bf31dcba90a16547450c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-02 16:23:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdbe6b1093bdc670b888d3ba513173243b4d7762","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-02 16:54:11.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/39/896939/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"bee2fceee616f8babc017920d37af7820b4c9a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:01:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8af80c450fd546869a5f9572e56d55ba\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f90902f5bbb344c580fe6d6a330dffdf : SUCCESS in 1h 30m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58d009f20d6046faa945b00066d8cd38 : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2b79760468416ca130fb0fffe1922a : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264742f5f1b249e5ae9bc941fc268246 : FAILURE in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/834f5340cddb456b8bce5d6695c4f019 : FAILURE in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b15131c600a4cc681692cd25131111e : FAILURE in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81e3c21074d3403aae959a99ef03eccf : FAILURE in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbb449850e34da8b765bb01ebe7fb33 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11b9fc0fd0f144cbabfd136ac2a25b6c : SUCCESS in 18m 10s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/790e303b0c9e448097e56e2d0adaa8dd : FAILURE in 6m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da45b73a46c4460d9bcdbd51de82ec0b : SUCCESS in 1h 09m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cfa528e3e82448dc8c3666a55fc6c19b : SUCCESS in 43m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3b96644a40924efe904b8e5a64661654 : SUCCESS in 1h 36m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b104f70579c6489b81c13d534e711746 : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":2},{"id":"e4d8184829c6a57a4f73593f826c154a1509c076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:47:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/434284c90a124085a1d325291193d50b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1bd448df9414bba9f559cee0178e54b : FAILURE in 10m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/090cebd3b8164f90bde5fbd711cc53ff : FAILURE in 10m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/24da698e91d1401e88da1633be681e1e : FAILURE in 10m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf9ea86cd3f642dbbbe5f116808f1582 : FAILURE in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e975a6a458ddbb135bb5d5f5030fa331fbd5b5e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 10:00:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"230c09ced50b0c11f3661795bb2dba13401a8dee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 10:30:29.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/39/896939/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"5d53bed7df5a2fe12faa17abfbcd1903c88d45e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 11:43:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6b8419654c4e469864b783b6f6b5dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c51464e14b54aa1a636283c92686eab : SUCCESS in 56m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf365f5cd2524978990e0c10942630e8 : SUCCESS in 46m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52ae19ae864d42808099c1d1b5973d77 : SUCCESS in 52m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2aa68bbe5bea4f908fd5cba079ed7eaf : SUCCESS in 47m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92bc7d436dbb2bf71baf4a7db1baf3332fcfdba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 12:11:31.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77921e9f1522490c819c7c3230a1434c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2e18c5ea93948429823dcea01f5fdd1 : SUCCESS in 1h 56m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/908c94fdb6d3466da656decf8367d4ee : SUCCESS in 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d29833b09e430498cc1bf18d4bdc8e : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e89fa032c4944da6a0edef568733ef1a : SUCCESS in 24m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfc8c219d2894dc3b45aa6c69d6c8777 : SUCCESS in 25m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e0b3c4e0d5fa49b1b1f78ddfe83223ed : SUCCESS in 27m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b418d28f58c4d38a5fd1476edb38c08 : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a77c1f725b44bcbb82ae66dfc688be6 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d257de8728df4164a5a5b6cea44413c7 : SUCCESS in 16m 41s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8ce50b5bbbf549f0a73f7528a101274e : SUCCESS in 31m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3663e6c7061d40bdb3262e1dc36e73f3 : FAILURE in 57m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/787b4e4aaf67474090b6dd772a5ebebe : SUCCESS in 1h 03m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7470bedc19ee4c8b930e6314b64822c1 : SUCCESS in 2h 05m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1b121659c1de45268648ed3b07dbdae6 : SUCCESS in 1h 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5b477729b9043fa8e7bed2ce7bb04d07b80af8e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 07:33:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"baeb511a196822c385c56261d7535e0b05be54db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 08:26:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67c09d4a68f244bbbf7bc365a7def794\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38c47ffc01984df7b10caca87505f83c : SUCCESS in 49m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24d3e039f381481f89652ad2fd7565cb : SUCCESS in 49m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b70755543e745b1bb61c0c39513ace2 : SUCCESS in 49m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e47704f00f11472a8162d12aef62667f : SUCCESS in 47m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d54753f41788be44c6c3c2bf55d6b4e2fa97e73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 09:20:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc613cc2d704ad3bfafa158fdbf0ad1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aed3aabbb5d4490181c8ecad90672e33 : SUCCESS in 1h 44m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88ccc418aa04087885757704785d2d5 : SUCCESS in 37m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9b154f7f5847999343289bedf8ffdd : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d94d3dc0684192b0280b013153e429 : SUCCESS in 25m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8401e6fb404440eb915c6185db7cd7e : SUCCESS in 19m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ea2fec457764204bec02102c0993e02 : SUCCESS in 23m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a15f8be1bda244f0b86621bd21b09d66 : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5f121b128f4e0e89914e3eea663d1a : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d5b08a594c8483e912f5b07f7d80abb : SUCCESS in 13m 31s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1922c7734abe4921b52ba30bd2234014 : SUCCESS in 29m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4dc35fd84ce45cba36caef92f6dde56 : SUCCESS in 1h 03m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a187ea2240134e74a4d39f351f826b5a : SUCCESS in 1h 05m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/63dad02b68b5449e84b9c140210164cd : SUCCESS in 1h 35m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ef7fb1be99974930905a5a1aef1e1c28 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b74f0383e0f5768ccecb0bd8c370f097bfb27406","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-09 07:47:08.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3ba4e3d52126d3780669eddcf3bf7b67c272e1fa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 08:47:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"47d4aa3121a13786685230dc85ae4609c07945a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 08:49:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0f82fd956c586aa3ca639826072f648a9d96324e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-09 09:04:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7cf830f36816ecfa34a79b8fcd390d4af863d1ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:42:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f9ffcb9aee84e8c8f89640de9f9bd2a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/018c28338c6040f887736bf0cce89e2a : SUCCESS in 1h 22m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d9890b8090a4d6987ae9cd801a0123b : SUCCESS in 38m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869cacf6fc4e484badd0412819d11a45 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e093ee0601c443d395481d647f895e0b : SUCCESS in 28m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d313b66b0a747b3b279b745ac3df3a1 : SUCCESS in 29m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffbd884ef5c948c28d136885970d5331 : SUCCESS in 21m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1146defd56ec47cfb6bb4190d1f53135 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa410b027d744917a72e6e28637510d0 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d42113b5f7544a9bcbb33a21a61611d : SUCCESS in 18m 25s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/23871a36a8264534b6d0f36711813eb8 : SUCCESS in 27m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c306990a13945b6a7528e98fdb02178 : SUCCESS in 56m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2da6bc68033746a79a3b9ec636fcab89 : SUCCESS in 1h 03m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f0e8460e778479c864c19c8b49c5e43 : SUCCESS in 1h 45m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d470cca129b408683a0e54964128b48 : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f94b6d17683897f7df608884f09785f6f6a7568d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 11:10:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab0e31e54dbd4654b36f46683974d515\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e7d89bcc0c64d5a92796886d17b931e : SUCCESS in 1h 20m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef366e1919814f89b7b46922576cb909 : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1521a9ce168478ca02946aeb970610a : SUCCESS in 41m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca70ac744c184be2a864d6e28a067b8e : SUCCESS in 38m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2731d94f5b81ba5953e59cd2e3b6c187af992fd5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-09 21:31:49.000000000","message":"Patch Set 5: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0e34a39a463d90c2fa52d4d3388d6a61aa195992","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-10 13:21:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6319e8d0697adb0b47c813fa4541638bce5907a8","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 15:00:19.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_6773\u003e\n","accounts_in_message":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}],"_revision_number":5},{"id":"e72b6b213aa86ccaca76647845f0a9ad6e5127ea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 15:00:29.000000000","message":"Patch Set 5:\n\n(11 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4815b84b5c31e145c26e58d5f71ae205f24c1d31","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 15:48:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b58cb5688a6b74dd08b3f6c0debb5cd1cb548535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 15:59:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"08f3c103aeacdead2db8a7ab6584dad864ab9204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 19:02:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee2707b2eb2e4642adf9bf86cdd4ae87\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/51debb38c15f4598ab6877084b9ccc9f : SUCCESS in 1h 29m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e19974dc82d04bf89565f241ba0a69f7 : SUCCESS in 31m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae1e8444acf74749ac42c56e521f8c92 : SUCCESS in 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f2f98bb8bf043bfb99c1d07a070c7b5 : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09c91fa5769940a4bf226394b1975c0e : SUCCESS in 28m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb4e3ad06f694dce9589eaf67dcaec41 : SUCCESS in 27m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4083aefe3add42208a99f7ee6abaa5b1 : SUCCESS in 29m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/100ff290f9514ca3bc91e37c321eb057 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899d78d6564f41a5a4798ad9e26c567f : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eadab2b7d2b24fffbbcd8ace1cc8599c : SUCCESS in 15m 26s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/94177e649f6c453dba27eb6a8ccbaecd : SUCCESS in 28m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd6881ebccd242dda8bc827e24ff9068 : SUCCESS in 53m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca1e0d9f71324815bee34795178de4a4 : TIMED_OUT in 3h 02m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0e4fec790d248b6985e375aa66fe64e : SUCCESS in 1h 00m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a7982ce0b2824e349c2a18e87ad36bd6 : SUCCESS in 1h 58m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7af322f49ae3459a9fa633168d7aaa45 : SUCCESS in 1h 17m 46s","accounts_in_message":[],"_revision_number":6},{"id":"c34e5f347e32c37ad8a980a18c3c41a36527acd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 19:09:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d05f124c4c740cabd0eb1a8551a67b1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/86db9cf9672045e4b1337e982fd84987 : SUCCESS in 40m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2542daf088b344d094a63bdd8226dd62 : SUCCESS in 46m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d73b2c4719e64913a8d1a618a883db5e : SUCCESS in 40m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17c64178b248470bbb12451960d03cc2 : SUCCESS in 37m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e12616c7ac1476be1676938ccb89e41c00cc97e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 08:45:49.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"cd98d640092a8f955d8815400496b0075dc8abaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:03:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a3f21b4bb140eebf3ba62ed3d14bc9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f4890043d984b47b1a318a7bf1f0cb5 : SUCCESS in 1h 54m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe2cf74f64c347c19d144105ef85f0bf : SUCCESS in 39m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e20d4cfbc864f3a9cf1303adc697d2b : SUCCESS in 44m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81456b077ed435192e4277e5f1b35c7 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a990aaa8d78c408aac9f6b90f3e4a235 : SUCCESS in 32m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a976ebb940e84f7086961dfa0d9d75bf : SUCCESS in 31m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ff9bb9238eb44deb64d09d101f72e5e : SUCCESS in 24m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66f60a82ad814a85bce7bcf974926da5 : SUCCESS in 23m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/114c4c4f85114d43b7c0b10ea45362e2 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/699a9a1d90c94b679e05d93a35bcc1ce : SUCCESS in 16m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/50ec0bd0bd1a49f380a21da718bbe213 : SUCCESS in 26m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/86860d966c85487aa70dcfc92fac4ded : SUCCESS in 52m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac10c618b7c244f29dbf7c596bbbe227 : SUCCESS in 2h 15m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b9fdc04997c54b5c83447ece0ccce88f : SUCCESS in 1h 00m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f78fd8f619f14a3382a45b664d734051 : SUCCESS in 1h 43m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/83765294e15f40b8a312faf4c4ccd787 : SUCCESS in 1h 06m 45s","accounts_in_message":[],"_revision_number":7},{"id":"827488bb97b80e9f0810d72a954bb7326db46cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:45:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86b136d3c84f46058cfc85d8ef5c7da5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e448712b30a24e65abe2236ad696cab9 : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b58da371aa18412bb048a5791ae04689 : SUCCESS in 1h 11m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a0bb70b47384b47afc5af48d68e2fbe : SUCCESS in 41m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e981624e83fb4389a981e677ac912f8a : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cf7f8234e087b0f17cba61727c8cc458141a2c4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-17 13:26:36.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b3e9ffaa5b6d0aa4608be4ee7735afc13062fd03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 14:22:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/907d12c1195341ffa3ea3c148a0e7c30\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c0b2f1e3cc452581681e7cf3245586 : SUCCESS in 49m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd753fe91e8a445aa8365cb968fa3ae2 : SUCCESS in 47m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d7cde05e1df4a8aba0d76adc48695ff : SUCCESS in 52m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6ce893948c34ffeaecdbb9b6c5d46f9 : SUCCESS in 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3c8c4b2e87347c5621d0e36a7b58b5990f8ddf5b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-17 14:30:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/39/896939/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"cab96023052de6f443da1e069a7673aee22f9e11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:37:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef3d77395846437ab1dca99ae3cdb333\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8b572addbde400d89c96f006b6c69a2 : SUCCESS in 2h 12m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06cd6107c4c6458bb3175314c9de9e17 : SUCCESS in 34m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d05d7a90ff1942938f7542e665f71ce7 : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c3bc21eb0348e987810b50702a2b4b : SUCCESS in 8m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58920c8c02854b63b84570d61b6256b3 : SUCCESS in 36m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40b2dab7ffbe4f1dbbe65e807c3463d8 : SUCCESS in 25m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/954d78ddd32e40bcb8da2df817678457 : SUCCESS in 27m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf30842694fc4d29a9e8cb7b2194b07d : SUCCESS in 24m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43cc2ee0d57f47148afe08415a59a0f5 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbe64fd71b3444a48ab75c51c40f0096 : SUCCESS in 14m 11s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/52ef6eede5ca4675be65dd94edbb4e46 : SUCCESS in 28m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d60f31820134d0faa15bf445a71031c : SUCCESS in 1h 02m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f25e3ec7090b4f4a865b50eab59cf0ae : TIMED_OUT in 3h 02m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/269a17e1375147698ae745669e9fb4bf : SUCCESS in 52m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8449a78a0c0b489296f62c9f37105408 : SUCCESS in 1h 47m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c396fbd9d7704c1881519f369e0ed958 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":8},{"id":"2029ed9a8dafd2b7bbfc32371654bc1e9e17c958","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-17 16:50:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e0bfd039dbf3f7d19613b63243cb8ba50465f1a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 18:50:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938a81b3f59f4fed84c4592b8a735325\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/660899a7032d4f9594f6ae4ba38d60ed : SUCCESS in 1h 15m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c06630d2353433895e1b4f2db581266 : SUCCESS in 1h 15m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45479dc756e649139d1c211597a1f2b2 : SUCCESS in 1h 15m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/18195b4027de4200b01d0c020ce99d2a : SUCCESS in 1h 08m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dd5b38e5bb0c69d130a42e9732e7abe4f3161a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 19:40:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb6a8de72c94f809677886775815733\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cc59d4cd5db84f329d0ab1b2b081c21b : SUCCESS in 1h 38m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1e33889cd9d4526b482dc3b5949201c : SUCCESS in 38m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/687338b153154857aac03ded5339ae3d : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65dc3a46a2ba44689b501808b0d44334 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d44e0527786347b5a9012917315115aa : SUCCESS in 22m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aebed6aaf3648cbbcbe6698ce888c2e : SUCCESS in 25m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b7c39381d8f42f7885cecca9e81d253 : SUCCESS in 22m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aca730c066a74adbb0e7623a1bfe406a : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e9c5c1ed1a4f2c8526f7ee1858d7e4 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03e8913f056c4bb8aed08188eea25553 : SUCCESS in 17m 56s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d2926f2e160f403bb6b64ee52f27dcf8 : SUCCESS in 25m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28284679616240968e9d49d0c8987cf3 : SUCCESS in 51m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d9c8b4cbe284f89a455ecc8eb3273ff : SUCCESS in 2h 20m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4343ea1ab8194e00a7cdb2861b4ca0c2 : SUCCESS in 1h 02m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b16a9206a9bd4cf991ec85be723470fa : SUCCESS in 2h 04m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26fc1fc0567f4ef9892107c4a9cf7c7e : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":8},{"id":"ed21dcaed8b5d760b4c156c05407f1dafecacfec","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-18 20:05:12.000000000","message":"Patch Set 8: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"57c884b10f25128719d360e0877f92d163d122f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-19 08:58:53.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"abb281f18bb9888419d289f5e2bf83271250a6bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-19 08:59:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"0cb943e42bea8fd02226fc0fed14ba1c7464494e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-19 09:48:29.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/39/896939/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e919f09e624c6ab4856df1bcf27de44a2d8ed664","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 10:56:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8833bacdb80410a9f4dc70b7299ba89\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d68e781d1ec4d5e9fc89e470161351b : TIMED_OUT in 1h 21m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84f7ac1a25f3426db6eb9f59c33859e5 : SUCCESS in 46m 22s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/211fc1945d6946df8087940d7c6638ca : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c9497fcd84d4fb6bbf8c5d9da01cc7b : TIMED_OUT in 1h 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dfd1dd0fab63716114be3753d42baadd2c7dc44d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:23:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0923524fd27849ed821202c5c2986c95\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f66648d3b8f40308e12328225e84430 : SUCCESS in 1h 47m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1845ba403864f3ebcc739cb854fb536 : SUCCESS in 33m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8aa6b5e9b2f426691f16bad9093fcfb : SUCCESS in 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6813a8818330491b9d41c047c0efb5a6 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6f6395212ba47f99c497d7edae3e590 : SUCCESS in 33m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14cc604e2f244514b3139289cfb8ddb4 : SUCCESS in 24m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fdf6fa2ac9c440d9cf4372de62304e9 : SUCCESS in 26m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b28c0f32edd24c29bbcfc64ff4fe07aa : SUCCESS in 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887e1d55512c4944a452101a7af37144 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d1c7a2ee430429289aa71414d8dddbe : SUCCESS in 12m 57s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8563585632904f9c83092edfd811ccad : SUCCESS in 23m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3276f2eef8454e89b9cf816e1b482c60 : SUCCESS in 55m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/610f941862a9461e8d4717f1645c91f0 : SUCCESS in 59m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7090b91bda2949e385a3c39c7fa306a0 : SUCCESS in 2h 18m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b537bc2635674b1faa7bd808626f475d : SUCCESS in 1h 04m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb3da05763524860ba471db23e19bc89 : SUCCESS in 1h 54m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/29f0820001cd4cf98b2c1bdf57b086a0 : SUCCESS in 1h 11m 21s","accounts_in_message":[],"_revision_number":9},{"id":"0e4eba7da50b0965faf00d227885405bbf74317f","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-19 12:53:40.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"7ba4e9ecb850abfe851210234f6dd492e08ffd7a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-19 13:23:22.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"79f97e0f21f4a9622c7b04ef5d7ff787dc5de612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-19 15:30:38.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3a63cd13801166377715f3a8b6e5587dda9b1b20","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-19 15:31:19.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"983dcc1144c9f6f1bc25087892a2bcfe52853b4e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-19 15:48:25.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f6b63b562803305d186955fb7c7fda014f2a8ab4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-19 17:20:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c9607833f78a72670606766a3cf40b37c661779a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 17:57:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47982dd0c414471ca5e7ac11fff976e3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49951d99988342168ec2ac82e556a450 : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/adf6aa0d5a644942b1bc692f656beb56 : SUCCESS in 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d1c9523c774d179d9b1d00c5b7a171 : SUCCESS in 40m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03dba92e95174edcaf7d521dfbb903bc : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f699987c2be4bde8a7108df75914b6a : SUCCESS in 31m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/140f8b49f0814be1a74f9ea723da13a2 : SUCCESS in 31m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4b8d2e20f6f4841af4a52b3692fdb0f : SUCCESS in 30m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff1ea492776a46eb8797eac6bbe9913b : SUCCESS in 24m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8151ccfe717d4a19b5dac36352f33a62 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f2240a4217c4eb4b6efac95134b4317 : SUCCESS in 18m 23s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ac6dc4540d474bd2a8f2a8829768ec10 : SUCCESS in 31m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/76a36d966e1943829839dbcc4de9dda7 : SUCCESS in 1h 01m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/190f79cfd552425485ba0f0e8bb78aad : SUCCESS in 1h 07m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cf34053b8c04257a92798ccc25e9264 : SUCCESS in 2h 21m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6a43ab5d2c2d49b986e02c2b04e3dfc8 : SUCCESS in 1h 02m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33f55e60c7ea43da9f2f55743e740c43 : SUCCESS in 2h 03m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/53b4cef155364c87906bfb6b9324e070 : SUCCESS in 1h 20m 41s","accounts_in_message":[],"_revision_number":10},{"id":"e709f356c41db5b61ddb3062c38c96160e66a7d0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-19 18:20:23.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"647c66c448c7d6d4ec13c960cfb86f48e63742b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 18:21:01.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"135e415ef363f924ed4c0550e4d829bf51967e64","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-19 18:39:29.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/39/896939/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"aec52a430d3cbef771a7790a0c5c64d3e3cb5549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 21:01:39.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eeda11385aa4a4ca16921599b470479\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/15633b53d7c64ebc85d0c129978017ac : SUCCESS in 1h 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a0dfae7eedf476181936ea11d02b59d : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a09e85c9b947486ea645794a6552b73f : SUCCESS in 19m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e0c006103374e3882fe8607c7871916 : SUCCESS in 26m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ad0d401285f45329ad6fd34cb4fc9dc : SUCCESS in 22m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6d10f3a2a1842c2b319768ae7617e6e : SUCCESS in 25m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30fbf24cb01c4125bf241aa4ef00f307 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f4d837bfb9749a3bd6cae6d86e5f4f3 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ea0ab80ed604abab3cab27480676333 : SUCCESS in 1h 00m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/430868611bd74b6b9ca1b818e43974d4 : SUCCESS in 2h 34m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/51bfb3f352c347aca43cc8a9b5cd5d39 : SUCCESS in 46m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a7b36162d6544eba98659eea0e4c3c0 : SUCCESS in 1h 21m 55s","accounts_in_message":[],"_revision_number":10},{"id":"b565759600552d25c95f35b2eb88bb2be9e645f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 21:01:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"a7628cc0c1bb48c726b109eeb6d334d1cc30959d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 21:03:34.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b1828a2492d4ec091bae67d44084cd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ceadefe0f444e281cf62d9c53d1c54 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/400d8725465346ba8f7f60dae77d46ff : SUCCESS in 52s","accounts_in_message":[],"_revision_number":10},{"id":"65da6519e4eb14238097bcf4be4e9c4eb019e733","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 22:02:27.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca3f7e53bfb41f28c8d5cec4d2aa97c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5fdc069aac14bc1944dc47d00222230 : TIMED_OUT in 1h 21m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5df2c2d425444c77a2c7d2facf763e3e : TIMED_OUT in 1h 21m 08s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3401ecdd72cd45799282f95882a4e973 : SUCCESS in 44m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a9eb136b042437eb1f22a6b08971b01 : SUCCESS in 1h 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"aead4aa99df4769462ec47045c8d973ae1b0a0a0","revisions":{"0c187e1d9c12bbe6c16940f4dba636145caa19e3":{"kind":"REWORK","_number":1,"created":"2023-09-29 13:38:05.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/1"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 22:24:16.000000000","tz":0},"subject":"WIP \u003d\u003d [OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"WIP \u003d\u003d [OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nTODO: testing, documentation, release note\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, ther user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c187e1d9c12bbe6c16940f4dba636145caa19e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c187e1d9c12bbe6c16940f4dba636145caa19e3"}]},"branch":"refs/heads/master"},"40241bd78ac90fc5645fb896d5d8cb2eeed1bd30":{"kind":"REWORK","_number":2,"created":"2023-10-02 16:23:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/2"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-29 05:51:09.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, ther user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40241bd78ac90fc5645fb896d5d8cb2eeed1bd30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40241bd78ac90fc5645fb896d5d8cb2eeed1bd30"}]},"branch":"refs/heads/master"},"4a40b5fdeea3444635a00d3ed20ff62ad93e2a1e":{"kind":"REWORK","_number":3,"created":"2023-10-05 10:00:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/3"}}},"commit":{"parents":[{"commit":"a76aa019772cc35b07a404c213b327169c188243","subject":"[OVN] Add the default condition check in ``PortBindingChassisEvent``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a76aa019772cc35b07a404c213b327169c188243"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 09:59:45.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, ther user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a40b5fdeea3444635a00d3ed20ff62ad93e2a1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a40b5fdeea3444635a00d3ed20ff62ad93e2a1e"}]},"branch":"refs/heads/master"},"3664df777f4dbf8d147c7b6338a4e9c8c4d4235d":{"kind":"REWORK","_number":4,"created":"2023-10-06 07:33:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/4"}}},"commit":{"parents":[{"commit":"2066dcaea228c4b872048a486fa3b526d211c178","subject":"[OVN] Populate the \"router.distributed\" flag in ML2/OVN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2066dcaea228c4b872048a486fa3b526d211c178"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 16:43:12.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, ther user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3664df777f4dbf8d147c7b6338a4e9c8c4d4235d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3664df777f4dbf8d147c7b6338a4e9c8c4d4235d"}]},"branch":"refs/heads/master"},"7a288f752b28269328106c43df190e221fd88917":{"kind":"REWORK","_number":5,"created":"2023-10-09 08:49:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/5"}}},"commit":{"parents":[{"commit":"a45263e146ff8fe3fd8aebe8a9a83d0827b7043f","subject":"Merge \"[S-RBAC] Add service role in neutron policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a45263e146ff8fe3fd8aebe8a9a83d0827b7043f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-09 08:49:31.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, ther user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a288f752b28269328106c43df190e221fd88917"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a288f752b28269328106c43df190e221fd88917"}]},"branch":"refs/heads/master"},"70f58ecb481064129f366e4a59147e84ab312a5c":{"kind":"REWORK","_number":6,"created":"2023-10-10 15:59:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/6"}}},"commit":{"parents":[{"commit":"78027da56ccb25d19ac2c3bc1c174acb2150e6a5","subject":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78027da56ccb25d19ac2c3bc1c174acb2150e6a5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-10 15:58:52.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, the user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70f58ecb481064129f366e4a59147e84ab312a5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/70f58ecb481064129f366e4a59147e84ab312a5c"}]},"branch":"refs/heads/master"},"3c1ef566e424cc94eed4f2c5ef2b8563b3a5cfca":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-10-11 08:45:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/7"}}},"commit":{"parents":[{"commit":"0e17bd15ebb5fd0819b183fdcc628c5f8f22f850","subject":"Merge \"[fullstack] Unify ``TestQoSPolicyIsDefault`` tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e17bd15ebb5fd0819b183fdcc628c5f8f22f850"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-11 08:45:49.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, the user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c1ef566e424cc94eed4f2c5ef2b8563b3a5cfca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3c1ef566e424cc94eed4f2c5ef2b8563b3a5cfca"}]},"branch":"refs/heads/master","description":"Rebase"},"615fe7733740a8ff0ad02cea25bf616e0ccbf93f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-10-17 13:26:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/8"}}},"commit":{"parents":[{"commit":"169deb24500caa43ce3f6c8cd837d2286d219d5f","subject":"Merge \"Add constant to identify OVN LB HM ports\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/169deb24500caa43ce3f6c8cd837d2286d219d5f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-17 13:26:36.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depeding on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, the user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/615fe7733740a8ff0ad02cea25bf616e0ccbf93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/615fe7733740a8ff0ad02cea25bf616e0ccbf93f"}]},"branch":"refs/heads/master","description":"Rebase"},"06a839fe88349d70fe3f41a6be8abfddcf76898d":{"kind":"REWORK","_number":9,"created":"2023-10-19 08:59:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/9"}}},"commit":{"parents":[{"commit":"55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17","subject":"Merge \"Add policy enforcer for \"tags\" service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-16 01:29:37.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depending on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, the user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06a839fe88349d70fe3f41a6be8abfddcf76898d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06a839fe88349d70fe3f41a6be8abfddcf76898d"}]},"branch":"refs/heads/master"},"aead4aa99df4769462ec47045c8d973ae1b0a0a0":{"kind":"REWORK","_number":10,"created":"2023-10-19 15:30:38.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/896939/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/896939/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/896939/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/896939/10"}}},"commit":{"parents":[{"commit":"1879d925330af5598a105a8893ab6cfda9dc37e6","subject":"\"ebtables-nft\" MAC rule deletion failing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1879d925330af5598a105a8893ab6cfda9dc37e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 18:43:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-16 08:01:00.000000000","tz":0},"subject":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"","message":"[OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n\nThe Neutron \"port.status\" field (\"ACTIVE\", \"DOWN\") is set depending on\nthe Logical Switch Port \"up\" and \"enabled\" flags. Before this patch,\nthe \"port.status\" depended only on the \"up\" flag. However, the user\ncan set the \"port.admin_state_up\" field that will modify the\n\"lsp.enabled\" flag. If the \"port.admin_state_up\" is DOWN, the port\ndoes not transmit and the status should be \"DOWN\".\n\nThe OVN backend is correctly disabling the port transmission; what\nwas incorrect in the Neutron API is only the \"port.status\" field.\n\nThis is currently working in other mechanism drivers like LB or OVS.\n\nCloses-Bug: #2036705\nChange-Id: I5b7b55b0b365df7246a571cea97a392cdf89bdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aead4aa99df4769462ec47045c8d973ae1b0a0a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aead4aa99df4769462ec47045c8d973ae1b0a0a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
