)]}'
{"id":"openstack%2Fneutron~897019","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I01fa481456c0a675b5857d9796b8323493e4fbc1","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2037717","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-11 08:43:23.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I01fa481456c0a675b5857d9796b8323493e4fbc1","subject":"[OVN] Add the default condition check in ``PortBindingChassisEvent``","status":"MERGED","created":"2023-10-02 12:52:16.000000000","updated":"2023-10-11 13:16:54.000000000","submitted":"2023-10-11 13:15:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897019-bug/2037717","meta_rev_id":"05f9ace8c2c6bca58a20bc16b30bcd6c0e6f65c6","_number":897019,"virtual_id_number":897019,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 13:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-05 08:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-10-05 17:51:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-05 17:51:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-02 14:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 08:21:05.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-05 17:51:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"d41e8b45ffc2188df69fd3dd470d07a23fdb280f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-02 12:52:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"285ab45bfd3ed11fbcb28723b2d42b399ed772e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 14:46:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bc97b652f34498bb88a3b4f44c68afb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d53987dff6a14d3a98125920adcfb4d2 : SUCCESS in 1h 33m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ce500ea7b9a4365b0fd5fa059e7bb7c : SUCCESS in 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cee7411b30484eabdfdb42bc3b832a : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8856830a1ae14effa95b413fc222e80c : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad9f6df214a04a439dd0fa5b0fb58ec9 : SUCCESS in 20m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eaf8ba23b6eb415fa4b63e394ae20123 : SUCCESS in 20m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ccb18e2878ca440bb0330690a99c2eb2 : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c104a07a0e6141e8b2b211b7da7a3d74 : SUCCESS in 11m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01d629d50bc54fb2a49116c6a9ce9057 : SUCCESS in 56m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7051c5f03d0b499aa39b6199e96a419a : SUCCESS in 58m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df470fd1bbaf4f1f90e9ac10f0054c6f : SUCCESS in 1h 49m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e16933b129e4256baf8333ea965470a : SUCCESS in 1h 16m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f5485e0bcb9cc2b0f918d88605d023ce55248f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:09:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d67d0b6c204cc8b3082d162ed52d27\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d80f5a12ba034bd2a094222496ee78c3 : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb7a7a6904f24025b397f76cc5a7991a : SUCCESS in 40m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f13841287674c779104949af8c27509 : SUCCESS in 45m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/569cb4aed5344834b362cae5c9ae1d6c : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35bdcba134bfc7b25c512dd9ef4fd1c340ce47db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-03 08:26:04.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"356600212c001d04939752b471b6b2388cb141df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 09:48:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"47c1d8d7a9def8820f9a77ced5ae886ff3484fbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:27:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e734af7c8122413ea6281ef6e7f9db3f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5acb4c47c4d7446698e4e04edacc8c70 : SUCCESS in 42m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/656e619bc7af48749507c92063738575 : SUCCESS in 40m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df908e4844ba431a960a06a4cd7f3e0d : SUCCESS in 47m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a9e4e422c0c40f7868e33594e95c7bd : SUCCESS in 1h 02m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c223097505259ad241c5e02babe0c1f5d353f474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 11:44:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9e143622494adeb4a59ee7143e8690\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/253ed538b3ea4653a2b7b45e69467501 : SUCCESS in 1h 54m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a95ac01b181b49968ce4867da627aaf2 : SUCCESS in 39m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb018c6505b4a88bcbb076f2683ba8a : SUCCESS in 13m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e57e4a7c48134a5aafc403b8152cedab : SUCCESS in 26m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2533531889a84976b937af1500be2827 : SUCCESS in 25m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e37dd03eb07145a3858f9ad1bc37f354 : SUCCESS in 25m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b86dcd06f764c9eba1d1b1ea5ab9c15 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4206df0ec9449e99222a9eda2f72664 : SUCCESS in 11m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fc8dfcd8c814f4f9858ae1d850770ca : SUCCESS in 58m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7bd708efac2d445b870b67b169c8477b : SUCCESS in 1h 03m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f06a9ccef4b4466a9ad57aed07ec5845 : SUCCESS in 1h 29m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c943b3cf381417796161daa4f8ac081 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7390a89b68921b4a78ca27dd38f583e1731e5387","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-05 08:21:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d979209703424e957e0af29fd6b35790ae119f0a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-05 17:51:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b381575b4e800b51411704566d51d0e3eb21c9ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:51:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84133f987a42e83ab1ca6b37e82aaf3c5d062cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 21:26:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b543ec6102b4373a6aec68ee516e0d5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2b6a7fd5f80461c8adadee49b9c9c91 : SUCCESS in 1h 49m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30341e23b9ed4b59ae5468379b4756c3 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/758dd92af2054ef38d6f55643d9e1f4a : SUCCESS in 23m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b56911bc14274585b7b81de5303c1287 : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7513a247817446b95009665981f9b94 : SUCCESS in 24m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60430ac7c3bb4192bc8f2118a0431bac : SUCCESS in 11m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a28bb0376304e918f2f79dcaecd63d2 : FAILURE in 54m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e5f2c89ff0294da88ac6636fb618efe1 : SUCCESS in 59m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cdfb110053254a108fed74b17ef4c73c : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e5f73df5d5062ceb9745f842f08a22b5abaa2364","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 08:43:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70b68134cc513574da485a33d170b22198956696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:14:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c81e15807c94ccab7058ad9ba3c1e1d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f630bd610bb94a469850d54483f81245 : SUCCESS in 2h 29m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/984ad46f65ef4dd187018eb5d136dd6b : SUCCESS in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e0ff42c9924c46b2eb852a9aac70fd : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5eb5c6acc944318a8b27e95dad096f7 : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/380066b3e16b4a18831c92b7d9240103 : SUCCESS in 26m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e9a3b82a6744967a6b009f7285b13e5 : SUCCESS in 25m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2def3eafe7d647b7bb7c26a5c036a11f : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/445f90335bab4c5c9a17011c68a78afb : SUCCESS in 12m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41b32859073d42c9b119ad37fb46353a : SUCCESS in 1h 01m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e4abc47be804c8a9d944dc331654606 : SUCCESS in 1h 03m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1799386a67a444a1b6a6eb0cdff0f19f : SUCCESS in 2h 07m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9b5ffa1495434bb2b1d9d40c6b02e63e : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":2},{"id":"882822b054a1bc91a09acd37164b88d78c04f97b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:14:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c719cbfafdcbb245b6d4ed071edab286a9e1c448","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:23:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b01019d8f19488fa2db4b54accaa7be\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58a53bcaf0414fd5871d69f91c7b546e : SUCCESS in 1h 20m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32be8c4d2cd04d82843fe10beb64670c : SUCCESS in 43m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1ea93f720ad4c759b2029858868b2c9 : SUCCESS in 40m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bffc7294f6b0475f97629cfea44b71e1 : SUCCESS in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6254c2f495ba0cd61a8c7e5e9012d48b181be99e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:15:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/612ea5f740fd4df5b830c2e69b2fd156\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c645c182d7554602a0ab2b48c7668b9f : SUCCESS in 1h 59m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd01733115c7478392c4dcd1b6976fb0 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96a5e791dc2e49c7a04218eaf57c94d9 : SUCCESS in 28m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fb5fcd3ee94559921947c324b091f6 : SUCCESS in 28m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aad0f1092aec4eb698910cafb64d2c52 : SUCCESS in 26m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cbd0f6ce89549969e506f59f9c8017e : SUCCESS in 13m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/05a795e7fa894fa19cd0b8a555c58fbb : SUCCESS in 53m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d265201c0c884787bd8d8c4801ee34cb : SUCCESS in 1h 11m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bc3541067a7145619cf7f8def140f46a : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":2},{"id":"61081d89aaac7c395d5675df60bc6a1c0cc0fb8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:15:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"05f9ace8c2c6bca58a20bc16b30bcd6c0e6f65c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:16:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fb37d7d3174e7e8cb0e6b467cf23dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf6763f4da346f6ba06daeb80ec1a7e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61f94ab4493c981cea4844b5947f5699352c7fe7","revisions":{"240c28c4ace8198616d2fd417a4bf0cefb6314c9":{"kind":"REWORK","_number":1,"created":"2023-10-02 12:52:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/897019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/897019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/897019/1"}}},"commit":{"parents":[{"commit":"04e9b063ad526778ba80438fc851e03fa16e5085","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 19:52:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-29 02:20:23.000000000","tz":0},"subject":"[OVN] Add the default condition check in ``PortBindingChassisEvent``","message":"[OVN] Add the default condition check in ``PortBindingChassisEvent``\n\nThe event ``PortBindingChassisEvent`` should be executed only if the\ninitialization condition \"type\u003dovn_const.OVN_CHASSIS_REDIRECT\" is met.\n\nThis condition was skipped since [1], where the ``match_fn`` was\noverriden.\n\n[1]https://review.opendev.org/q/I3b7c5d73d2b0d20fb06527ade30af8939b249d75\n\nCloses-Bug: #2037717\nChange-Id: I01fa481456c0a675b5857d9796b8323493e4fbc1\n(cherry picked from commit 5da65c4803bfab80662a41b4cd20c6246b84ec31)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/240c28c4ace8198616d2fd417a4bf0cefb6314c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/240c28c4ace8198616d2fd417a4bf0cefb6314c9"}]},"branch":"refs/heads/stable/2023.2"},"61f94ab4493c981cea4844b5947f5699352c7fe7":{"kind":"REWORK","_number":2,"created":"2023-10-04 09:48:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/19/897019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/19/897019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/19/897019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/19/897019/2"}}},"commit":{"parents":[{"commit":"04e9b063ad526778ba80438fc851e03fa16e5085","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 19:52:41.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-03 18:04:25.000000000","tz":0},"subject":"[OVN] Add the default condition check in ``PortBindingChassisEvent``","message":"[OVN] Add the default condition check in ``PortBindingChassisEvent``\n\nThe event ``PortBindingChassisEvent`` should be executed only if the\ninitialization condition \"type\u003dovn_const.OVN_CHASSIS_REDIRECT\" is met.\n\nThis condition was skipped since [1], where the ``match_fn`` was\noverriden.\n\n[1]https://review.opendev.org/q/I3b7c5d73d2b0d20fb06527ade30af8939b249d75\n\nCloses-Bug: #2037717\nChange-Id: I01fa481456c0a675b5857d9796b8323493e4fbc1\n(cherry picked from commit a76aa019772cc35b07a404c213b327169c188243)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61f94ab4493c981cea4844b5947f5699352c7fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61f94ab4493c981cea4844b5947f5699352c7fe7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
