)]}'
{"id":"openstack%2Fneutron~897438","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ia8393f0e4f736dafb3ba7fb50cd2b679c24b1a01","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2038541","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-10 15:36:30.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-10 10:18:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-10-10 10:57:05.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-10 10:19:32.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-10 15:36:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8393f0e4f736dafb3ba7fb50cd2b679c24b1a01","subject":"Add \"jammy\" distribution release to the legacy ebtables installation","status":"MERGED","created":"2023-10-05 14:41:46.000000000","updated":"2023-10-10 15:37:47.000000000","submitted":"2023-10-10 15:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897438-bug/2038541","meta_rev_id":"19d5c2a2e80e19a42285032cddbeaefb6af63a9b","_number":897438,"virtual_id_number":897438,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-10 15:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-10 10:18:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-10-10 10:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-10-10 10:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-10-10 10:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-10 10:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 16:51:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-05 17:07:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 07:36:42.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-09 04:01:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-10-10 07:06:07.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"ccc59340a1a9f4b4ab87df66155ee7ef6eb81361","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 14:41:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1649c6b7a464c72fdd7cd40076574ee934c44cb0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 14:42:11.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"f640b75a866b167497db18bf324e54d8dd8f89e5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-05 16:51:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"107c5f6ee61d7e0575b789185793783b65cce509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:07:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9bf2ea4490542bd8096c30ca8200780\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/33b798bd5e1548dca7e5370eda53c393 : SUCCESS in 1h 57m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1c6662f3c8484e539a69196eb289f15d : SUCCESS in 1h 42m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18592c5fa2c141308a98b105251dcb70 : SUCCESS in 1h 44m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a491a1a5057c484aa39d5f225fe69326 : FAILURE in 2h 12m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf8c45ac7e0e49ab8340e26761cada67 : SUCCESS in 40m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382e40b0c09b49fbaa69ce597ca64b0a : SUCCESS in 11m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ece6e879a54454a7a7c45d3b7ede4a : SUCCESS in 13m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b475370d3d804dcda9d2bd5520dfa0d2 : SUCCESS in 1h 05m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/426493baf47b45e2b58932bcb0400b99 : SUCCESS in 2h 20m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/073f6835af2d4a489c099349a082ac0c : SUCCESS in 41m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/11e50e40a6994c24b57c167641a1b707 : SUCCESS in 55m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6a39a80730e74c58beb7894df9e23794 : SUCCESS in 1h 26m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d4ac9096f6b446918285ab3aca064a6b : SUCCESS in 1h 10m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/95e4cd9a7fe14777a178c503ab1696c5 : SUCCESS in 53m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2c82927600a144eea3b9a08a45f90668 : SUCCESS in 1h 49m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2b2463a916e4f1fa373c52584738500 : SUCCESS in 1h 31m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e51f98c47334f3dbc734c9c4b4e6ad4 : SUCCESS in 1h 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"98e7879aa54094764028c29fc9bcf746955b65bb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-05 17:49:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d37243c8bfce20b39155cb1af3d347f06bfc7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:25:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1517b444814ac781f7b48a44dc5256\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/72e4f5e6eca244f1b02ed3505f91ad94 : SUCCESS in 1h 12m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/66e64e8abe684fe0aba1ebd8bfe7aa77 : SUCCESS in 1h 29m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96ccb27ccbd54d26a7b4a6f248466dcf : SUCCESS in 1h 10m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4cb4a836423475599cd7da2e2e49b27 : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9509c43b40c49d5bababf677bca2280 : SUCCESS in 32m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e07eed7e8c4d4e9db87a0877434a88 : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cf70112c6447ae9383e77df575f952 : SUCCESS in 13m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8dec99d3e1b41d48d26a67b744d4aad : SUCCESS in 50m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e558683f33343cda14ebcfb13ecef95 : SUCCESS in 2h 28m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7a3ef10f7d145eea815e50f9b861521 : SUCCESS in 1h 00m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3e95808a177a412c87989db15abf386e : SUCCESS in 55m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ae2890607ff4b7ea13bdd593bd9074f : SUCCESS in 1h 29m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/210e90324613443abdeda0f9ae33d810 : SUCCESS in 1h 28m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c43a211ccff4c9887ae08952a1c5a20 : SUCCESS in 48m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/08f040c647a84614b32dcc9ebe69e562 : SUCCESS in 2h 00m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eece3f6ed01e4fed97cb62108c98a5bc : SUCCESS in 1h 22m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c7c0e61bfac4a398e8f62b423e0df84 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9f9d0b2491a8662e36e5a04a9a275c520afb91ba","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-06 07:36:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"534ed8ff2be286ab8dfb060ae7512b8fd381f943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 07:37:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02293d76449ed5824f89ee3685116af5f702b01b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 10:40:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2ff96c0d3149c49859f0280b8fc255\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e8465325c464febb0e8293486d8975d : FAILURE in 17m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6aeb206a03942689a4ad4ad43f83dd8 : SUCCESS in 2h 05m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aaf5468911a84d8ba64e314f9648d849 : SUCCESS in 1h 51m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4d5e118921a9494486cc781274e4cf82 : SUCCESS in 1h 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e381c91c5d4a4cd1ba21792ada747e3f : SUCCESS in 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3805f6cc0641a1a913a6c1bf28111f : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7147770108ea4ea78f2bbe8ddcc8399f : FAILURE in 48m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/db2dadf5a23047bb8a8f19cd7b3db1e0 : TIMED_OUT in 3h 02m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee5588847c7544a6b780a4543782c256 : SUCCESS in 1h 22m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/008a47c166f540c596351e2a3bf97829 : SUCCESS in 45m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b70f31409d5c47bc9694c1c36565aa53 : SUCCESS in 49m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5515d42ce43f4af5a37fb85d4f552010 : SUCCESS in 47m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b004be3f173a40cb9613e341df028477 : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7ba2586ddf47fe4ff5b931bb366d8ae708fa48e2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 14:25:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d99839eb846ece67b3a35b62ff24844e6b806d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:56:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0ab1680db094d89bb15098ddd65cd00\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4484090194a64e94be5102c24b447f0e : SUCCESS in 1h 30m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e0049f7f404b45f4bdb6e69cf18f9aa6 : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf997920b12c47c5874bb82dcec220b1 : SUCCESS in 1h 39m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/24dc6c38b7b24a57960c7138a316a9de : SUCCESS in 2h 16m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/88ce8df6dd844222b7562b504b9545ab : SUCCESS in 40m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385c8932eb6e48cfa1b179630e6e8a12 : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6b07e3ee094429a5149ba9b269b972 : SUCCESS in 12m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/701967f0fcf74527bb08b4c2166aa2ed : SUCCESS in 58m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/970b28c9906a468e9a63136aeeef94af : SUCCESS in 2h 26m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/796cc461527c4fccbb66b1c83ce4bb00 : SUCCESS in 1h 02m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9a333623bd2d4d029eafaaf014f75f63 : SUCCESS in 51m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10837596406a42b099fa489897fc2c82 : SUCCESS in 1h 37m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a3133e14dfd9412e9d715cebec20c119 : SUCCESS in 1h 15m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1abfbf26117b46e996f00adfa8e1caa4 : SUCCESS in 1h 07m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/95e7dd83357347b2931ace759032fcce : SUCCESS in 1h 55m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af6565e9e001449b96f7c62fa1d656f1 : SUCCESS in 2h 02m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ccb44508104e446783cd18d814c88cd7 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"230b2345a4836174fa61559e0185483243e59ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:57:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6f8a3ea160d1de48042a95a0bcfced7cca257c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 20:17:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee1ecf590727446a8751cd8294218c2f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d20320b4d9304adea467927d54b70a8a : SUCCESS in 1h 43m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ded27ebea7af43e898f1368e7ccebcd3 : SUCCESS in 1h 40m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3faeab53516244178d817f3482e43135 : SUCCESS in 1h 40m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c41ed81ad36b450189cc3be4b598adf7 : SUCCESS in 1h 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e17d28ee58c482eb51ab0518b0489f6 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adad56a45dc5498d93e4cae5aa015131 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/562d27155c2743969c6bdcc8fbf2a127 : FAILURE in 57m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/837bc2c4c0174d1096818781b9810fe4 : TIMED_OUT in 3h 02m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6371ad40ebee40a3914074450934acbe : SUCCESS in 1h 23m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ce3e51e8d294823abc403c1ab20fe1a : SUCCESS in 45m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e303a4131d5473e9bcd3b5e070a9845 : SUCCESS in 58m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f6840b6f71f44b5ab4526d8b35d7677b : SUCCESS in 45m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7ef80de20fd45ff924558b5e240030b : SUCCESS in 1h 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"46ad4728642f3d9dfdbd0b27b7a015f03178a22d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-07 09:22:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"371a5432ccf8bd99136607f7475f4e978011b514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-07 12:23:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae26609555f546fc85ee35e9d6b6326f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/737a23fb639140f187a5496ac48a7635 : TIMED_OUT in 2h 50m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c0c780b3420b4b4ba583ea387056d0ee : SUCCESS in 2h 00m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4775f80c1b454bdfa4dd1ce35414a2ec : SUCCESS in 1h 40m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d4e6c46d6ee48d8a0f7415b081a08a9 : SUCCESS in 1h 29m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02402bd5ab21466da6b4aba1b0eb7477 : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172871b73542476d9475ddc48e10c2da : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9413039ef0d476e8c78c5a594765601 : SUCCESS in 11m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/292b4bc5bbf7475baa87cd39c62dd404 : SUCCESS in 59m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68859ff9b4bd4e27b19db3c0fe298d8e : SUCCESS in 2h 18m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3934ab9ad103431d92a3a925a6d4206c : SUCCESS in 57m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/844636d710d940469a7cc7760b2ec872 : SUCCESS in 1h 02m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef8d14faddd94d5c96a1aee212b9f160 : SUCCESS in 1h 05m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bbab88ebbf4441cda2e9c8a019001f2c : SUCCESS in 54m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/904e685ce3d14c15ab6cbe039ce9a97f : SUCCESS in 48m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/06822ef96c82481d807ea66a5c4488a0 : SUCCESS in 1h 15m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/151466a1c8104dc9b89245daa4cf51fc : SUCCESS in 1h 22m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7cad9c1f47434d62a3a4b2c1d44c3f50 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":1},{"id":"26aa17470e73ac98bf47e8e5dd765896d1ebe43a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-09 04:01:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dd27f274556799323733d557af0f8d29a5edbc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 06:14:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12b5810fc3734ea68fc6bb7030b816ed\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/09c605b4fc5b480fa7cfece3c1bbe542 : SUCCESS in 1h 18m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ab793c674ca49748c2cd3f744aee0b7 : SUCCESS in 1h 25m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4d9e4a574b324d87bab1ebe87ce102f8 : SUCCESS in 1h 23m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/80e42b9adfc14da5b35239b74814b4d3 : SUCCESS in 2h 11m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/418886cf53f0456988b14c5315a8e8cc : SUCCESS in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcff48f813ba4ea1a1b3415affe9d91f : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295c88876f0c4f2db46da460d099c633 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a632fa668cc4ee485961d837dbf67f6 : SUCCESS in 58m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/37b29a08f0404fd692858677e3f3f7d5 : SUCCESS in 2h 07m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdb06806c2614bc0a3615852d01eedcc : SUCCESS in 41m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a1ec1c19ce64404bae82aed44bcb7f31 : SUCCESS in 1h 00m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d645f10747b6417ebf1de4c0634b008b : SUCCESS in 1h 13m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a508e9a6d8a847659cb28fff0e6aace7 : SUCCESS in 1h 28m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21b4bf22018f46cf8239af5e6c840d61 : SUCCESS in 1h 06m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de7501c33303427e90750a8712158d0b : SUCCESS in 2h 01m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/07727082dbe24cc5818b98a9fe9ac1b4 : SUCCESS in 1h 14m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1d41202f606344e2a32de8c9ab4e0616 : FAILURE in 53m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8deb110768ff898006954fde4f69d8cc317b3c6b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-09 06:17:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d025befa054ed32b59aab14b23fcf9f055d34699","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 08:03:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bfc426743fcbffdfcb24dfef3b8223ffd00a2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:21:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6ee608ce1d2c4137a053e267f389b3bc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d9bd82c30584487b0161f18563f511d : SUCCESS in 1h 43m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/681a2b32c9b64d1d82733ddec0da8bdd : TIMED_OUT in 2h 50m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/610cf3fb27854ec2a502c7d9043f71ae : SUCCESS in 1h 45m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6a761081a3643cd8d22cfeba8b771d3 : SUCCESS in 1h 55m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e9a1df3e4b64d00931263db364d661a : SUCCESS in 39m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a828656c8e5b4f2290f51408e1eb4edc : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b886ad92da4fb297e17b666142c3a1 : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/de80d6e9cea744feaf077a4c6103fa8f : SUCCESS in 49m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e6f277ebeb84b9796ec96d4270c0807 : TIMED_OUT in 3h 02m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ee90575eb2e4f03a1cc44d30e9c57cb : SUCCESS in 1h 00m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/972fea510e2a498fb1c694037f8444ea : SUCCESS in 1h 02m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fb2544d5f47412d94f57fa75ae83479 : SUCCESS in 1h 27m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f12c900fa25c45eeaa5dcd2538a06c71 : SUCCESS in 1h 27m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cfe705dc69c542948debb18c6d3088ea : SUCCESS in 56m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b77742d19cb247258006fec233e2b38f : SUCCESS in 1h 59m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22307f8608324d7c95c7e243cec2afbe : SUCCESS in 1h 21m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6926daf07214f2c8130e97120aa0b73 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":1},{"id":"53a118a3e35f8e94d8752e5c9d86a739281de448","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 09:47:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff867aede0a394805b895eca3796612fba7b3a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:50:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7dd570ee2024e0a89c80033fb457a6d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7502b280cf954561b93d9efa139bab1f : SUCCESS in 1h 27m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f75a443bb26a477383e1348dbd72b873 : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7f58508526e34092b516ff1e97297bf3 : TIMED_OUT in 2h 50m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/49e52bf0aa6244a5bc0bbee7d12d4b4e : SUCCESS in 1h 43m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f75fd0d35a7f4b6aada4d80201580f98 : SUCCESS in 43m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41db92965e1144cca0c4bd9c529dc03d : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee3d47b6d604721a67acbe2c1ba7acf : SUCCESS in 12m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/126979df133c44f8a4c04e186c63c9ca : SUCCESS in 1h 02m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/27f85c5f791c447db7ffdce74672f3b7 : TIMED_OUT in 3h 02m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a63f1df35fe4fcfa00928445e1064fc : SUCCESS in 48m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ad18b5977444e928eb4ce9a4a1a5422 : SUCCESS in 1h 04m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9fefabfaead8405e93e3afc969229cb7 : SUCCESS in 1h 28m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/302601e1644a4ed1b3345ca114074187 : SUCCESS in 1h 21m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0a86b7e95a75470289eee8978b53ac90 : SUCCESS in 58m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/203101f391ba4ae8a8dc3d03929137f3 : SUCCESS in 1h 50m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce2bb1158f93478baeba24bde8a29ca4 : SUCCESS in 1h 48m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8b7f0626f199439b9a8f134d2bf0044b : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a03acf1c4d915049062013e9cc8ab03634705eac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 14:08:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9301e8e4aac030ffa5708d9b6da7b6e8995fd5e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 16:58:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dac5b8c8147a426fb17c28c8a7abcb96\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/24ba8534b79748ff835ff1d82fa750db : SUCCESS in 1h 53m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6c896dc84d1340629928d9bbdaf25747 : SUCCESS in 1h 38m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/092e28b2fad84a56bbc75502cf3837da : SUCCESS in 1h 22m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/54cf3364e2bf4d0b9c65b6aacf75a044 : SUCCESS in 1h 36m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ca1f8d7e40a4604b26eae6bd01b7fd0 : SUCCESS in 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d40446eb6c944898f149ee728411c12 : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489f8166ddd441d5a076a6c00a14a370 : SUCCESS in 12m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc40ca69bed54f1295864ce5a36a48b3 : FAILURE in 42m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e50fc67e6a334391be12693a117b5451 : SUCCESS in 1h 59m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfd124dd37b44c5bbc1c41edab324e63 : SUCCESS in 59m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d443043d0a6e4d86beb4da51987969ef : SUCCESS in 57m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e6f31e6fafd442a99b93016103844d42 : SUCCESS in 1h 16m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5e8179c0d6c54280a3e836abb28f5452 : SUCCESS in 1h 25m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f49ce99617cf4313b0fbfc5cae541e0b : SUCCESS in 47m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/63f45fdab518413080bcdb3dda8e42d0 : SUCCESS in 1h 39m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/10072445a32c4d88a21335c585a8b7d1 : SUCCESS in 1h 47m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd2f84cf5a994f4ba8a4d859a13b92a2 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ac8fb1d1d0e68fca63fd07eb35f64148be6a0691","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 06:12:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f5b552e229bebb69f27f943152f290868c915f1","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-10 07:06:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66a2c576de8b55d1a56753f4adac850c5c95e056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 08:46:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/657a40439b504fdca4dd3ab1686be7cb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4116f5c54894ef5920752d41c6be5e2 : SUCCESS in 1h 43m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dd895d44fa76490e861e17473b642df6 : SUCCESS in 1h 51m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bfa4e9dc4ed1458e91d0f2473a14ab9f : SUCCESS in 1h 29m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bead182e41d4495cb8511bf69432a3fe : SUCCESS in 2h 00m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a527b56899454f64ae0b1bbcf73cf8e4 : SUCCESS in 43m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb928acddfd24a40b8a4efa63e1f8660 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da4ba1cdf4649cdb58dc4ec225cdd8d : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b8409ead8b9433fb037123d4f3f4c8a : SUCCESS in 42m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60754cc2f1b34c9785564f9e49d666a5 : SUCCESS in 2h 28m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab17810ce2b849f3b0bd3036212d6629 : SUCCESS in 58m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ac9868009dbb4d3fb84c9242e6cccb40 : SUCCESS in 56m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e8899612e5964c6a9b74a890386bbbbd : SUCCESS in 1h 12m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3915c9c8163444acb7692c6f74e99ad2 : SUCCESS in 1h 22m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eae71a04e7b14602960c63c6a6a73581 : SUCCESS in 57m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0e090cbd67a54e02868922130e5db551 : SUCCESS in 2h 07m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7cd31f9eb6d4e1bb9333aae48236966 : SUCCESS in 2h 03m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6afc53452e3040e6a015e52bb155e0be : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":1},{"id":"30e6ffb2996e257a3fa785f634e79bc2590490ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 08:46:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aab6e77833f7a17465bc4087856e20c8082cce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 10:02:43.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3973cbb06578376572d8d5f52c99768f9f4438ff","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-10 10:18:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"97e89ed401e6c7ba22683ee102d28234b0ed235c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 10:19:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"060e7cb22c67f71bb087184558f8a32061ac3b5a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-10 10:57:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63fbfa88506e8a482257d4133ff1deff05b70189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:42:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd8f47b10dd46069861614c41544a12\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bece5853a66d45b987c35665218bc06f : SUCCESS in 1h 48m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a8b9cf07f0aa47479cd419b7d7f513d2 : SUCCESS in 1h 59m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c455b7d016e41ba8e21dd4073f73ba5 : SUCCESS in 1h 38m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fa252d982c5245088c70721d0156dd58 : SUCCESS in 1h 52m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73b45027f6554681885d757f8f456523 : SUCCESS in 45m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7934d417d2473cae0ebb8dbda010e2 : SUCCESS in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93fd39dbe40a42c9b65e1c232b4d153c : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8690e8010c54346bf015e13da9d4d7c : SUCCESS in 1h 00m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/46f337487c364c46b78f9125081dba3c : SUCCESS in 2h 28m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c25ace64aa043a8acd13e752919709b : SUCCESS in 1h 01m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99ea00001292499295b823b6f52c9b73 : SUCCESS in 1h 04m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ad00a7d4a504ad3bf9c650765593bb6 : SUCCESS in 1h 38m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4f07f8f5aba24d5ea99d00150374c834 : SUCCESS in 1h 28m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9807f68b8d314c4c935721ba8836dac2 : SUCCESS in 1h 08m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b9749983c40c45afa535a4da7bfedaf4 : SUCCESS in 1h 32m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b5e14e2200e4102a644e6485f43c60e : SUCCESS in 1h 38m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0bdf3547bfba49f48a21bd0c1637abb5 : SUCCESS in 1h 08m 21s","accounts_in_message":[],"_revision_number":2},{"id":"af47f7c695b13d80450fc69b57e4b65392d14b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:43:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4c8720c0b266ba9ff3e283f3a120d9afbc04b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:36:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7491c3d1a1834b4bb94f9626965b92a2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a6d21e06f894810b3afe725a408c13b : SUCCESS in 1h 59m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5f225d99524c4f049597e5d67e1169c2 : SUCCESS in 1h 49m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/750e5101d9f94f42b840ee09ea1fec64 : SUCCESS in 1h 46m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/151984ec678d42488b76288323a76a8c : SUCCESS in 1h 47m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8dedb8415494c1f9324a0beeb303811 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0335c1f973d497c93e2883e970ecef7 : SUCCESS in 12m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9d6bc0facf3401db5a87dff12a7714f : SUCCESS in 1h 03m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8640add795e64ff880337d64412413a1 : SUCCESS in 2h 46m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8bf14c4ac3c14b80a786bf0625882efe : SUCCESS in 1h 31m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5162fa0454f041bcbfb0b5fd1e8b0d85 : SUCCESS in 1h 02m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa4bbb37b0264648881e0062e673f2f2 : SUCCESS in 1h 09m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bcdc73c64bed475e8e9b24c26a651c4f : SUCCESS in 46m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff82014d64224dd8836bbbda319f0b15 : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":2},{"id":"5f49aa21717efa58f49a7f470c21f4b3be5bbffb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:36:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"19d5c2a2e80e19a42285032cddbeaefb6af63a9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:37:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2885575150ea4619bd51abac9b9d0206\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/110f5f9f6a8d43348beda84363c919c0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4266dce979a8e168a82b0e27c338b9248bdb1c3f","revisions":{"53df16033c1faab9effd48884f247bed6fa1fdeb":{"kind":"REWORK","_number":1,"created":"2023-10-05 14:41:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/897438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/897438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/897438/1"}}},"commit":{"parents":[{"commit":"04e9b063ad526778ba80438fc851e03fa16e5085","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 10:49:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 14:41:17.000000000","tz":0},"subject":"Add \"jammy\" distribution release to the legacy ebtables installation","message":"Add \"jammy\" distribution release to the legacy ebtables installation\n\nChange-Id: Ia8393f0e4f736dafb3ba7fb50cd2b679c24b1a01\n(cherry picked from commit acc6ff05806b414e552525a7e6d758e253d6ea23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53df16033c1faab9effd48884f247bed6fa1fdeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53df16033c1faab9effd48884f247bed6fa1fdeb"}]},"branch":"refs/heads/stable/2023.2"},"4266dce979a8e168a82b0e27c338b9248bdb1c3f":{"kind":"REWORK","_number":2,"created":"2023-10-10 10:02:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/897438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/897438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/897438/2"}}},"commit":{"parents":[{"commit":"04e9b063ad526778ba80438fc851e03fa16e5085","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 10:49:23.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-10 10:02:07.000000000","tz":0},"subject":"Add \"jammy\" distribution release to the legacy ebtables installation","message":"Add \"jammy\" distribution release to the legacy ebtables installation\n\nRelated-Bug: #2038541\n\nChange-Id: Ia8393f0e4f736dafb3ba7fb50cd2b679c24b1a01\n(cherry picked from commit acc6ff05806b414e552525a7e6d758e253d6ea23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4266dce979a8e168a82b0e27c338b9248bdb1c3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4266dce979a8e168a82b0e27c338b9248bdb1c3f"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
