)]}'
{"id":"openstack%2Fneutron~897439","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Iea77d20bec330b692e3e8c9e38b3a62e2047b4f4","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2037239","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-10 15:41:33.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-10-10 15:41:33.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-10 15:41:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea77d20bec330b692e3e8c9e38b3a62e2047b4f4","subject":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"","status":"MERGED","created":"2023-10-05 14:42:07.000000000","updated":"2023-10-10 15:42:36.000000000","submitted":"2023-10-10 15:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897439-bug/2037239","meta_rev_id":"44cc1c1b75e7211222209b47ddb35789fdfc5ccd","_number":897439,"virtual_id_number":897439,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-10 15:41:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-10 10:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-10-10 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-10-10 10:19:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-10-10 10:19:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-10 10:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 17:19:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 07:37:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-06 13:11:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-09 04:15:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-10-09 04:15:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-09 04:15:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"7ce9f692e3157d9c2e1edcbbe49aa6143d65a4f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 14:42:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1baeb4ec4e4392a47e870d95f8fcf91a56b0c6af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-05 14:42:15.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"6ec2545ed15f646b32cae4916dc9e18b64e53b3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:19:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3d5d6b5031c4731b9353d8e02aa909f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8723bfe4e04c454289244de3eb1cf326 : SUCCESS in 1h 08m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87f81f92e3be4005a1d0588062d8bfe8 : SUCCESS in 1h 25m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87b29c43bcb04748a15ab95e87ed118e : FAILURE in 1h 20m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1cf78d479694466581a6be66e2ec8419 : SUCCESS in 1h 44m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42b97e14cd0b42779adc9fb1b5101639 : SUCCESS in 37m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8efeb70f3a041d8b61f2ebb2d8c751f : SUCCESS in 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c15c74045f4b2d8efdaecdf81ca7b0 : SUCCESS in 12m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf27a47fcc16431ba346b7c4ca7adbf6 : SUCCESS in 23m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79cf5144780b42da9b105bb4d63563e6 : SUCCESS in 23m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad163a9b2b584017a0906cfbd5032679 : SUCCESS in 28m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ecfda66f83c482c9f1ddbe21c4dcee4 : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e253001a59de45df8a5b9c42238202bd : SUCCESS in 10m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3c5c951b3b69446f85e6729edc44770b : SUCCESS in 3m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85ddcb17d75140cb95a3fd6713f5b193 : SUCCESS in 54m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ec62fcffbd0b49339be34919ebd67106 : SUCCESS in 2h 29m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e841bf4a41a4708b96fef9d5656c4e4 : SUCCESS in 52m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/33d4561636704db3bbde42d8364de81c : SUCCESS in 1h 04m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/096ffeaba7834483aa43574b642bff23 : SUCCESS in 1h 28m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8144e20584c74ae3845cfad2681761d8 : SUCCESS in 1h 11m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d307091cf6184194972045368e018b14 : SUCCESS in 55m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/18be1b7b1b8a477ca39893953a83e5de : SUCCESS in 2h 07m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/52a1d917d5164acda82505b3d6c35c73 : SUCCESS in 1h 48m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cc6d73dcc2d40e88e8c0de528c054b2 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9c773467b14b48c699d9f6a7d3599b892c309055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 18:40:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8510c0b047fd48659dc392e52ee3de3b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b27e3738c1546388b9a5cfcbea7f3a3 : SUCCESS in 51m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d4eee10e1a3497cb827408bd5759c2b : SUCCESS in 41m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75ea936ea9484c0f95332e6fc30183d9 : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fe1518058fa149e3bb7b50348e8e499b : SUCCESS in 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46c4c253ad433b175af71786152f668db4ee26eb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 07:34:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44ba6bc0104d643b59f0bc1d89172139e04fb910","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-06 07:37:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d47144a067344aa9b231a9555bee9e5c8833bf17","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-06 07:38:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bdb21227c0e64335cb2102c21af9252d45f21280","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 09:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a5fc5c050947019465f8ccfec2552d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec60715845404e56882e7a0fe11e5c99 : SUCCESS in 44m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f65714945984cd89e024061ff725d75 : SUCCESS in 45m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb19c1dd9f2d4f1b93161f349c2863d7 : SUCCESS in 46m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71875324e52948b0994c5f2f7e99bcbd : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98cb3a74c378bfbbc612b6d5c8a908ce943ccfab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 10:38:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccd8f9c01ed04baba5d07b132890fbea\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7dfc2bf84b142caaed2d0c86f5d4685 : SUCCESS in 1h 32m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/87a1b4ad83854a9e952d1ef6d30525ff : SUCCESS in 2h 12m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5b7d5e29993c4da89e6ad8a030e1e09e : SUCCESS in 1h 43m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2e97faa395be457f9d4758e456beca59 : SUCCESS in 1h 46m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce3204212ed640918b6cbd5ab51906c0 : SUCCESS in 34m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac6c2c303ac743c1907cbf93fca87d7a : SUCCESS in 36m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bea93cb815a47c0a8d2289ce040b879 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee809071dcb643218311478323bba170 : SUCCESS in 24m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/739ca6827f544b79a5c1e866cb5b60cc : SUCCESS in 27m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c07941a72bee40e38ee913c6731bf60b : SUCCESS in 27m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f9a826abc7a403082e46f0d4e3c4414 : SUCCESS in 28m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/022f6bb6393844bbb613dccb3e999264 : SUCCESS in 11m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6352f7aea88a4c61ae6f6887dcf4fd39 : SUCCESS in 2m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/28b819bedefd4e65b04c0425b40220f9 : SUCCESS in 1h 01m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/172059684c2e4531b63bd19deaa5013b : TIMED_OUT in 3h 02m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/251ff989026342c8918c1eb2f1f7264d : SUCCESS in 50m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e86b1b2fe9e14f9085cafac04a4a72ed : SUCCESS in 49m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/70f3fedd7a224dfca39239043342f20e : SUCCESS in 1h 24m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/94361b461d2f455ba0c81a6c4dc56f6c : SUCCESS in 1h 19m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aeca59524d4c45fab7fb40a1d67be28a : SUCCESS in 1h 01m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4308fb8bc6864d1496744c692be2df0c : SUCCESS in 1h 35m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3405a9c55ca14f518798abf2cda19f0c : SUCCESS in 1h 39m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/07e1ed780c314f468861e94be27cae7c : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e602d22bb3a5330194c560c2480e387b70595a19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-06 13:11:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cad4c6caf4bfe8635b049ac98041bd11aa83d86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 14:05:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850923a98ba94a869b5dc7bbd1dc5068\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70fa19adf62748929f589b78cfeb77d3 : SUCCESS in 52m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eebd30d78b6949848f7c1e632c0a624d : SUCCESS in 51m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39da7e3d0c604c9cb6c4fb6f40dd9085 : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cabddc3b945e42789375960c203bd8dd : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f2ce40bee5203b5c7f2a228473076c28e30ee1e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 14:25:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6282da9edba6a97d68293676c2908be924df19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad17e71130dc4e4dbc95ce802bba5393\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1cb4c47d6dd548edb7cac4dddd93652b : SUCCESS in 1h 21m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/60e15215a7ce41b8b8ad1ebaff21e5a6 : SUCCESS in 1h 51m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c6d0d3ffd06d4459a422e6ce31319d5a : SUCCESS in 1h 36m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2ed9c3790c854d1e8e03b62412bb335a : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a91cda1c224944cba580b16867538342 : SUCCESS in 44m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b07066cebd14a9eb3853f49e729f7d3 : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329e50c926514f8c84c17f2f196affd6 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04af2731a4324981b28935ab9980d01f : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61299f697f384567852974a6609dc40b : SUCCESS in 23m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75f99ce833c94909a0df77188c1fbdfe : SUCCESS in 26m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d76124d29ff4eb2af770dd833c6ae53 : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0468a98ccca4d058fb05f716f6ec188 : SUCCESS in 9m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7490b8e07aa349ee80668f7b9e42e768 : SUCCESS in 2m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0097c8ac06604081859b687fbf5a062d : SUCCESS in 58m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9223984e36f24b3d89baa5c8a97a1a18 : SUCCESS in 2h 47m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78e798ac9a294bf3a94ff24c641951d7 : SUCCESS in 1h 05m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0785f664319424f83844577c7a28245 : SUCCESS in 1h 01m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d8e6011b5eb6442a93c3156c4751de8a : SUCCESS in 1h 23m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4dd0025a1a94c31bbbb22b105d8b9f7 : SUCCESS in 1h 05m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8bdba2a902a04fe79f7672e3afb311f6 : SUCCESS in 51m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/49c8c35895384039990cc1db4a64d007 : SUCCESS in 1h 49m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9dcc32c34f6d4ecb9e9e106bfa3e5248 : SUCCESS in 1h 56m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ea4a327c4e78494e9f3d6a30b71eb611 : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5118c151ec876845d333923d74ddc3061edf1e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:08:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b980a97a574be2a687ed0eda237e56\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6213a1249bae4986b8a020ef0ab61d3b : SUCCESS in 53m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ce8a7edc17b488fafcbde9f682099bc : SUCCESS in 39m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e4be90be5a54ba1a3fe3c751ffce4c6 : SUCCESS in 52m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c977133685b4ad28bb71ebf9a4c34ae : SUCCESS in 48m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"725f22c1fec400f5dc2e24abb85a2316702ec29b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-09 04:15:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ea50314db87e6bd1b7cdb0f423f778474a0be09d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-10 07:07:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a38815d2e26ddb8633d4b010a23b641de3bc3266","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-10 07:13:32.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"15acf0805bd2b681b9cbb619977358591ef14eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 08:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cc39b1dc2a56de091379f7a458424c19f0847eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 10:02:49.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0a4815dae03b47b32a13e779deb9f5c34a144e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 10:03:00.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/897438 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"1833c2e293b4438cbe20f5b3311f210bd94dfede","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-10 10:18:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ee029ec56ba44afe9ae79ba82fe18d60b037c49d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 10:19:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d7e4f11d7ba38a59c381cb5537859875baacb3f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:54:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ecd671b771482290cf3aea950ce9e9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f3ed1bc75b64bcb921ed90c33110632 : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db0a66f8662a41e28992b4e1b61c213e : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c48ebc1ae22b4fb0a584781521cebc6f : SUCCESS in 44m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7d27a255f454499d96870d2d96ee5395 : SUCCESS in 57m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b7d7ef51221d6633b2a834354240080815e269bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:27:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe38220885d46269bc0cbfc87512598\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/95f745fa02ac41abbd8e678f52442609 : SUCCESS in 1h 37m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8a159f6e57b4c6aa812106dbd782e7e : SUCCESS in 1h 49m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba5746fd24bd485ab3c7a63a5bd6f91b : SUCCESS in 1h 20m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4ad37e9f64fe4919a07d7a503bd270be : SUCCESS in 2h 05m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f0a7f83ee7e4b2286da7471c4f2fcce : SUCCESS in 43m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af55d1bc019b4a3f84704c28634d1289 : SUCCESS in 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe1eeec27f84488b27ca4c6e28da11e : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bd9e08d1b4743cf8bd6e66586048405 : SUCCESS in 29m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35002740f2d24f0a9e07fb37dc760d62 : SUCCESS in 25m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61698432d56c42a898cf3c32ef99067b : SUCCESS in 25m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6215d2be468749d893abd837322130c6 : SUCCESS in 19m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc624e877a24208a2d7687cc141f2fe : SUCCESS in 10m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b7bdcb3f9b64475faee59cf8b9993189 : SUCCESS in 3m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2cec3b84afb41be8a8da94520f686d3 : SUCCESS in 1h 00m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/01ed9a09675e42c7adc0388c4e10be55 : SUCCESS in 2h 16m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da4e1c724d2741f0bfdb71a49c10e82b : SUCCESS in 44m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8e1cd6175b6b43f09692c3f946af0ed1 : SUCCESS in 1h 05m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6d67a1a853a54d84bee11204b36f8f16 : SUCCESS in 1h 34m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/541fd055e9574a2895db0df0d2fe5190 : SUCCESS in 1h 24m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/792e0f649172407ea16306d037614fb6 : SUCCESS in 48m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fd579cd9d95d4e0aba124012ad0e94e3 : SUCCESS in 1h 27m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1b564bbf1f144d98daabd1409c28ac6 : SUCCESS in 1h 38m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11e1b379ff8f476ab935552d6713fdd1 : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1b1fa24c24dc7b67bdea1828ad7ad83a0702b4de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:43:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"290fb68440c654da2b305937f6fe42025b377e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:41:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eccbf876ebb1404f9d0d4201671e8619\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e16bc84ca0bf4508b0e3db49da3206d8 : SUCCESS in 1h 33m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/580c7f91e6b94bb69472a24cdd193c9f : SUCCESS in 1h 58m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/661dd3dd94694987a8d8184901e11176 : SUCCESS in 2h 13m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0164456fcc5747cd8663ef5afc389c8a : SUCCESS in 1h 47m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc09049b1d2472aa957a54281f37646 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b24e183e43e94e5e91ee71c5630b6051 : SUCCESS in 28m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124bccd43e914c6a8cad64c9c87bb356 : SUCCESS in 24m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a51f7ae229b4e3aa2568052f40d4d15 : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355a232f58a14b6c91834c2a5ea960d9 : SUCCESS in 14m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2dd470c2de5b477196624e641b1b0364 : SUCCESS in 3m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c06740726fd429190db051d878b61c9 : SUCCESS in 59m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4ee9810242246a88f72b408b220573d : SUCCESS in 2h 52m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5de70fbf9fc45dcbe4a97cb9ea64419 : SUCCESS in 1h 36m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/539d7548c831455281972187d36be00e : SUCCESS in 1h 03m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45deaf29c9fa4c56aa2c7b04ccc652ce : SUCCESS in 1h 11m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c038523bd654b2bb07077a96a80362c : SUCCESS in 1h 04m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ee2b8a99e65428c9a9f580286dd6d04 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4ede0aa72e22474005e76f635a121b0ed6fc67d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:41:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44cc1c1b75e7211222209b47ddb35789fdfc5ccd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:42:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1b4b0ec10d4e01bd8dd5b9caafc838\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42db75de4dcf404b82c213355b123af7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de121943ee8edcdbf8c4e8c0212f89f3e2526fd5","revisions":{"01a929c4cb9c20d0c9fecd03e2d4d92877b74c96":{"kind":"REWORK","_number":1,"created":"2023-10-05 14:42:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/897439/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/897439/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/897439/1"}}},"commit":{"parents":[{"commit":"53df16033c1faab9effd48884f247bed6fa1fdeb","subject":"Add \"jammy\" distribution release to the legacy ebtables installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53df16033c1faab9effd48884f247bed6fa1fdeb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-03 21:20:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 14:41:54.000000000","tz":0},"subject":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"","message":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"\n\nSince [1], present in oslo.log 5.3.0, the ``log.setup`` method is\nunpatching the evenlet thread module. That is causing several problems\nin some Neutron services, in particular the keepalived-state-change\nservice.\n\nWithin this oslo.log version, the patch [2] is provided to call this\nmethod without unpatching any eventlet module.\n\nThis patch is also bumping the minimum required version of oslo.log\nto 5.3.0, in order to call the ``log.setup`` method with the kwarg\n\"fix_eventlet\u003dFalse\".\n\n[1]https://review.opendev.org/c/openstack/oslo.log/+/852443\n[2]I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76\n\nCloses-Bug: #2037239\n\nChange-Id: Iea77d20bec330b692e3e8c9e38b3a62e2047b4f4\n(cherry picked from commit 0aa154b5ce9dc8da73309fb212843a2b69b68696)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01a929c4cb9c20d0c9fecd03e2d4d92877b74c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01a929c4cb9c20d0c9fecd03e2d4d92877b74c96"}]},"branch":"refs/heads/stable/2023.2"},"de121943ee8edcdbf8c4e8c0212f89f3e2526fd5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-10 10:02:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/897439/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/897439/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/897439/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/897439/2"}}},"commit":{"parents":[{"commit":"4266dce979a8e168a82b0e27c338b9248bdb1c3f","subject":"Add \"jammy\" distribution release to the legacy ebtables installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4266dce979a8e168a82b0e27c338b9248bdb1c3f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-03 21:20:34.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-10 10:02:49.000000000","tz":0},"subject":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"","message":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"\n\nSince [1], present in oslo.log 5.3.0, the ``log.setup`` method is\nunpatching the evenlet thread module. That is causing several problems\nin some Neutron services, in particular the keepalived-state-change\nservice.\n\nWithin this oslo.log version, the patch [2] is provided to call this\nmethod without unpatching any eventlet module.\n\nThis patch is also bumping the minimum required version of oslo.log\nto 5.3.0, in order to call the ``log.setup`` method with the kwarg\n\"fix_eventlet\u003dFalse\".\n\n[1]https://review.opendev.org/c/openstack/oslo.log/+/852443\n[2]I4bbcfe7db6d75188e61b9084cb02b2dd2aaa0c76\n\nCloses-Bug: #2037239\n\nChange-Id: Iea77d20bec330b692e3e8c9e38b3a62e2047b4f4\n(cherry picked from commit 0aa154b5ce9dc8da73309fb212843a2b69b68696)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de121943ee8edcdbf8c4e8c0212f89f3e2526fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de121943ee8edcdbf8c4e8c0212f89f3e2526fd5"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
