)]}'
{"id":"openstack%2Fneutron~897464","triplet_id":"openstack%2Fneutron~master~I5d954a725d5f427898bc23d6adf10ee022fc621f","project":"openstack/neutron","branch":"master","topic":"fix-dhcp-agent-router-port-deletion","attention_set":{},"removed_from_attention_set":{"36276":{"account":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"last_update":"2023-12-04 11:03:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5d954a725d5f427898bc23d6adf10ee022fc621f","subject":"Fix router port deletion issue in DHCP agent","status":"ABANDONED","created":"2023-10-05 14:56:27.000000000","updated":"2023-12-04 11:03:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7e161c898fad3456603f513778d8efac46b78911","_number":897464,"virtual_id_number":897464,"owner":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-05 18:22:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-05 19:06:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2023-10-06 15:26:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 16:36:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 19:06:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 20:36:45.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-06 15:26:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-12-04 11:03:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"19a3c71efb924ef5b7fa75344e25356c5b57a0b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"date":"2023-10-05 14:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c5bb84f58ae9b75815bede2c51b3df4f48f0710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"date":"2023-10-05 15:06:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8d6c6d46e86f277d23fd7d31599f4ee56f8ac551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"date":"2023-10-05 16:02:34.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4e2638a4512638d2b6525f5dcf475d304b14e167","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 16:36:08.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/64/897464/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c6a109fb2889adfadaeb26ed27284ebff10b45d6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 16:57:36.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/64/897464/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7ae939931849c57b06c579441eb76481e8c5417c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 18:22:04.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/64/897464/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"da6a706f38db3d63cc754011cfe97566caa39c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 19:06:18.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e090b6e8525f429eb43a040253cf7905\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/645866efe3d34441bf01ea36850d6b04 : SUCCESS in 1h 16m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/371451621c854f94af12b187ffdda2c6 : SUCCESS in 1h 29m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da6ab7854a8249a081da5881eaaf6e4e : SUCCESS in 1h 39m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2a3893c3f6c4d13a0fe0a777a6bb8dc : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d031cdff6e245009930d0c3da5abf66 : FAILURE in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b92e8d1b087425aa76c45f1e8ceed0b : SUCCESS in 33m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b1318848104451b16a33af06db7964 : SUCCESS in 25m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cfb015174e64dcb88c1a98b3f44c721 : SUCCESS in 30m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a823b63f658942e98c834c096ac8ee9b : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc64ea830864c75b2f4035728ca43aa : SUCCESS in 12m 52s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/27ed7c7007fa4678bd2b31fdc9a16509 : SUCCESS in 29m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5752a4e7cdbd4ca2869dea3616cb71f9 : FAILURE in 55m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f7465f336d84188a50dfddf04ac5fbb : TIMED_OUT in 3h 02m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db601dbca1994ce9a6e453f7a4ce8ada : SUCCESS in 1h 07m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46354c2c12c9416fb3444b298083e256 : SUCCESS in 57m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/30d7ffa596624b8b8365925a2e9a2024 : SUCCESS in 1h 17m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3c70fa2fbac645b2b1dbc34b7a733d79 : FAILURE in 1h 09m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ee32d10c010e44d89787bfa1ae3888be : SUCCESS in 1h 51m 04s\n\nWarning:\n  Comments left for invalid file neutron/agent/common/resource_processing_queue.py","accounts_in_message":[],"_revision_number":3},{"id":"cda311ba0a5ab4fa6e6bd13599ee937d44b96a62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 21:06:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e10902e1fb24365afc80eb6383a656c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeb77ee5810b4b12b48fc0456118ca95 : SUCCESS in 42m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d323bf708284df2a3ce52ca77bd8db0 : SUCCESS in 1h 08m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f393f5aa20414000ab4f6f6b52e9544f : SUCCESS in 1h 15m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d30ca580c746436a887c8303f9878a05 : SUCCESS in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c7d06dbad5c40d996fb791325dbe0ac7a75c26a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 15:26:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e161c898fad3456603f513778d8efac46b78911","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:03:39.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"141ee7e4ceeda426382896dc0fbbecde90ba9900","revisions":{"f1c397f60c842bfc3af5b790aa87fea8714b4401":{"kind":"REWORK","_number":1,"created":"2023-10-05 14:56:27.000000000","uploader":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"ref":"refs/changes/64/897464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/897464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/897464/1"}}},"commit":{"parents":[{"commit":"b58d36a6a4c8aff1a494e3dc0eceec05fe161847","subject":"Adds protection in agent worker for failed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b58d36a6a4c8aff1a494e3dc0eceec05fe161847"}]}],"author":{"name":"grishatsukerman","email":"grisha.tsukerman@zadara.com","date":"2023-10-05 14:25:57.000000000","tz":180},"committer":{"name":"Grishatsu","email":"grishatsuker@gmail.com","date":"2023-10-05 14:56:07.000000000","tz":180},"subject":"Fix router port deletion issue in DHCP agent","message":"Fix router port deletion issue in DHCP agent\n\nUpon router deletion, it triggers port deletion in DHCP agent but it fails to access \"fixed_ips\" key inside the object and therefor fails\n\nCloses-bug: #2037876\nChange-Id: I5d954a725d5f427898bc23d6adf10ee022fc621f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1c397f60c842bfc3af5b790aa87fea8714b4401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1c397f60c842bfc3af5b790aa87fea8714b4401"}]},"branch":"refs/heads/master"},"3400ccb6d16c4aab1613baf6ac20b3f9ed7294e8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-05 15:06:45.000000000","uploader":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"ref":"refs/changes/64/897464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/897464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/897464/2"}}},"commit":{"parents":[{"commit":"3e57d709c5d76386f3f8a490ac4799c7ce0b9718","subject":"Adds protection in agent worker for failed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e57d709c5d76386f3f8a490ac4799c7ce0b9718"}]}],"author":{"name":"grishatsukerman","email":"grisha.tsukerman@zadara.com","date":"2023-10-05 14:25:57.000000000","tz":180},"committer":{"name":"Grishatsu","email":"grishatsuker@gmail.com","date":"2023-10-05 15:06:33.000000000","tz":180},"subject":"Fix router port deletion issue in DHCP agent","message":"Fix router port deletion issue in DHCP agent\n\nUpon router deletion, it triggers port deletion in DHCP agent but it fails to access \"fixed_ips\" key inside the object and therefor fails\n\nCloses-bug: #2037876\nChange-Id: I5d954a725d5f427898bc23d6adf10ee022fc621f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3400ccb6d16c4aab1613baf6ac20b3f9ed7294e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3400ccb6d16c4aab1613baf6ac20b3f9ed7294e8"}]},"branch":"refs/heads/master"},"141ee7e4ceeda426382896dc0fbbecde90ba9900":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-05 16:02:34.000000000","uploader":{"_account_id":36276,"name":"Grisha Tsukerman","display_name":"grishatsu","email":"grishatsuker@gmail.com","username":"grishatsu","status":"Zadara"},"ref":"refs/changes/64/897464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/897464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/897464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/897464/3"}}},"commit":{"parents":[{"commit":"898e92375ce401eec0cc52a3e95dcb7abb2b5010","subject":"Adds protection in agent worker for failed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/898e92375ce401eec0cc52a3e95dcb7abb2b5010"}]}],"author":{"name":"grishatsukerman","email":"grisha.tsukerman@zadara.com","date":"2023-10-05 14:25:57.000000000","tz":180},"committer":{"name":"Grishatsu","email":"grishatsuker@gmail.com","date":"2023-10-05 16:01:46.000000000","tz":180},"subject":"Fix router port deletion issue in DHCP agent","message":"Fix router port deletion issue in DHCP agent\n\nUpon router deletion, it triggers port deletion in DHCP agent but it fails to access \"fixed_ips\" key inside the object and therefor fails\n\nCloses-bug: #2037876\nChange-Id: I5d954a725d5f427898bc23d6adf10ee022fc621f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141ee7e4ceeda426382896dc0fbbecde90ba9900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/141ee7e4ceeda426382896dc0fbbecde90ba9900"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
