)]}'
{"id":"openstack%2Fneutron~897528","triplet_id":"openstack%2Fneutron~master~I09a37ca451d44607b7dde344c93ace060c7bda01","project":"openstack/neutron","branch":"master","topic":"dhcp-scheduler-with-ovn","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-13 08:51:34.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-13 05:44:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-13 11:17:42.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-13 11:17:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09a37ca451d44607b7dde344c93ace060c7bda01","subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","status":"MERGED","created":"2023-10-06 09:02:23.000000000","updated":"2023-10-13 11:18:57.000000000","submitted":"2023-10-13 11:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897528-dhcp-scheduler-with-ovn","meta_rev_id":"93d2fe7728be661b9314365b2caf3c5ec5a740f9","_number":897528,"virtual_id_number":897528,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 11:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-13 08:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-13 05:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-10-13 06:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-13 06:04:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-10-13 08:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 09:13:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-06 09:14:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-10-06 09:45:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-06 10:11:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-06 11:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-13 06:04:31.000000000","updated_by":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"reviewer":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"state":"REVIEWER"},{"updated":"2023-10-13 06:14:41.000000000","updated_by":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"reviewer":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"state":"REVIEWER"}],"messages":[{"id":"157391de0de3833f0796caf0aad7c750ec806051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-06 09:02:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce9b2c369552ae354032a57900a207159f7c5036","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 09:13:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6cf528313e08ec73f3bcf0d541678197ab4d14a2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-06 09:14:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83f4d502cb946bda28a68512b46108ff52a99028","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-06 09:45:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a266e1fc2919c5e74ab28d40a2c811b7f24c254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 10:11:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f23e7a57aa9d492b9e4113f4cf9b710a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8fd8be1e81a4faf9aab5c3b43073759 : SUCCESS in 41m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec35de5b9a3b4c28bf03281eb967e631 : SUCCESS in 48m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/028398d79e02490fbfb6a9bb18eb22fd : SUCCESS in 39m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b031c5982694ec995dd0110818d4131 : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e91afd52241bfe72a73bdc97fafc06b9b0b87a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 10:53:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"582450a3f68414147701d326f902e2f901cbd736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 11:08:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be19ef6f5caa4749b749cbe95b7ed511\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac5189079233472383e094d57188ecb1 : SUCCESS in 1h 30m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eaec21d7c74247c9b6c79a6bb27dca3d : SUCCESS in 32m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/635b5dcc14f041e6ba2a6086a1716eaa : SUCCESS in 41m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ceaf067ea44f5aa5320d3ce64d077e : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f16d6d1bd9248718024929d47885b59 : SUCCESS in 26m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83f344966a204c879627b0d3494c9c60 : SUCCESS in 25m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c418608619134a22abea958d1a426dc2 : SUCCESS in 26m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aabdd87d5e494b71906d51450f47c721 : SUCCESS in 29m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4075f6b8594e9094966c0ff4db4cff : SUCCESS in 13m 48s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1897f7d8893041edacd768312b2901bd : SUCCESS in 24m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13c8bc10a3084dd096717ac90b157490 : SUCCESS in 1h 09m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d7a87199fdf418bb91bb26b18d43b50 : SUCCESS in 2h 04m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a00bd2b2c8494b83a752d8bdedd7aed2 : FAILURE in 53m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b1ba7cb56ea744e2b8e469470e7c6390 : SUCCESS in 1h 57m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5eb27fb61a544803819abe818be3333a : FAILURE in 59m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c07af01ceefc1eb0eb705dff16779bb6d7af66a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 10:08:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b6951b4b51415b3aea096a449cf3c6fbf5aef91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 11:36:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c23fc291a684b3eaadd09ffd7f8a3ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b9cb4e851b348a8877f736d782e715e : SUCCESS in 38m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d01044a914dd4cf7949173a6cd28e462 : SUCCESS in 1h 10m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5d1c625ee51d4704a7129ff49a7e6fd1 : SUCCESS in 1h 08m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/efd11661782f406988da283096737774 : SUCCESS in 59m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11173756b4670f182a7403a748f83b02d43abe01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 12:20:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a47419c28d34563acd1414262433598\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/692aa154ce024419a90c8402b60c7ebd : SUCCESS in 1h 56m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/97fa9f6488c14883adf8287109823887 : SUCCESS in 39m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7668fc797c7f43998abbb341ea4689ec : SUCCESS in 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b2af76c8e54704912cebdcdca2ac96 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53e99a88e95348829a1fefaef3112cb3 : SUCCESS in 27m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0783fa949ff5410f92e886353d61d987 : SUCCESS in 27m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6360ba50b7c494d932ea1b195bfca78 : SUCCESS in 25m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e798f836d6bf4b91841f7efd92404792 : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce079986e6f4c76923885f884431820 : SUCCESS in 10m 36s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2845eecdbacf4f8188830be5e7abe23f : SUCCESS in 29m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b917e2bb39834be188ba92a960e71db3 : FAILURE in 42m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b520b6ed0fa34fd5a7673dc8689d54a2 : SUCCESS in 2h 07m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/297497bdf3ad4e53997043957394aeb5 : FAILURE in 39m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a98ef44231a4c7fa10c8a5b39a869fe : SUCCESS in 1h 38m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45236eab373f4f1a8d951881bcff1be2 : FAILURE in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f07c0fdbb4c017f357241ef25f1e7472bb5b4906","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-09 12:52:01.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9b065ed24704732c2ee4c9765f98982132bfd962","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 13:25:41.000000000","message":"Patch Set 1: -Code-Review\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"0986bbdde160debd132fd700c507da89bc9b00b3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-10 13:25:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8ac066981c50c2a1eea8c9617f7eedbf431211ce","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-12 08:21:25.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fac079d97321ca1918d433043a1c50882094b6c6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-12 08:21:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c72f8099389b916fd8bdc662980a8806a2e0955e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 09:36:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18758087e4fc4cfb91033388907e7a5d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c1428d1d6bb4de79a6ce5a7cc949a77 : SUCCESS in 1h 12m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3c7c1992ce8422f85d1628f985ff616 : SUCCESS in 1h 10m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/879a012eb9fb41b8aeff37e8db7075b9 : SUCCESS in 1h 09m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/846cd1af645b445eb796df17b5b3691a : SUCCESS in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fd91033961577c05981e0dc092d063a36fc66d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 10:42:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd08b37bdd8f4d88b6036d0858417c40\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4a9eb8578c3a456f8d160118f1710f03 : SUCCESS in 2h 02m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acc58a7605f94a22a9427832d1faf1b9 : SUCCESS in 46m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15b17ea8b7114cfbacac11b5355a9471 : SUCCESS in 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62b829c26324ae6a292734e211f85ef : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b67ab79046f14bff820c530ed7d8fe82 : SUCCESS in 29m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4da98379aba484b886fa3abb2f4b266 : SUCCESS in 24m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8deba2f6b8249dca4a7384bf24468c1 : SUCCESS in 25m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a58e6bcc24f409f828242073e10b5eb : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2ece17b110473bb077ac7f9329af0e : SUCCESS in 10m 58s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6bca0fc759bd467b9657db9cd5bb0bcc : SUCCESS in 38m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0caffa34fe34a85aad92d7d10680a09 : SUCCESS in 46m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e62b9d8f4d3c425cb72921fb25f846b6 : SUCCESS in 2h 18m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/af32e2c3f8c1487f844fe528e1efc034 : SUCCESS in 1h 00m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c1d7d77383843a9b69fd6222165f95b : SUCCESS in 1h 51m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b8debd9bff994842b21776795f677050 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"6569b9f503f73f2efe234b4583f6a33336a5c929","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-13 05:44:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0b31ddf69887ad86001ca1b124cc2385f17e380a","author":{"_account_id":36348,"name":"Rahul Pawar","display_name":"Rahul Pawar","email":"rahul.p@ibm.com","username":"rahulp"},"date":"2023-10-13 06:04:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63359435510d5ef4c846db4f3d4ceeeafd9ca5b5","author":{"_account_id":36439,"name":"Vivek Pandey","email":"vivek.pandey11@ibm.com","username":"vivek.pandey11"},"date":"2023-10-13 06:14:41.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7e4e5076bb18890ea8b2bc13e06c031c4538466","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-13 08:51:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7d36b1be2ac4d99daf69ea66dd9caf2b7ca41ca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 08:52:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab6b34a048842174e27b2eda0ed42d4eb5b7f5c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 11:17:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80e890459bb44ecaab80f1cccef9a24f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee378299f1ec4d02aa4961a59b53f6e7 : SUCCESS in 2h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dffcff8f61d546bcb4c57f6a533cf95c : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94928d60a45f4b24a11ce953c3256a9e : SUCCESS in 26m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3fde68eb1b44525a0567ac4df5311dd : SUCCESS in 21m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccbefccf61b24fa2b94e9a1b24cff662 : SUCCESS in 31m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa880cb15c749e381e9096e45c5a0f4 : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95f2e4fd44c54461866de0370a306a89 : SUCCESS in 1h 01m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1683323958224270a8dbc5afc4ea5f91 : SUCCESS in 2h 20m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b9136de18e54f669d2e54c56423b3c6 : SUCCESS in 1h 04m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ea16dc9dc27a4dcd96dcc79e528b2400 : SUCCESS in 1h 15m 40s","accounts_in_message":[],"_revision_number":2},{"id":"feefc5599673fea475923fff166a8ae2f3b9d2b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 11:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"93d2fe7728be661b9314365b2caf3c5ec5a740f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 11:18:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d224a1faa9f1458f842556d37409e8b6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a2bf5344704f55a5634e50091ec91c : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f006d29251abe3c138ee0dea5b549496b66b8fa7","revisions":{"4cc7adbf5e639345fe5b1ae3b56cbab3091b5af9":{"kind":"REWORK","_number":1,"created":"2023-10-06 09:02:23.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/28/897528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/897528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/897528/1"}}},"commit":{"parents":[{"commit":"8cba97016e421e4b01b96de70b4b194972d0186f","subject":"Merge \"[UT] Add chassis.external_ids dict in ``TestOvnSbIdlNotifyHandler``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba97016e421e4b01b96de70b4b194972d0186f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-06 08:58:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-06 08:58:57.000000000","tz":120},"subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","message":"Add dhcpagentscheduler API extension to the ML2/OVN extensions\n\nIn most typical use cases ML2/OVN backend don\u0027t needs to run DHCP agent\nas OVN provides DHCP functionality natively. But there are some use\ncases like Baremetal provisioning over IPv6 or Spine Leaf architecture\nwith DHCP relays where DHCP agent is necessary and it can work perfectly\nfine with ML2/OVN backend.\nThe problem was that dhcpagentscheduler API extension wasn\u0027t listed as\nsupported by the OVN backend so it was filtered out from the list of\nsupported extensions during start of the neutron server. This caused\nproblems with API to get/set/delete networks to/from DHCP agent.\n\nThis patch adds this API extension to the list of the extensions\nsupported by the OVN driver to fix that issue.\n\nChange-Id: I09a37ca451d44607b7dde344c93ace060c7bda01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cc7adbf5e639345fe5b1ae3b56cbab3091b5af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cc7adbf5e639345fe5b1ae3b56cbab3091b5af9"}]},"branch":"refs/heads/master"},"f006d29251abe3c138ee0dea5b549496b66b8fa7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-12 08:21:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/28/897528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/897528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/897528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/897528/2"}}},"commit":{"parents":[{"commit":"8cba97016e421e4b01b96de70b4b194972d0186f","subject":"Merge \"[UT] Add chassis.external_ids dict in ``TestOvnSbIdlNotifyHandler``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba97016e421e4b01b96de70b4b194972d0186f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-06 08:58:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-12 08:20:29.000000000","tz":120},"subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","message":"Add dhcpagentscheduler API extension to the ML2/OVN extensions\n\nIn most typical use cases ML2/OVN backend don\u0027t needs to run DHCP agent\nas OVN provides DHCP functionality natively. But there are some use\ncases like Baremetal provisioning over IPv6 or Spine Leaf architecture\nwith DHCP relays where DHCP agent is necessary and it can work perfectly\nfine with ML2/OVN backend.\nThe problem was that dhcpagentscheduler API extension wasn\u0027t listed as\nsupported by the OVN backend so it was filtered out from the list of\nsupported extensions during start of the neutron server. This caused\nproblems with API to get/set/delete networks to/from DHCP agent.\n\nThis patch adds this API extension to the list of the extensions\nsupported by the OVN driver to fix that issue.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/898090\n\nCloses-bug: #2038655\n\nChange-Id: I09a37ca451d44607b7dde344c93ace060c7bda01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f006d29251abe3c138ee0dea5b549496b66b8fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f006d29251abe3c138ee0dea5b549496b66b8fa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
