)]}'
{"id":"openstack%2Fneutron~897540","triplet_id":"openstack%2Fneutron~master~Iae2e3a31eb65d68dc0d3d0f9dd9fc8cf83260769","project":"openstack/neutron","branch":"master","topic":"bug/2038646","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-11 04:32:33.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-10-11 04:32:33.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-10 14:37:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-11 04:32:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae2e3a31eb65d68dc0d3d0f9dd9fc8cf83260769","subject":"[RBAC] Update the subnet policies","status":"MERGED","created":"2023-10-06 10:50:57.000000000","updated":"2023-10-11 04:33:55.000000000","submitted":"2023-10-11 04:32:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897540-bug/2038646","meta_rev_id":"d02e148bb5ce3be6a48f1485ca66607ace90f71b","_number":897540,"virtual_id_number":897540,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 04:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-10-10 14:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-11 02:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-10-11 02:09:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 12:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-06 13:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 14:23:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-10-06 14:23:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-06 14:23:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-11 02:09:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"94b7623028618902b8dfd21c1d38e325652b058d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 10:50:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42be5e88d87be65dbbd6df3b30c0b2d5379fd124","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 12:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d9328c6a88a48cbbf676549057e5b1f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef1e802d28c3457caf35ccc06fbeba7d : SUCCESS in 40m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa8414fba0054fc8b60490ea271e4c62 : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea16d0e365264f5f8eeedcae7010a2ec : SUCCESS in 51m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0ac918eb52b43d6b1d7fbe79f33991f : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57570e0efef50c9a51db38096c264c098fe1bc12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 13:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066c08d5812541fdba0a879af0a7ff06\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/829d26440197404dbfb03d56c0b14965 : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77119ed58e684fc6a29e188de10a9f25 : SUCCESS in 1h 32m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ebeb145e406443f992085fee79584461 : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9da8fff806814503bddf498d4996ac25 : SUCCESS in 1h 53m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/499351c35b834355b89b4e9fe4486948 : SUCCESS in 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a13b2b0efd541d080a306c23e3cb4b2 : SUCCESS in 34m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec19e12a0f8948ffafa2ce31ac4e5cd3 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03275776674140f1a6973f9f4beb348d : SUCCESS in 27m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd7690fa807d47cc8ac0a31f0de06bdc : SUCCESS in 20m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d910abe866645c2a3371fedbc98c9b3 : SUCCESS in 27m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/403b95e1d0a54f6790a5033d7e2d727d : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d3acbab6ee4bf7961a697f23c28bca : SUCCESS in 11m 37s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3a2d2c0de39c48dc8b57ee48d563d552 : SUCCESS in 29m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c58333b546dc48d986c91479700605e3 : SUCCESS in 1h 01m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/27c73ea3edec4bc493eac43cdb36e3f1 : SUCCESS in 2h 16m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/531609dd98af44d48f023086a8d58da1 : SUCCESS in 55m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9737fd13c1fa4012ace63e8204dcb4bf : SUCCESS in 52m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b317aab0294746ae80b971b9d74476fe : SUCCESS in 1h 29m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9416c9d518034227bed7876cb43ca1ba : SUCCESS in 1h 33m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8791db6aab1446e4bb7136b21ad5e27f : SUCCESS in 1h 01m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/25669c3b15f6424c9140c3403312b4a8 : SUCCESS in 1h 46m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6bbac7c244e4de9958bdbb4cc610e80 : SUCCESS in 2h 05m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ca57041a5c694340989588ce019d5918 : SUCCESS in 1h 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c4853732bfb0e6855d67210fa246fd923069c410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 08:04:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd5d7996333f83df0c92cf48ee6a0fde74a3a7fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:16:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d9c0b2065c46b1afa0ab7601468df4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa4da04cfedf4292a5987ca84ca8ac06 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dbcb367a9504a12b792dcbc41a2216a : SUCCESS in 39m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1182e86321ea4264a29359f468616d6e : SUCCESS in 43m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c1c0fe072add4137a2001463957a335f : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"26d3f10320ab7e46adbd21c6e03053b503af5576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:49:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82cf93b58b454272bc1195a5775e7167\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/58bcc9734f9947ddb3f43b7a36aadcbd : SUCCESS in 1h 15m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7ce360bdbd07498bb10b415a006b4475 : SUCCESS in 2h 10m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43530b3feb8344c08324f63f5d7c2642 : SUCCESS in 1h 28m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ace028d7c56c453c9311e987917d5ed4 : SUCCESS in 1h 20m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4de1a451153b4bda93b37fa2f7aa4050 : SUCCESS in 45m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2418b1df56da496abd0039eca54e023f : SUCCESS in 43m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a948c664980145f0982f188de91f2ee4 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c95fd46c9953403f8806ed037c952eb9 : SUCCESS in 25m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7efd78976ffb49609872cc0d80fb5e47 : SUCCESS in 28m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edf71507a3e34ce09c011fdbd50f638b : SUCCESS in 25m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa441345c5844aac992e927dc98c3c5f : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbf97f7922945629307f26720ae8545 : SUCCESS in 12m 04s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8031e65581d74d189b25944b262b2b67 : SUCCESS in 27m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/446c6f4313af4a24a26a50ac0d2354a6 : SUCCESS in 47m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1730148db92d4c8eac39fd10a0a0cdaf : SUCCESS in 2h 36m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6246444572e04b989538ba27949b7015 : FAILURE in 29m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/957b265205a64a18b4ca244944105a97 : SUCCESS in 49m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/254014f443834ebc9ae9f4cd157e583b : SUCCESS in 1h 22m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9f9a44dc7f204355a1addd5d97d95b11 : SUCCESS in 1h 30m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/929763060c9f4aab8e0f9b89f70f0311 : SUCCESS in 58m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d9f463f2de04b03b356e21748c86116 : SUCCESS in 1h 51m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac5ee8296f5140d884c0ac7fbd0b3260 : SUCCESS in 2h 22m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3775498331eb41ff94e79cb1ebe57cd3 : SUCCESS in 1h 09m 55s","accounts_in_message":[],"_revision_number":2},{"id":"28fa55c9e87d1bb6e890f859e7b812015505c840","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 14:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d247e81aa14b12ba70cffb986f94911393b4443","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 15:01:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfe1ad1885a4d1691e37196682e817f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfcd21c15563449385602a8f429c0b7e : SUCCESS in 49m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a610ebe28b73464fb8094c4091033f37 : SUCCESS in 49m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af5e5d3f8c8547abab48ae8a4b678412 : SUCCESS in 49m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3af19f221c314d19bb0104e58cd6cdaa : SUCCESS in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1fbb0e2486432af8297f61d9c15715c1d923114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 17:13:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68ef722c94d74979bd718a3a91886afc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b9284d2bf6e14c3e8706440c8ca3a39a : SUCCESS in 1h 16m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/58a100d65f384ad1a08d56473813dbd9 : SUCCESS in 1h 35m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/efa9a065b4c84208b1b08ebfc226a0b4 : SUCCESS in 1h 24m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4e36c4c8649b4d14bf891d317d395ff6 : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1a1c69ad27334e8da6da1c378b15ab7b : SUCCESS in 34m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed71c2260a3148859dfb0bf4622712ae : SUCCESS in 39m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7fef5f033244c28e799129250d64cb : SUCCESS in 10m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1400993ec18e483794d7f4853fcbcd71 : SUCCESS in 25m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4771683d83c4a82a022eddd96876d7f : SUCCESS in 28m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7bab54dc45f43db8c47c24ba17b0cef : SUCCESS in 27m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc06d1ad7ba24dc580a627546b4027ee : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc13281732484bf3b0c72e6669524815 : SUCCESS in 11m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/43a864e881c54d94af8132710cd4b1bb : SUCCESS in 28m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f33697f3b934b06b80b1d2a850a81c2 : SUCCESS in 1h 11m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e554667ecd984c49869fb0f5b2258f70 : TIMED_OUT in 3h 02m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1219a67ca77d404295e3424d072b3cd8 : SUCCESS in 52m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a454c0d81e8d4854bb4185aec574be55 : SUCCESS in 50m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/380f96a3264a478ca339c6fa85d24156 : SUCCESS in 1h 25m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7feb5d01ecf5402c8c904a48fc886325 : SUCCESS in 1h 14m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/68f580f2b5eb46dda670f7d00dcbed95 : SUCCESS in 59m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/768d4654ebd24487bbd7a3fab5286fd6 : SUCCESS in 1h 58m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d4ca2980c394609a63bb7594984c344 : SUCCESS in 2h 08m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dc3c8c5dcfc544dcad372803e3f5c85d : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":2},{"id":"24493669bc15408e3d2be6c226bc844da4c7102b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 10:13:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec2a20700b85fa14e7b7fc9a55691c7649dc3ecc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:57:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d99ca7039645119832c527c74eb10b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c389d1c8420e4705819cdd67b10cb5d2 : SUCCESS in 1h 00m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bd058cf368747c8ab28b4e8060c49f2 : SUCCESS in 39m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b5fa0581fcc14a1098916de778421a4d : SUCCESS in 38m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ef92fb016984d889e155413f1b6a482 : SUCCESS in 57m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12a42dc6c7299919748c79529359c79d6ec2a01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:36:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b3d6530fa7c46239ee2ee065d4601af\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/138c71d4b3b6470b98612edfae9ef5c9 : SUCCESS in 1h 44m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/235c9c26729645d885083e8cf6ba07fa : SUCCESS in 1h 48m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77547ad49f474a23b91e9ccf2d266686 : SUCCESS in 2h 03m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee1591c7a3704358b2d26777322f54bd : SUCCESS in 1h 54m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/13120cb6e0e44d17b8736be54bc47cac : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c59816047bd46708b695ea115bb2f05 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6e16f742f7466d92d16aec29d04452 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54e23aa30344baebfde5184408bf75f : SUCCESS in 26m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9867cf1b31b74bf8aac43ca59821bab0 : SUCCESS in 23m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f34b1660241489da6479f1b56aeb529 : SUCCESS in 27m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a610f98877d46fc913dd30cbd7943c3 : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1676805460476fac0387aa371941c1 : SUCCESS in 10m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c83e0b1a959446879ba6bf4593a39e9b : SUCCESS in 27m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93cbc4c9fe264cd6b257407f3d01e904 : SUCCESS in 1h 00m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e12fdea586e84d5e9509a5c785e5e7a9 : SUCCESS in 2h 17m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1283d09a97d8465cbe843b5b745bf874 : SUCCESS in 1h 03m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7355fcb4b7e4456bae27a93c2cd3da0e : SUCCESS in 57m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6ab6fa4c7aa940378f7b84c53144cfa2 : SUCCESS in 1h 21m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/232c2ac0b45b4954b7c72e1d65987ba9 : FAILURE in 1h 16m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d990f0d5e99346eb9cc825ff618d4264 : SUCCESS in 1h 01m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/290b0ca37c6542de9f037f166f0fc244 : SUCCESS in 1h 34m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77d1dbbf088e4186b555d44c83cb7c78 : SUCCESS in 1h 50m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2fb8930048e410e8af648d46210213a : SUCCESS in 1h 09m 05s","accounts_in_message":[],"_revision_number":2},{"id":"36fdaeaa6021edfdbd4e221b0d8622110503f719","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-10 14:27:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"300ca992a930530ca9a268fa72067ba0281a88f3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-10 14:37:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f8da9cf6f8b5c820a485fe8e6f4e02f4a8a17781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:48:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4e2c3a5fa6a48a88364d3442b0bf6e4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7ae016de01b44776866f1a40d5e5394c : SUCCESS in 1h 25m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2eb820ae2b014eacb74a6a6ca83a888c : SUCCESS in 1h 27m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b22c1bcaf65a454ba1306795d6c70d5a : SUCCESS in 1h 24m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95b81e42d4c74e7eaef0f4cc1fa53087 : SUCCESS in 1h 22m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/feafe8884f1e42d297e491ee32b75a3f : SUCCESS in 36m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3f8d96c8d5b4f9da0b7fb69e12cf315 : SUCCESS in 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3dc14de79704cb0821569ace8101228 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5190a684074413aadac40cecaa8bc3 : SUCCESS in 25m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96258a3c88514ac482545a4db6d4412a : SUCCESS in 26m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8ba68d59da34294ab88196446cd5a5c : SUCCESS in 24m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78f4418eb38f4a9aa45b3dc6c0073b7f : SUCCESS in 23m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d600299bfe54089932c7f3572fc2b4e : SUCCESS in 11m 49s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/522ab394720a4a1db5b6645ccf74c3df : SUCCESS in 26m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fd13619622c4fac8b4846705100e17c : SUCCESS in 1h 15m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f3f4f22b8a243e9a04839a5a78032db : SUCCESS in 2h 15m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/73263cf5526048d88511f5b3b1f000ef : SUCCESS in 58m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dddbd784d43840f1bf11bdd64206714d : SUCCESS in 58m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d18c860359554eba8e80d131e0683539 : SUCCESS in 1h 25m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/62c29ee83a12444e992b272fba09b627 : SUCCESS in 1h 19m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/321e8f87c59449d0b80ca709e344b726 : SUCCESS in 1h 04m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fe3523f1c76499289e82d3f5486f51b : SUCCESS in 1h 17m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/701f45a9b3bc4a5a9509070c1e4ec642 : SUCCESS in 1h 25m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fa7013bbe3244f56bc3a1c8a4f39f9ea : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"80c6007e2568b3c7c18cfca7b5363f2ed6d03950","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 18:48:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07ed2b86364c4b74a302ef094b43e6fc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3c678f5720045cab2dbf4e507c82ebb : SUCCESS in 55m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9cc578cb47ab4dfdb10f196e5720afeb : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/757e5a76a35e48dab37e225392498be6 : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f24d1f41e25f4c1096690b023aedadfd : SUCCESS in 34m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9319ddf846d607c426e82b5859716d580e635bad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-11 02:09:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c972be9f71346d59bc67f3cf666c4974965949a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 02:13:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4989c640dc5bbdf43bdf171655826e383764cda9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:32:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dd3220465f64fc390b8029a07705ee2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bca665c401a345fb95c3ca759dcff6c9 : SUCCESS in 1h 24m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/88f58d8874704d048254f67c79053e6f : SUCCESS in 1h 27m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bffaf39d059b4a3c99cfcdad70100c36 : SUCCESS in 1h 46m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/112aa06e42d347f0bf38d61d32135a55 : SUCCESS in 1h 31m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99387ebe8754ffb86c66f7675098a74 : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c3bdb4f047438aa82e23d1bb0e22d6 : SUCCESS in 23m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdac975bd0374c5295e5e36d316550d8 : SUCCESS in 28m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f19249218c10485fb9b2305c6b3e10f9 : SUCCESS in 23m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5339bf96de244b680212f7a823c9b7f : SUCCESS in 13m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/916f1756c2fc4368b4f0864f363f15e4 : SUCCESS in 56m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/681402a93c2346d0b573b5d5a116f4c2 : SUCCESS in 2h 12m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ad881a49b60486ca0ddf89c6e885e54 : SUCCESS in 1h 21m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1acf25f2208645979a59aee49651b447 : SUCCESS in 49m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/57616f27299f45e883a8ac3cfc302e15 : SUCCESS in 1h 06m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/61bb0cdb3d5742faae1c52dabe5de0c4 : SUCCESS in 53m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/733283ce7e0c4412bfe528e74f0409db : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":2},{"id":"8da0418ae9c0ad12e272fc518dc88d3597fd800b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:32:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d02e148bb5ce3be6a48f1485ca66607ace90f71b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 04:33:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93cb050340d445419d9da2b1d76399e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387ea1fa3fd1416fb0bb307fffa178f2 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"294e1c60b41d3422bb830758e2ea6b6cf554ac46","revisions":{"618d3c3a30ac0013b256812b8edd35eb2f59ab15":{"kind":"REWORK","_number":1,"created":"2023-10-06 10:50:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/897540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/897540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/897540/1"}}},"commit":{"parents":[{"commit":"0aa154b5ce9dc8da73309fb212843a2b69b68696","subject":"Fix the ``log.setup`` method call with \"fix_eventlet\u003dFalse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0aa154b5ce9dc8da73309fb212843a2b69b68696"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 19:32:32.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 20:02:06.000000000","tz":0},"subject":"[RBAC] Update the subnet policies","message":"[RBAC] Update the subnet policies\n\n* get_subnet: the network owner can retrieve the subnet too.\n* update_subnet: any project member can update the subnet.\n* delete_subnet: any project member can delete the subnet.\n\nCloses-Bug: #2038646\nChange-Id: Iae2e3a31eb65d68dc0d3d0f9dd9fc8cf83260769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618d3c3a30ac0013b256812b8edd35eb2f59ab15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618d3c3a30ac0013b256812b8edd35eb2f59ab15"}]},"branch":"refs/heads/master"},"294e1c60b41d3422bb830758e2ea6b6cf554ac46":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-09 08:04:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/897540/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/897540/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/897540/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/897540/2"}}},"commit":{"parents":[{"commit":"a45263e146ff8fe3fd8aebe8a9a83d0827b7043f","subject":"Merge \"[S-RBAC] Add service role in neutron policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a45263e146ff8fe3fd8aebe8a9a83d0827b7043f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-05 19:32:32.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-09 08:04:30.000000000","tz":0},"subject":"[RBAC] Update the subnet policies","message":"[RBAC] Update the subnet policies\n\n* get_subnet: the network owner can retrieve the subnet too.\n* update_subnet: any project member can update the subnet.\n* delete_subnet: any project member can delete the subnet.\n\nCloses-Bug: #2038646\nChange-Id: Iae2e3a31eb65d68dc0d3d0f9dd9fc8cf83260769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/294e1c60b41d3422bb830758e2ea6b6cf554ac46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/294e1c60b41d3422bb830758e2ea6b6cf554ac46"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
