)]}'
{"id":"openstack%2Fneutron~897565","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I8af6b1dc0552f0aba44400b41a542818b6f01fcf","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2038541","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-06 19:30:54.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-06 23:44:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8af6b1dc0552f0aba44400b41a542818b6f01fcf","subject":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily","status":"MERGED","created":"2023-10-06 15:10:07.000000000","updated":"2023-10-07 09:21:08.000000000","submitted":"2023-10-06 23:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897565-bug/2038541","meta_rev_id":"8358fbe2dadc37e3911c24f49d916c7ecbab806d","_number":897565,"virtual_id_number":897565,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 23:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-06 22:17:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-10-06 19:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 15:12:27.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-10-06 19:15:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"778a39a4e7966bbee77090ea0b04725f44094fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 15:10:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58cd95ef8f5489804ab1aa122a7b27a92963f829","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 15:11:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4709376e00e315913a982c7646dbcf5733c48034","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-06 15:12:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c72e6f7179457cdcc7961d37a0270d7f45db6f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 17:02:27.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c1252cc0c4043d85b08abc59cade0a01e7283b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 17:02:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b50984350c62a092b5a7163b0b72d0b2fe64f80e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:15:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c9ac96024c34cd0ba60e4f3e13127c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/406bf076b3494614870c159f9ac1fbcf : SUCCESS in 11m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb6cd456c164d949cd2edb02cdb1677 : SUCCESS in 10m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac02401d979d475ba9149847ff619118 : FAILURE in 1h 56m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6754287b207f48558e0969cb578fb67d : SUCCESS in 1h 02m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d874c674ca943608d1bca9d8312d8b6 : SUCCESS in 1h 03m 57s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bcf27d59f81b4e78b89834cd692553e4 : SUCCESS in 54m 37s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ed75b8b52930450fbf0a1140c447ca55 : SUCCESS in 1h 06m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/16391f227647418d89d674fc3c9df722 : SUCCESS in 1h 30m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dfa2b82ea0b54043890b4799d6a45d47 : SUCCESS in 1h 22m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/df01f7d3246246098f8cb189047e176f : SUCCESS in 1h 03m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/67038f9885ab46adb45e88a054d1e2a4 : SUCCESS in 1h 11m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8e74cbf2e3f04e4299dcabfe531a98d2 : SUCCESS in 2h 07m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c7e0d38f176743f481b2db01d532ffd3 : SUCCESS in 1h 03m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ee938030c3bb4bb2bc4219071c78892c : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":2},{"id":"32acce195914fcd34c37584843a8811e9d02195d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 19:30:49.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ec232d8e76a65637bdb7f4e0093c70f14f8c2050","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 19:30:54.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d042a31b9a3b52c596e9316e49d48eb7eabf264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 21:35:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb47a7126504716bbbb3649e77fb49a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b47949d1594b65a5b0d96ddca80293 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b5a4929a144e639ea8acbdb937fca7 : SUCCESS in 10m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/304afef5fbe945f0b601247bb37f6fce : SUCCESS in 1h 03m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/adbdd9d9a48e4fee984eebd3a29f3e99 : SUCCESS in 48m 00s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a128fde4b41f47df8eef18487fa9f2ef : SUCCESS in 1h 22m 34s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/833834f585794718a7fdfa2aebb7bfdb : SUCCESS in 56m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/afbebe68319047a780224c9019667a7f : SUCCESS in 1h 11m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/815fa88c6c46437e945f97e36295bc0e : SUCCESS in 53m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/371e2a04cf534a4295962528027530de : SUCCESS in 1h 05m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/21ec907f0abf4e4990247a65c4cfc496 : SUCCESS in 1h 58m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/efc2fead22c6401f8283385339034a97 : SUCCESS in 1h 58m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/40b7cb35988749dbb972391046b4d955 : SUCCESS in 53m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/922fbcb52b8247ef80ca02553b7f5be3 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":3},{"id":"41fb9458f7348dd90bec22b4ffaa94707e444c69","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 22:17:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3760d443273aee09c6a878ce8207e6b9fa58610f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 22:18:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c881e73697b5c6e16e23753b159142a340bb788","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 23:44:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83124414ab047e08dabb3a392d2a580\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2afb060132a04d959faf039742091035 : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c668c3efe214d9fbfb7838e6c070a9d : SUCCESS in 8m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c8d05da9e1144e0890c90b331ed91827 : SUCCESS in 1h 20m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7c8d767a07e48a8afdc412544bb2dc2 : SUCCESS in 43m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/41e3169e17aa405d9d6b8d5b55ebb6ba : SUCCESS in 42m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0f5ff79dd9b8435ab773cea7e804afc3 : SUCCESS in 1h 04m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a7d69a2a016c4f249944e9d2c8cc5837 : SUCCESS in 49m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/23057fa1bf5a41a5986bcec1939fe5b0 : SUCCESS in 42m 51s","accounts_in_message":[],"_revision_number":3},{"id":"11d6ed2a051e3be6ae14a90e342b876a6d191316","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 23:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a32f7ae6d63423152dea20e0c0cf806c6a15ffd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 23:46:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5c20153f3040eaad10075c136bb025\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4bd1f453a54c4094bd0d16d9c437e8 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"8358fbe2dadc37e3911c24f49d916c7ecbab806d","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-07 09:21:08.000000000","message":"Created a revert of this change as If2d1d54bea639e922ea46e9b8fae12c98c6f640c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27a19fb88afba2900c611741f5975d7461455a0f","revisions":{"37fd3ab073df832a6dc233ca0624001de9b1c9ac":{"kind":"REWORK","_number":1,"created":"2023-10-06 15:10:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/897565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/897565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/897565/1"}}},"commit":{"parents":[{"commit":"2319e7c35f79798cba199f52a5a7c4a4ef7bc088","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2319e7c35f79798cba199f52a5a7c4a4ef7bc088"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 00:20:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 00:20:14.000000000","tz":0},"subject":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily","message":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily\n\nThere is a cyclic dependency between two patches:\n* https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/897529\n* https://review.opendev.org/c/openstack/neutron/+/897440\n\nIn order to merge the Neutron one first, it is needed to disable the\n\"neutron-tempest-plugin-jobs-2023-1\" jobs.\n\nThis patch is proposed for 2023.1 only.\n\nRelated-Bug: #2038541\nChange-Id: I8af6b1dc0552f0aba44400b41a542818b6f01fcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37fd3ab073df832a6dc233ca0624001de9b1c9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37fd3ab073df832a6dc233ca0624001de9b1c9ac"}]},"branch":"refs/heads/stable/2023.1"},"fac3086a005b42570a4c2e4a8d6b983136a893d4":{"kind":"REWORK","_number":2,"created":"2023-10-06 17:02:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/897565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/897565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/897565/2"}}},"commit":{"parents":[{"commit":"2319e7c35f79798cba199f52a5a7c4a4ef7bc088","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2319e7c35f79798cba199f52a5a7c4a4ef7bc088"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 00:20:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-06 17:02:26.000000000","tz":0},"subject":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily","message":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily\n\nThere is a cyclic dependency between two patches:\n* https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/897529\n* https://review.opendev.org/c/openstack/neutron/+/897440\n\nIn order to merge the Neutron one first, it is needed to disable the\n\"neutron-tempest-plugin-jobs-2023-1\" jobs.\n\nThis patch is also disabling the functional jobs that will be fixed\nby the upper Neutron patch.\n\nThis patch is proposed for 2023.1 only.\n\nRelated-Bug: #2038541\nChange-Id: I8af6b1dc0552f0aba44400b41a542818b6f01fcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fac3086a005b42570a4c2e4a8d6b983136a893d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fac3086a005b42570a4c2e4a8d6b983136a893d4"}]},"branch":"refs/heads/stable/2023.1"},"27a19fb88afba2900c611741f5975d7461455a0f":{"kind":"REWORK","_number":3,"created":"2023-10-06 19:30:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/897565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/897565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/897565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/897565/3"}}},"commit":{"parents":[{"commit":"2319e7c35f79798cba199f52a5a7c4a4ef7bc088","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2319e7c35f79798cba199f52a5a7c4a4ef7bc088"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 00:20:14.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-06 19:30:49.000000000","tz":0},"subject":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily","message":"[stable-only] Disable \"neutron-tempest-plugin-jobs-2023-1\" temporarily\n\nThere is a cyclic dependency between two patches:\n* https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/897529\n* https://review.opendev.org/c/openstack/neutron/+/897440\n\nIn order to merge the Neutron one first, it is needed to disable the\n\"neutron-tempest-plugin-jobs-2023-1\" jobs.\n\nThis patch is also disabling the functional and fullstack jobs that\nwill be fixed by the upper Neutron patch.\n\nThis patch is proposed for 2023.1 only.\n\nRelated-Bug: #2038541\nChange-Id: I8af6b1dc0552f0aba44400b41a542818b6f01fcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27a19fb88afba2900c611741f5975d7461455a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27a19fb88afba2900c611741f5975d7461455a0f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
