)]}'
{"id":"openstack%2Fneutron~897578","triplet_id":"openstack%2Fneutron~master~Ia23d68628e1fe1579c274aa54d4ac40f6953fa19","project":"openstack/neutron","branch":"master","topic":"bug/2037107","hashtags":[],"change_id":"Ia23d68628e1fe1579c274aa54d4ac40f6953fa19","subject":"DNM WIP Remove network RBACs from subnet view","status":"ABANDONED","created":"2023-10-06 16:59:26.000000000","updated":"2023-12-18 14:48:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1ad21677c4ac1a4df3ebc428bb4b0378bb66a656","_number":897578,"virtual_id_number":897578,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-06 19:57:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"}]},"reviewer_updates":[{"updated":"2023-10-06 19:13:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-06 19:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 10:19:08.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-11-29 10:42:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2023-11-29 10:42:42.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32366,"name":"Waldemar","display_name":"Waldemar Reger","email":"waldemar.reger@inovex.de","username":"wreger"},"state":"CC"},{"updated":"2023-11-29 10:42:51.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":36350,"name":"David","display_name":"David Hitze","email":"ubuntuone@hitze.cologne","username":"dhitze","status":"inovex GmbH"},"state":"CC"}],"messages":[{"id":"431684392a04764215337bd3a4be78015943e71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 16:59:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"440c9e00423b453fd6a1ea1b4de20fc30c5d77ac","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-06 16:59:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"589cb9515b7b3de00924ae12165cabe41d9819cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c449f6e8bf6e4c6483bd0e6093295617\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8251b5969d4418fbb8993b6faaad15f : FAILURE in 38m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/798f07c29d274140ba31061fb425a2dd : FAILURE in 46m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53148fe8f3cd4934a8ab8e0b9fa652b3 : FAILURE in 41m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c0b9d12cb294b09b08bfd3c8a56c38a : FAILURE in 39m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9343818c86cf58484d91f0c372fbdac7182fd6bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:57:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5559093b9599462a9c73de5f67f7ee42\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/02b7627c08ae4b68b5ceea9345145b91 : TIMED_OUT in 2h 53m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f0c2454fe23344d59e112e61fb6ca8b2 : TIMED_OUT in 2h 55m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41f30d5c8e634e4080b35b9885c6fcf7 : TIMED_OUT in 2h 53m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6edabe5b4fb8452da6d03862b21f1d29 : FAILURE in 1h 51m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ebca2c5a5f3649b6a598ec27d3b85812 : SUCCESS in 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecdb5d8c86964300a0712dea07b2f0fa : FAILURE in 37m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd467af9554d469aa4fec3417ee1856d : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31c1d22bc0b142598159295a1bbe6d3b : FAILURE in 26m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aa26b8149f9476fb6c704bbeda50f64 : FAILURE in 29m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47f98c7dd77f40bbaea6404430ddc371 : FAILURE in 25m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e944e816420746ea8159fb575ef87eee : FAILURE in 26m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c64c71c75b74f929c83eb64873fdd31 : SUCCESS in 10m 27s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/75f24c97a9fd4f7c80d7540ccf9be3fa : FAILURE in 31m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5321b765e6794e9282d5c404dffc53f2 : FAILURE in 54m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a598ec0ce3f41089b737b582e5891de : FAILURE in 2h 22m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3afa45da49c046f5842e70235f700d7b : FAILURE in 1h 02m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0528c493f0b4b419b0028f0418f9ce3 : FAILURE in 1h 07m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6dea2be208e447a98366ceedefda2320 : FAILURE in 1h 21m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2957180f0e0f48b5bb98df6481ac7c01 : FAILURE in 1h 14m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb5063552c4d44378d0bbdd1564098b9 : SUCCESS in 51m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f65be241c2264b50a8936be5259e9b13 : SUCCESS in 1h 20m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc0c7cf60fba4efc86a181a2feaf2823 : SUCCESS in 1h 35m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6b7fb3b537634f039346c8c31dc5303b : SUCCESS in 1h 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1ad21677c4ac1a4df3ebc428bb4b0378bb66a656","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-18 14:48:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"daedeef313946528ef79605e4ec908211d885599","revisions":{"daedeef313946528ef79605e4ec908211d885599":{"kind":"REWORK","_number":1,"created":"2023-10-06 16:59:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/897578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/897578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/897578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/897578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/897578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/897578/1"}}},"commit":{"parents":[{"commit":"a45263e146ff8fe3fd8aebe8a9a83d0827b7043f","subject":"Merge \"[S-RBAC] Add service role in neutron policy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a45263e146ff8fe3fd8aebe8a9a83d0827b7043f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 02:11:16.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-06 02:11:16.000000000","tz":0},"subject":"DNM WIP Remove network RBACs from subnet view","message":"DNM WIP Remove network RBACs from subnet view\n\nLP 2037107\n\nChange-Id: Ia23d68628e1fe1579c274aa54d4ac40f6953fa19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daedeef313946528ef79605e4ec908211d885599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/daedeef313946528ef79605e4ec908211d885599"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
