)]}'
{"id":"openstack%2Fneutron~897937","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I546152312c1dfda69f90154cea5d87392e16a7ae","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2037596","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-16 14:16:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I546152312c1dfda69f90154cea5d87392e16a7ae","subject":"Parameter filters may be None, which cannot be called with **","status":"MERGED","created":"2023-10-11 10:54:19.000000000","updated":"2023-10-16 14:18:21.000000000","submitted":"2023-10-16 14:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897937-bug/2037596","meta_rev_id":"15d03bee314e997de298ab21e38e6c9978bac2f8","_number":897937,"virtual_id_number":897937,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 14:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-10-16 11:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-10-13 17:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-10-16 11:41:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 10:54:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"state":"CC"},{"updated":"2023-10-11 12:47:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-13 17:11:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-16 11:41:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"8676b745afa91501c5fdff3385ea6a86e215b00b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 10:54:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6782191cd61d7fb178975b3d91c3d4ca128260b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:47:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9de929bd649479691a27b10cefb3de9\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8824cbe3ba8c4cb781942c036c802ea0 : SUCCESS in 51m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/f46b87570e864690a2b64cb0ace7d75d : SUCCESS in 1h 36m 13s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/1416168bc9604c42a69a01cdcee1203f : SUCCESS in 1h 05m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/b8d856581ab044a5a418390f13f494ff : SUCCESS in 1h 23m 23s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/9322412f750c466f8e8aa8ddce14f674 : SUCCESS in 1h 10m 23s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/af1f049bf6d94f5395cc55ac94b32178 : SUCCESS in 30m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7431b5cdb286452fafdd7f46d03c6849 : SUCCESS in 37m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ecc73b71f040c7a127e79abc4a1e05 : SUCCESS in 14m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/490a505c274942a2976003448ac50f60 : SUCCESS in 26m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cf58aca627a4ce6a43444468bd3b616 : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c906521002014233a29b24aae8e1c4ec : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc4295cdbcff4e94836fbc8c08fa61df : SUCCESS in 53m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/000437fd6a0f47929d6bdc8cca586022 : SUCCESS in 1h 45m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f5c7040d8b45489ba87b12e55388595a : SUCCESS in 1h 09m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0637f19f5b645be9468f03c73b01377 : SUCCESS in 59m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6a69a0388e94e9fb3563c6af4b65d09 : SUCCESS in 54m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/83def2ccff91490888b11cec607721d7 : SUCCESS in 1h 00m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/10463fd4f0ae49448e6d43da08771b73 : FAILURE in 1h 08m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/917ef395d2fc4c0d9db5dd9352c5edcf : SUCCESS in 1h 00m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1061273adf984990b8d02a5e41692ebf : SUCCESS in 1h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ce45bf8e616c945f35608946b36a93d821009b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:19:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56534606bdb245ecae52fcbbfa1adc92\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5223c298f9c4ff890f3e9938068bb50 : SUCCESS in 38m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a1a509d389c4fc2bf47efc2f6e15d37 : SUCCESS in 38m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56fa24d8d6d8dd9eae6f3deac0b7377cc01bf9c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-13 10:57:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b6b7e35bb366f4d4b778bb99316d318409eb1d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 11:45:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8116177fd344fe0a9154b691a96c8f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fa91b037f69456abc4fb3500ce43109 : SUCCESS in 41m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31b644e3a2c34adeb2e49fd382ba1287 : SUCCESS in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44732a1347920dd0b0d3ff471f734de519867050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 12:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0628fe26962049be896e148d676ec0fb\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/53c9369024c545d1952a73f3ee78af06 : SUCCESS in 57m 47s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/15e6e7cdd6ca40ab93a774b7848fae44 : SUCCESS in 1h 54m 07s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/1232fc10b9d549339a136c2a6d613297 : SUCCESS in 1h 24m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/aee760bf3f8c4d0a8fc855e562fb8eb0 : SUCCESS in 1h 23m 07s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/aeeecccddf964ae1be6b9a2464e1af30 : SUCCESS in 1h 27m 55s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/5c20293de2e04e188dd65deb02fde9da : SUCCESS in 41m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21760b4a6a8d493a81cb815c63c80455 : SUCCESS in 36m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e5f703e5b140908f93c9455fd67f79 : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7956f1a0bfd0462389a26f0116276f75 : SUCCESS in 30m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffc0e2a52070447493c1267b4275aa17 : SUCCESS in 22m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d01b04018ff481cbe973a18c38d3809 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a69cc7cbede4ccb889eee597b0f899a : SUCCESS in 54m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4772f0b41d584f8e978440bf103976f9 : SUCCESS in 1h 39m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/28ae03ee086b41c1894c9185eeecefdd : SUCCESS in 1h 05m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff3b3cdf71394a31a0bb30121705f297 : SUCCESS in 46m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b4664c99df9e4331ae4300d18e3132f9 : SUCCESS in 1h 01m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e32af499f0774fd5aeaa0df23422e786 : SUCCESS in 1h 11m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/86cf0c982abf4b9780e37d0a28322d02 : SUCCESS in 53m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3858733fb4a64ea9a5cc094a4039b9fe : SUCCESS in 46m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbd23199120c48039bcf76753f739627 : SUCCESS in 48m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c81ef6b2052282f78144090c63d0086368e31de0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-13 17:11:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"122c92bf45d8364e8e938fe4dcd804905b15c04a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-16 11:41:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1c5a371667e5db40738fc5aca970acaef6ee556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 11:41:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82f84c1bac5a148f0d45b1537717134fff7f0482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:16:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb606165951d46d4a09fbee7f0289210\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/0786efef066b4284a5a96a2cd76fa3ed : SUCCESS in 57m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe5161a52f84865a6a1b20166c50eab : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bed2077a88c4ecc9ced155b7c10303b : SUCCESS in 28m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98740e235883414e9d7deb709db184a7 : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b78f152cca846a192fd0e94bce8b5e5 : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/668b8ee5a7204792b7304d85dce015f6 : SUCCESS in 57m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d6899ef4bd540e789d3cf3268c20ffe : SUCCESS in 1h 20m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/827cda3296ce41ed8e2dd1c3d224f33d : SUCCESS in 1h 09m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7552306c6e1344c58d7a74731db4399a : SUCCESS in 1h 05m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9fbd9e1f08b448d9b47df3512ec6f9e0 : SUCCESS in 57m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b00f8e0f9ace49cba0fa11f81e5b01ab : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"afd4cdb8514100ed3a2f17327987c0ea898a8b0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:16:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15d03bee314e997de298ab21e38e6c9978bac2f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:18:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc20c52c112d404d85d669cc81b0e309\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8533376b03574419bb037dfe7e19ef73 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69c4712d8ddbeaf3108b97501a5da34e808e09fd","revisions":{"69c4712d8ddbeaf3108b97501a5da34e808e09fd":{"kind":"REWORK","_number":1,"created":"2023-10-11 10:54:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/37/897937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/897937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/897937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/897937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/897937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/897937/1"}}},"commit":{"parents":[{"commit":"cf096344b07b80524c3888e44e0b895465598a74","subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf096344b07b80524c3888e44e0b895465598a74"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-09-27 11:52:31.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-10 19:08:27.000000000","tz":0},"subject":"Parameter filters may be None, which cannot be called with **","message":"Parameter filters may be None, which cannot be called with **\n\nChange-Id: I546152312c1dfda69f90154cea5d87392e16a7ae\nCloses-Bug: #2037596\n(cherry picked from commit 7e10a01874b7d2e7f4a28b740a2ebfcd193ae797)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69c4712d8ddbeaf3108b97501a5da34e808e09fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69c4712d8ddbeaf3108b97501a5da34e808e09fd"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
