)]}'
{"id":"openstack%2Fneutron~897938","triplet_id":"openstack%2Fneutron~stable%2Fxena~I546152312c1dfda69f90154cea5d87392e16a7ae","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2037596","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-16 14:16:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I546152312c1dfda69f90154cea5d87392e16a7ae","subject":"Parameter filters may be None, which cannot be called with **","status":"MERGED","created":"2023-10-11 10:54:30.000000000","updated":"2023-10-16 14:18:16.000000000","submitted":"2023-10-16 14:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897938-bug/2037596","meta_rev_id":"adf7f1e9538f30cbd8e974784a29916a25dc548d","_number":897938,"virtual_id_number":897938,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 14:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-10-16 11:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-10-13 17:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-10-16 11:41:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 10:54:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"state":"CC"},{"updated":"2023-10-11 13:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-13 17:11:35.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-16 11:41:25.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"751df7c2b0f3e8ab4cc4ff76f823193a68ea7fbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 10:54:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b78febf30d498a30a81c36d3bf40e641b68b431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:54:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9c03436dda542599c2a82cc2e59fe4f\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/1d4f3be633e1451c83c21735a3b8b285 : SUCCESS in 1h 02m 40s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/c632286b365a4c10a24e91a289dc1e22 : TIMED_OUT in 2h 49m 38s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/3c2d3eb79cda45ccbba8b7dd301aa9c0 : SUCCESS in 2h 36m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/6f8b3d6f64b1409f80f65dbad504cacc : SUCCESS in 2h 07m 37s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/dd18f1ad31db4dff9b7ba997bd74c45a : SUCCESS in 2h 54m 08s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/9e5e3d7a36494dceb4e7be4a167fcc1f : SUCCESS in 53m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5f74869c61b4f859eb067c5c6a6a8df : SUCCESS in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404e2345803f4dcbb2309d57ca87d928 : SUCCESS in 12m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9a6e68f1bf24b38bedd8bd5e77cf533 : SUCCESS in 28m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57d7156c4b3c4465954c876ee1ba0c28 : SUCCESS in 20m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2850610d6ea42299a473c5e9e332feb : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486745ddd94b40f8b7b6e6d18bb32bf5 : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b73f69cfa9f41448651508ee48a5541 : SUCCESS in 58m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28c20399e7f34d4e908eae6ee52e44af : SUCCESS in 1h 15m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f76e9782eec14bdc981175cef55de876 : SUCCESS in 1h 12m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6db9eb88d3cd4fbf8df15cd82eba60ae : SUCCESS in 1h 16m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3361fe8eb5e45249be4c35711637484 : SUCCESS in 1h 08m 12s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/162be76c15814e0697290bf3c6dc77d3 : SUCCESS in 1h 13m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55d03618246945d1b5e7e9aa60c11d14 : SUCCESS in 1h 15m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3da3afed48624395a50e5d2ddde59d38 : SUCCESS in 1h 00m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfecd1675c654a8285902113be4e954d : SUCCESS in 1h 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a560f914740dd43d4394e3efe0302e85227f843f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-11 14:05:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9788979e59e74c0140dbfd0a41bea22f3106de60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 14:27:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c5470007864494d8701bcec54985fe0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/375ee820ab084a5286bae1d5424b7227 : SUCCESS in 38m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8abc03153d3048588d762bbe4c8679c7 : SUCCESS in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"895df4212ff19abd66b6d2b7819564adb7ef4820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:56:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e585c4dad8d4663b6d3ea7f508071be\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/cd385850886e4d80b2ff674aa8d2790a : SUCCESS in 46m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/9999dac4c81b4046b2804f5cf9e85dd2 : SUCCESS in 2h 24m 56s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/1c6d3425d0584114a50ffd1a1062c5ed : SUCCESS in 2h 29m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/bb42dbb1d7f64533942b64b50f891c4d : SUCCESS in 2h 15m 32s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/690f6912dfed41c08a0d83800642f583 : SUCCESS in 2h 39m 05s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/76e1366820d64197b8939f3df06f8df8 : SUCCESS in 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94837ce17e254ec3846cbd1f9b516a48 : SUCCESS in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c026f82a8e804f919a6e0bc83436108c : SUCCESS in 15m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a49c094935e4103b4a359c26e1e3eb5 : SUCCESS in 27m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7824ac96f3da450494e654abc57df1ef : SUCCESS in 22m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c2c79b7bd74518b1ccf45e2b093981 : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f923ad7278b41a392215b7a1cecdf50 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84c29e78398f458fa5d24f2864f2c8d8 : SUCCESS in 51m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7edda515893c47f0a3c1cb2f25a70a81 : SUCCESS in 1h 27m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d5708181b3854cffa6fe8ded4ab22ab7 : SUCCESS in 1h 19m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a626e0b6718f4872a33f91203fb7d912 : SUCCESS in 1h 05m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f24b50459a0d4f259e1f0e42241dc30c : SUCCESS in 1h 13m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9815fa7f6b3041c1a7e1704b73ed5d1f : SUCCESS in 1h 10m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/76585bc832b8436c938a27eecb476c56 : SUCCESS in 1h 24m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8675b602c2fc4311bcf8c76c0e3bfd0f : SUCCESS in 46m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/45edbee4dda34c948368ea53d84bcdfd : SUCCESS in 1h 14m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7cd0d3bb32fcd11d8749463b410a1abf291aed4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-13 17:11:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78f7732cc92ca613423bc4b459db7b78f785d7e8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-10-16 11:41:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"572674c602626cf6d57b66b6ef84f03e9fa40df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 11:42:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45183d8d8e0aaa31363ef129365753ed73f7fbd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:16:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0948b811bcb24b9a9f26528ddaa08a7e\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/238bfe1b211f4c8f84bb50c32c868d07 : SUCCESS in 38m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c2f4f7bea5b4cb3893264f88165e649 : SUCCESS in 12m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8eb429446a494d2f9e1eb01ce71d4ff1 : SUCCESS in 26m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c308d8fd42c348e492b231f2641890a3 : SUCCESS in 23m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21682a9a6a14c6c88083b626dc29cf8 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48020fe9ae4c457b904c0b0d5702acca : SUCCESS in 52m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3ef90eb0013442bbbacac912ab9b3f8 : SUCCESS in 1h 26m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef49ff70d3294b809aaa0bedb4301dfc : SUCCESS in 1h 00m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/01cf18b0b9714881b0403aaf034d1858 : SUCCESS in 46m 18s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67844be6333b4809b6323af62de8c6b5 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5c2247e6cf6d76cfef39f959942cb1785ad0d9db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:16:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adf7f1e9538f30cbd8e974784a29916a25dc548d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:18:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49dda1ffce74f5fae582c99cae046f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00f152f095e4044b80926467183588d : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bd6497909236114e8929c2bd6bee9a7e51876c7","revisions":{"7bd6497909236114e8929c2bd6bee9a7e51876c7":{"kind":"REWORK","_number":1,"created":"2023-10-11 10:54:30.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/897938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/897938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/897938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/897938/1"}}},"commit":{"parents":[{"commit":"6172f4233ef0bd1199deacefbf81f3104b5ccc7a","subject":"[OVN] Fix rate and burst for stateless security groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6172f4233ef0bd1199deacefbf81f3104b5ccc7a"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2023-09-27 11:52:31.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-10-10 19:08:40.000000000","tz":0},"subject":"Parameter filters may be None, which cannot be called with **","message":"Parameter filters may be None, which cannot be called with **\n\nChange-Id: I546152312c1dfda69f90154cea5d87392e16a7ae\nCloses-Bug: #2037596\n(cherry picked from commit 7e10a01874b7d2e7f4a28b740a2ebfcd193ae797)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bd6497909236114e8929c2bd6bee9a7e51876c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7bd6497909236114e8929c2bd6bee9a7e51876c7"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
