)]}'
{"id":"openstack%2Fneutron~898293","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I09a37ca451d44607b7dde344c93ace060c7bda01","project":"openstack/neutron","branch":"stable/2023.1","topic":"dhcp-scheduler-with-ovn-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-16 21:48:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09a37ca451d44607b7dde344c93ace060c7bda01","subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","status":"MERGED","created":"2023-10-16 12:19:25.000000000","updated":"2023-10-16 21:49:23.000000000","submitted":"2023-10-16 21:48:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898293-dhcp-scheduler-with-ovn-stable/2023.2-stable/2023.1","cherry_pick_of_change":898292,"cherry_pick_of_patch_set":1,"meta_rev_id":"005ca17e3e453aa0d4aff23801838cb4d8f282b6","_number":898293,"virtual_id_number":898293,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 21:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-16 13:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-16 16:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-16 16:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 13:05:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-16 14:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 16:44:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"de2516e47cf87ee7c0e9b526b1e67fa674b96434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-16 12:19:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"f6e8f0264fdb84dda40078b5c0fbf73672b37dbb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-16 13:05:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43357bedeae1c085182310e79bc959d449792d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:06:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11fa3c40bf224505aeba4539822f9faf\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/f0ec93fe699542d3a90b16be697375c0 : SUCCESS in 1h 23m 05s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/455afa4d5272490f8c6539e936f1047e : SUCCESS in 40m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8baebd8e22041fbacf05a29f9a71f88 : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8543eae30b1744feb5cfaf13bbf10b70 : SUCCESS in 13m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22f1997a8f0a4fb3ad4e89120532ad68 : SUCCESS in 22m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83b8632ec79746eda9d458e5112d3e32 : SUCCESS in 22m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d932c75b04294b73a6d4c36de3e104fd : SUCCESS in 10m 05s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7ea5d6a7a59d4b79a8314cc2384d400e : SUCCESS in 19m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5dc09d8f742a451cb2418918ccc55502 : SUCCESS in 54m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc61bb6475854f3fa83bb779ae477bb9 : SUCCESS in 1h 44m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10ee976b84d74dc7b7993f46826e1a48 : SUCCESS in 1h 18m 02s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/31dec0568da844b381788c343d17edc2 : SUCCESS in 1h 12m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/beb80342789b4428829d3dc48f20e489 : SUCCESS in 1h 28m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4a8b75a792bb4463be457a10bfd8cdfb : SUCCESS in 1h 28m 43s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/70537301dff64fd5be2a459ae989dcbf : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4551363fe120253fb88e8c1c3fdfc462d538292f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:15:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202b36d51c464efdbd3956aa92c18c4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b005c9ee98f4779a14d36c41ba30274 : SUCCESS in 47m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4ea35d280684ba5b5629a37d59b75a9 : SUCCESS in 46m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca271c16ad3ee5ce71c59e59bd5f6d5cd907294b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-16 16:44:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5db374562144be1c536aa728c0c94cc743ccafaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 16:44:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4889359e2eb4cf3126d1de985eef8527358e8803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:48:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4342247b279d4c45bb726572b00ef89c\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/167444a0adeb403281cbbcba46b43069 : SUCCESS in 1h 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/370557ec8b6946689f5569a6503647e7 : SUCCESS in 13m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4389cfc52ba746d5a63eaaff36a249f3 : SUCCESS in 22m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeef746ff3c74d1b8205f67c6a1903d9 : SUCCESS in 21m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc89fc676d2445659a04aeab6714bf40 : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d582c54695743328309467fd8c40a4f : SUCCESS in 50m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de839cfb7150493fbff707ea71084d44 : SUCCESS in 1h 55m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2c4fe3a9b254448da501f40284041e57 : SUCCESS in 1h 05m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f174b03930c64797a1d37e94aab83265 : SUCCESS in 1h 11m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/603bd0d4f1b845ffa55bf3533f570d69 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dc7cdc559f3d3422925a40936ca7410cecfd1077","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:48:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"005ca17e3e453aa0d4aff23801838cb4d8f282b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:49:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b9395218319429a95985acf3c5f4a1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529a222d1d8f45beafcf3c0e3b8c1c48 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3123df2672dac33edb22742bba2b6a0812fdd1a0","revisions":{"3123df2672dac33edb22742bba2b6a0812fdd1a0":{"kind":"REWORK","_number":1,"created":"2023-10-16 12:19:25.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/93/898293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/898293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/898293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/898293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/898293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/898293/1"}}},"commit":{"parents":[{"commit":"571ac477364057f3c876d9def66f2f4ce32b5891","subject":"Merge \"[OVN] Add the default condition check in ``PortBindingChassisEvent``\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/571ac477364057f3c876d9def66f2f4ce32b5891"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-06 08:58:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-16 12:19:25.000000000","tz":0},"subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","message":"Add dhcpagentscheduler API extension to the ML2/OVN extensions\n\nIn most typical use cases ML2/OVN backend don\u0027t needs to run DHCP agent\nas OVN provides DHCP functionality natively. But there are some use\ncases like Baremetal provisioning over IPv6 or Spine Leaf architecture\nwith DHCP relays where DHCP agent is necessary and it can work perfectly\nfine with ML2/OVN backend.\nThe problem was that dhcpagentscheduler API extension wasn\u0027t listed as\nsupported by the OVN backend so it was filtered out from the list of\nsupported extensions during start of the neutron server. This caused\nproblems with API to get/set/delete networks to/from DHCP agent.\n\nThis patch adds this API extension to the list of the extensions\nsupported by the OVN driver to fix that issue.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/898090\n\nCloses-bug: #2038655\n\nChange-Id: I09a37ca451d44607b7dde344c93ace060c7bda01\n(cherry picked from commit f006d29251abe3c138ee0dea5b549496b66b8fa7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3123df2672dac33edb22742bba2b6a0812fdd1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3123df2672dac33edb22742bba2b6a0812fdd1a0"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
