)]}'
{"id":"openstack%2Fneutron~898296","triplet_id":"openstack%2Fneutron~stable%2Fxena~I09a37ca451d44607b7dde344c93ace060c7bda01","project":"openstack/neutron","branch":"stable/xena","topic":"dhcp-scheduler-with-ovn-stable/2023.2-stable/2023.1-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-16 22:35:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09a37ca451d44607b7dde344c93ace060c7bda01","subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","status":"MERGED","created":"2023-10-16 12:19:58.000000000","updated":"2023-10-16 22:44:25.000000000","submitted":"2023-10-16 22:35:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898296-dhcp-scheduler-with-ovn-stable/2023.2-stable/2023.1-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":898295,"cherry_pick_of_patch_set":1,"meta_rev_id":"a24c0efea9a2ce6eebf7a8f8575b1456e2ac6af1","_number":898296,"virtual_id_number":898296,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 22:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-16 13:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-16 16:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-16 16:44:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 13:12:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-16 14:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 16:44:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"dc536c1e9bd9d3cea7dda5b5d334d53673dd4da1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-16 12:19:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"b34380a1dfc5195bbcd93531cd486e15e752952f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-16 13:12:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f3a4d6de43a759e72af30b8c82faf00b9f87fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:18:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8860c6f2d64fe8946c9874830095af\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3f8fc1ba35404c2180ce9229235df113 : SUCCESS in 48m 24s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/d3b486f8245245b6bb02f11992221d1b : SUCCESS in 1h 52m 59s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/46e4f71ab0ac465e9085c989c8bc31d2 : SUCCESS in 50m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b952db26c6d6433e88916cc1a2e9e48d : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2750d4e9b7f84c6d9b0c7148d4276405 : SUCCESS in 13m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857a1a1d1c034dbea74ef4c93d187ae1 : SUCCESS in 28m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a3f3f01fa74bcd99877e244f857519 : SUCCESS in 25m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952efbdefbdf48ca8fa0300b63212f26 : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a40cb8740d4f0790c6bdfabbecd8f4 : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2520ab1246c4f77bb7a933dc5d639c2 : SUCCESS in 51m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1287f28e32b144379603f150d1b17a18 : SUCCESS in 1h 17m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f6f27d2ed694a459200bbd5912e18b7 : SUCCESS in 48m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbdda643fba74501946bcb5b8263ab34 : SUCCESS in 1h 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf4ef2e2c0181423cd4bf87c14ed2f5efad2e9be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 14:22:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230fb7681d734f72b9c8372f8b3a845c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79bb2f8651c44e0d8c11c271a913efd0 : SUCCESS in 45m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56a8b8bdd7cc4fe3b63df847787bea47 : SUCCESS in 43m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31039a68fc2162d9298538a07aa17334aeff3b84","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-16 16:44:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e2b4ae33fced28d6231713b0fa8b15e11a1580cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 16:45:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8ef1ed6d373434a425f52dcaf5e86e26c482822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 22:35:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecd308e651ec4cc79ab442ba4dd16698\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/8f9900d73f1e4f558e5a3379de72ec71 : SUCCESS in 55m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b27ccadd76d48a4b9dae1ba144f135c : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70ef7c34a8ad457882fef3c3586c04bc : SUCCESS in 25m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/722228c062584797a0617c1bac1f1b35 : SUCCESS in 22m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffa19c691674891ad4491535383037c : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/820d479c69c048e6ab4fad308e18b63e : SUCCESS in 52m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d7f850ee186452e9b6d6b3d1367a9a8 : SUCCESS in 1h 18m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e2cfd827dbe40c88956d48a6c964e90 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"015d1d7a7b73894a1e5e21a37ae97613560f4fdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 22:35:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a24c0efea9a2ce6eebf7a8f8575b1456e2ac6af1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 22:44:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4987b02c28084ee08a71864dd8eac5e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caaef17d85704f40902a6be96911ed08 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efc3ff75a4ec89dc2d5cc4d27fe4c7fc6139cd54","revisions":{"efc3ff75a4ec89dc2d5cc4d27fe4c7fc6139cd54":{"kind":"REWORK","_number":1,"created":"2023-10-16 12:19:58.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/96/898296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/898296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/898296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/898296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/898296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/898296/1"}}},"commit":{"parents":[{"commit":"2135a0c9dc91578dc5a0e130a0f500d9b7bb520d","subject":"Merge \"[OVN] Add the default condition check in ``PortBindingChassisEvent``\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2135a0c9dc91578dc5a0e130a0f500d9b7bb520d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-06 08:58:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-16 12:19:58.000000000","tz":0},"subject":"Add dhcpagentscheduler API extension to the ML2/OVN extensions","message":"Add dhcpagentscheduler API extension to the ML2/OVN extensions\n\nIn most typical use cases ML2/OVN backend don\u0027t needs to run DHCP agent\nas OVN provides DHCP functionality natively. But there are some use\ncases like Baremetal provisioning over IPv6 or Spine Leaf architecture\nwith DHCP relays where DHCP agent is necessary and it can work perfectly\nfine with ML2/OVN backend.\nThe problem was that dhcpagentscheduler API extension wasn\u0027t listed as\nsupported by the OVN backend so it was filtered out from the list of\nsupported extensions during start of the neutron server. This caused\nproblems with API to get/set/delete networks to/from DHCP agent.\n\nThis patch adds this API extension to the list of the extensions\nsupported by the OVN driver to fix that issue.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/898090\n\nCloses-bug: #2038655\n\nChange-Id: I09a37ca451d44607b7dde344c93ace060c7bda01\n(cherry picked from commit f006d29251abe3c138ee0dea5b549496b66b8fa7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efc3ff75a4ec89dc2d5cc4d27fe4c7fc6139cd54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/efc3ff75a4ec89dc2d5cc4d27fe4c7fc6139cd54"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
