)]}'
{"id":"openstack%2Fneutron~898567","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I2f8130dc3cf3244be2a44a4ecbdbaa9c7f865731","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/2034540-stable/victoria","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-18 08:37:57.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"I2f8130dc3cf3244be2a44a4ecbdbaa9c7f865731","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","status":"MERGED","created":"2023-10-17 14:14:12.000000000","updated":"2023-10-18 11:56:30.000000000","submitted":"2023-10-18 11:55:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898567-bug/2034540-stable/victoria","cherry_pick_of_change":894012,"cherry_pick_of_patch_set":1,"meta_rev_id":"9cb4efd7ccc14a1d8931382bb87a1fcbaf1b415c","_number":898567,"virtual_id_number":898567,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-18 11:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-17 15:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-10-17 15:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-17 17:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-10-17 17:08:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 14:14:12.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-10-17 15:49:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-17 15:49:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-17 16:00:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-17 17:08:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8f6c360ca5b61960d5f95234d3aed278dc3a1768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-17 14:14:12.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"8d77187d4fbfd87d795945f366d6a0d9b88de125","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-17 15:49:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e586826eed187e2c5ae7330df679535d0fd3bba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-17 15:49:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cf18d14d11930d74fa1859a27ed1eab263e3de9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 16:00:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a57ba15da0e048e9bb47008031f06a37\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f708d0f777cd4370b896e6575c532fba : SUCCESS in 37m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aff20e85eef4310b0e1a310214f926b : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5db6c7cdd44111b3433547c1ae7184 : SUCCESS in 24m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be62d19765e47f0b89dfa36c20e4b5c : SUCCESS in 30m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116d78d41e67456e8e63f2dc4c912867 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa5636d1fd0a4efdb0357d061b7c055f : SUCCESS in 1h 12m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8c459e56d454375ab67d163aa69338b : SUCCESS in 1h 40m 07s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ce3f6ff5eb53455896049b4bae7e575f : SUCCESS in 1h 39m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/98cecd942dee4191965271fb667f7a76 : SUCCESS in 1h 08m 22s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/933f80504d4145048896fbc61af41c9c : SUCCESS in 1h 10m 18s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/3bc498b8e0d345d1aa54439f83268554 : SUCCESS in 1h 19m 04s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/008d92c007be42578f980610605a560e : SUCCESS in 1h 00m 32s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/3171e9edf24742648ca861fc6f127940 : SUCCESS in 45m 10s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/71de013cb29944b1aad0e1d6e302f2c4 : SUCCESS in 1h 05m 16s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/339fa711349f4128ab944f9481e7913c : SUCCESS in 1h 25m 58s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/084ecaae22514c5fa9006d9ffe2c2f82 : SUCCESS in 58m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee145e7b7eae440bb0441637775c16fc : SUCCESS in 45m 07s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54514743176043f8a56e273c8610f38b : SUCCESS in 1h 19m 08s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0b924ad67f146c98155e5756f16e02e : SUCCESS in 57m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a195fd82e4014a59ad25fc14f462e542 : SUCCESS in 49m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d6e612c3f41e4d44b0fc5ce8721707b8 : FAILURE in 36m 42s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e8622ac1e63d4c24b4eaa994346d0c17 : SUCCESS in 1h 14m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/98e17d31cc9447638b7928c1809a4a2f : SUCCESS in 1h 16m 59s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/113c5ac5815a4136975712ad4bf6c374 : FAILURE in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbd52a06f1af503f129ef1a1b7bfd1996ed20e72","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-17 17:08:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cd305630a43980b69f59a3e81c99f776261e627","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 17:21:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6796386f64e415eca3fd8e4462ca4fb86535fb4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 19:09:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e0f6d17c2fe45bdbd4cc87bea5cb5a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e584d8724113452790ff82e6df692ba7 : SUCCESS in 14m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0c8ecb3542c496a8d0c9eac7ad12699 : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8643539d4e984a438046eafec516abf1 : SUCCESS in 24m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08997671c3f4959989bb47ef14aa74c : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20938cd84f3d4f12a3fc6b5d6bf1fbc1 : SUCCESS in 1h 11m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3ec7e4da2744202b0056643ebae896b : SUCCESS in 1h 29m 23s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd5441308a384b10b4ead8df7ab16334 : SUCCESS in 1h 14m 01s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a84f76d1d9a9471993e85da7bb963879 : SUCCESS in 1h 10m 27s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/435cc94138c94e5bbc997325aa91e2c2 : FAILURE in 55m 55s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2b71a1d98ee8440f86e93970f1abceb8 : SUCCESS in 1h 15m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f525826829e74fd19cd5a8e65ae3239c : SUCCESS in 1h 02m 09s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/af15af75985b4dbc8d58624e1aaab767 : SUCCESS in 1h 11m 13s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/250567cad56747009b9d8609b7ac6ecf : SUCCESS in 1h 42m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/708273accb624629ad63b60e243fa8eb : SUCCESS in 45m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ffb0f909b861436aa620800a9c433db3 : SUCCESS in 1h 07m 39s","accounts_in_message":[],"_revision_number":1},{"id":"be166b999cc10a42020d4f9d2f769642f7aab42e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-18 05:03:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4005b31d6b41144c0fa166808ec21bd7a4dc90e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 06:48:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b48c5209973542cc895ff8847b47f4d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccdd5526a0c942f79a1bf8f15cca741b : SUCCESS in 34m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7620b4ed85945ed9521a48e0727a15f : SUCCESS in 12m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e37c672d2c7148e3ab935788efc41c65 : SUCCESS in 19m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93597c53f3d8466b9500c3c0d7f705e1 : SUCCESS in 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8507898318854908b1b43c03344a16cf : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7c86b54eea24b05b356ff0e609644f2 : SUCCESS in 1h 14m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/84996156675848ce81c8a32a21fbe063 : SUCCESS in 1h 19m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4c2756b32c704177b683065e2d171f18 : SUCCESS in 1h 29m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c116281e3dc7463b9174276035e3df1f : SUCCESS in 1h 16m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/28ef44440cac4a8ca0c02ee48eb48073 : SUCCESS in 1h 04m 33s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/2261deebc70d4b1282bdf443cd7dd791 : SUCCESS in 1h 15m 38s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d391f73eeed04197945507a34aac94fe : SUCCESS in 57m 53s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/836420ed7cec44b3bfb690d385d02b73 : SUCCESS in 49m 52s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/06730a1cf04a4e129f31e0af8cd054d2 : SUCCESS in 1h 14m 55s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f77c3b65bb83423a8562909180336d7f : SUCCESS in 1h 11m 37s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37e1738c4e7c4e7cb7c4a0db31b2cc9e : SUCCESS in 1h 12m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2e8c694dcd314fa6b3ff889fe852c6a3 : SUCCESS in 51m 05s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a77e0318eec1452598ad7f51682696f8 : SUCCESS in 1h 39m 44s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/777082dd51424e829d2f632029ba2d1c : SUCCESS in 57m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c6bb0d04f59e44e1b382c7fd92fd42e4 : SUCCESS in 59m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bb795dadaee141bd987a51db70847aae : FAILURE in 24m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1fdd877f12af4262aa47c3183f4df273 : SUCCESS in 1h 02m 19s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/68d44df5865b4a798f43f45f547d9c0b : SUCCESS in 1h 09m 05s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/fb2d1e62b2954bd397f081962d3154b5 : FAILURE in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bea9236f54649b543102e265078ea48ee12921e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 06:49:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb349c036ea53033ddae19eef61f40f4f82ff2d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 08:33:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3fba9d49488467e9d7acc4512f95da4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87460a511b64e07add3057705bb9879 : SUCCESS in 13m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4d3fa6fed7465f984ecdcbaf12bfd5 : SUCCESS in 28m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a294bcd772a94e1a88e86e7dff12149e : SUCCESS in 20m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2532961adc6a4b6c87f47c23d7ed6178 : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1290050717e7416db9a4bfd57b66e337 : SUCCESS in 1h 10m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/99d223c1d20340be958abf76af443e47 : FAILURE in 1h 06m 29s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/607b6b066ce74a5397e162c256d4109b : SUCCESS in 59m 53s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ba16ce75adc94538a377f52fb458a52b : FAILURE in 1h 04m 26s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aefb115c5ade4c6691aa4ce1b9416729 : SUCCESS in 1h 09m 17s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b02eaac985cb4d52ba38d6a008871f6f : SUCCESS in 1h 18m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51205fe464424861bd497b6eb4952d11 : SUCCESS in 1h 10m 44s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cd2deb53132646d6a8cb001aaeac2a6c : SUCCESS in 1h 01m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db4dd1af9c094aa1acbaa6b39b901310 : SUCCESS in 1h 37m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb379b1156ed4cb3ae9d2f87ff024635 : SUCCESS in 1h 04m 43s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/5c06c9d7b9d14e7f99a07910069fe8fc : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"aeb154692d334e3285a067d90e1aaade1c100702","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-18 08:37:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb05c546e77b4f4da4944ec2ad155aad08dbafa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 10:23:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ccd17b9f2a421aa68984faf43495e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f178126e9ec480dbc06a0c941fce32f : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4861ac426bb45fda56aab5f3cd38a1c : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a75bed732ef4e60aba98658ddffc2b6 : SUCCESS in 26m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7470b480a7c746f5a392d07b6d653f3c : SUCCESS in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b761759ec14aa59be7632e7e68c1d7 : SUCCESS in 11m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e846049023e4941968964b113f2885b : SUCCESS in 1h 08m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e24e4d440e94cf19beaca1a096a6f32 : SUCCESS in 1h 16m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/ecc15b69fc9447c9970077c7d9b94a30 : SUCCESS in 1h 26m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/21fa7316e4194fc884dd81cf028cdeb7 : SUCCESS in 1h 12m 15s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01c8841c8d3c40609dc9ac65faddc721 : SUCCESS in 1h 09m 04s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/dfc15144158d48dea28c72285a79d3b0 : SUCCESS in 59m 55s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4323c44667ee486299250e3333c8d154 : SUCCESS in 1h 09m 46s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f105c53a5ce47bf8b649c5fae23b273 : SUCCESS in 1h 07m 36s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6813b225eaca40baa8afd3f7ae8c56c1 : SUCCESS in 58m 29s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/177e0ae93fbb4d09bb26a5138a1b85a1 : SUCCESS in 1h 17m 13s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a31ec95c9dc74356bd8904aa1f2f742d : SUCCESS in 1h 15m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7be192ab284a4323ad830871807b39f3 : SUCCESS in 59m 29s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c026137a43da4d06b87139f640f154d4 : SUCCESS in 1h 13m 33s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fe564befc9040b791bda9ce4d5a6411 : SUCCESS in 1h 08m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/8cac693a2b56449ab69077b289dec17e : SUCCESS in 1h 09m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/68b3e218f1d94792b9efac0cdf6ae6bd : FAILURE in 41m 05s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6a642325b38842039caae4feb4c4fd52 : SUCCESS in 1h 16m 14s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4abe63dc6dec4434bc183d870c91e928 : SUCCESS in 1h 39m 42s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/205bb7a463064c3cb0e10613f7725999 : FAILURE in 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44c3ad4b5070b6dc870b1faf25d39bfa2ed5ac9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 10:23:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfaa3a59fafebbd39ff24d2031e4a51f3252ba7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:55:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c567ea51603414a92cb3f67a610db89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28a9b4ca9fe48a899757539711cdb00 : SUCCESS in 12m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef42e86c5c6b451fbe4b20f71e6a9dd2 : SUCCESS in 24m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21faafafa158456f89dbc01f4d4587cd : SUCCESS in 24m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ec59a16cf246659506f02c7f0b7b47 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3272836af084318a971265d14066177 : SUCCESS in 1h 11m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e530e07827064938ae5f8dd5e7029bd4 : SUCCESS in 1h 25m 17s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/471270c093914af28e0586773a2062ca : SUCCESS in 1h 06m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/70de9ce1b8b24d26b6a937750c88b7ed : SUCCESS in 50m 27s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ca03c347989f417fa78540a67c398324 : SUCCESS in 1h 11m 24s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c4ce46c70f46475e98ebc83dbf9f4216 : SUCCESS in 1h 15m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/178f92eb4b254fe7a696f6e958393bff : SUCCESS in 1h 06m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d31d5a6f53045d09d3519dbef7bc636 : SUCCESS in 57m 56s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed930df7087a4aeba4aaefe20f76541f : SUCCESS in 1h 25m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/905caffd93f74f3bbc4852a577fe09cf : SUCCESS in 1h 07m 26s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/19efcaf325fa4a618d587d1acd64f6ba : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":1},{"id":"dbb7b2168ffdb715ea71a35ec7f455f50055aff4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:55:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9cb4efd7ccc14a1d8931382bb87a1fcbaf1b415c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 11:56:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d9483e522e4fc487cce1daf7a57975\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8796105653d44cf6a6a248a6df3ec0f6 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3e4d677680d7f7cd404b40a4fb5e0e984072e89","revisions":{"e3e4d677680d7f7cd404b40a4fb5e0e984072e89":{"kind":"REWORK","_number":1,"created":"2023-10-17 14:14:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/67/898567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/898567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/898567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/898567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/898567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/898567/1"}}},"commit":{"parents":[{"commit":"8d45ee7f34f339c683c03f5391800b9848aaf11e","subject":"[OVN] ovn-db-sync check for router port differences","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d45ee7f34f339c683c03f5391800b9848aaf11e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-06 15:58:31.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-10-17 14:14:12.000000000","tz":0},"subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","message":"Call the \"tc qdisc\" command for ingress qdisc without parent\n\nThe \"tc qdisc\" command to create an ingress qdisc does not require\nto define a parent qdisc [1]. The ingress qdisc \"differs from other\nqdiscs in that it does not occupy the root of a device\".\n\nThe previous command was replacing the default root egress qdisc:\n  root@dev20:~# ip netns exec ns01 tc qdisc show\n  qdisc noqueue 0: dev dummy root refcnt 2\n\nThis operation is now not permitted with the new kernel.\n\nThis patch is also changing how the TC funtional tests interact with\nthe namespace interface created. Now the interface is set to UP before\nthe test starts. That changes the \"tc qdisc show\" command because now\nthe default qdisc (see the aforementioned CLI output) is now present,\nas it should be in a live environment (where the interfaces are UP).\n\nCloses-Bug: #2034540\n\n[1]https://tldp.org/HOWTO/Adv-Routing-HOWTO/lartc.adv-qdisc.ingress.html\n\nConflicts:\n    neutron/tests/functional/privileged/agent/linux/test_tc_lib.py\n\nChange-Id: I2f8130dc3cf3244be2a44a4ecbdbaa9c7f865731\n(cherry picked from commit 8cba9a2ee86cb3b65645674ef315c14cfb261143)\n(cherry picked from commit 37d09f95cd58143d0b2de9586158d3a6dfda800b)\n(cherry picked from commit 2179c6978d18c74fe6e58f718d842808a01eb4a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3e4d677680d7f7cd404b40a4fb5e0e984072e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e3e4d677680d7f7cd404b40a4fb5e0e984072e89"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
