)]}'
{"id":"openstack%2Fneutron~898939","triplet_id":"openstack%2Fneutron~stable%2Fzed~I6373a3c213b24ec957e4d2ea7fc42524517d10d5","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"35676":{"account":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"last_update":"2023-11-08 16:14:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6373a3c213b24ec957e4d2ea7fc42524517d10d5","subject":"Spread OVN metadata agent heartbeat response in time","status":"MERGED","created":"2023-10-20 12:40:16.000000000","updated":"2023-11-08 16:15:50.000000000","submitted":"2023-11-08 16:14:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898939","meta_rev_id":"5c05adaeb7e4a2a61cb3129dcbae69cfb4e15445","_number":898939,"virtual_id_number":898939,"owner":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 16:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-08 14:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-10-20 20:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-11-08 14:26:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 12:40:16.000000000","updated_by":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"reviewer":{"_account_id":30463,"name":"Krzysztof Tomaszewski","display_name":"Krzysztof Tomaszewski","email":"ktomaszewski@cloudferro.com","username":"labedz"},"state":"CC"},{"updated":"2023-10-20 13:52:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-20 14:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-20 20:23:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-11-08 14:26:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"4cd8485785b4ba28a3c9c390dcdecc55b0f8b9cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-10-20 12:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"505812795b844965d94868d2894f95615331cdf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-10-20 12:47:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4de69e956087c5824adcbae17202d7d108a172d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 13:52:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb68e80ec934470af185e893ffed9d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03532ad0d7fd42ae90b555a0578e30ed : SUCCESS in 43m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a5c05c6f6c84138aee916c9b024b1ef : SUCCESS in 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e508fda883aa995a82a6e3d8e0f65e80f92ba130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 14:09:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7b5a09a456e4ff290f2890c966f7a46\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/f233b18f97a84af888fe12abce0346be : SUCCESS in 1h 21m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1741bce6ae334bde89dab548e4ee0473 : SUCCESS in 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979e45a293fe4daabcbe0a7e25a9361f : SUCCESS in 13m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5395222aa9843f3b984b84c55ca0444 : SUCCESS in 28m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59779f5657054465812a8e7f5367e29e : FAILURE in 21m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf541620809b4a5ead93121972db1d26 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40aa32e40f9c4d5a82fef151be8b1a81 : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/395bf0f7849e444eb85a02519ea57245 : FAILURE in 50m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83a26091a5f2419f80f45a8e30472d8c : SUCCESS in 54m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1e087b3e73634a6f88dbd7e202c5dc73 : SUCCESS in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4826ac66c1efde2e482ab459920cbc248e863963","author":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"date":"2023-10-20 15:46:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abb21cca9f944d577b5eec156d4b9a6a0eed2e5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 16:28:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b254fd1d3613467880bff9a2201ae60d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/850e40b896e94ea496c93cd19de2a6e8 : SUCCESS in 39m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de5aec04319f4e64a7cd6c06b400cc67 : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb73063457b73c31dff9e2bd3d54f93f73cf6d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 17:54:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ab013f4da284c5dbc8707e01a4540c0\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e6750f4138934792a88f7623e9ea7959 : SUCCESS in 2h 05m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9f466c914984e148f9537f23656f6cf : SUCCESS in 26m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8a1089a33a4b54959301c8b2e44377 : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c92c0ed9274271b3bc852a6c1723e1 : SUCCESS in 23m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b2f3a29bdb04b6388414dfda8e61bee : SUCCESS in 27m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0b20bdb62074f848ad124494e267172 : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913b67dfaaee480d9737118f38a763a9 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7bbde6f8b5b849bb9bf1d5204caea8b7 : SUCCESS in 47m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48d194421c7e46af800107eeede53849 : SUCCESS in 54m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee46ea5bb9e548b194010d74aa7ac8b5 : SUCCESS in 1h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a38680a1aeb54e8334e1396d9bd070139f8646b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-20 20:23:12.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e7b5362be482a105cc867f20527f095dc073cac","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-08 14:26:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f36faaa8e06d509c9c033770f8daba70a9ab39a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:26:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c9bfc99fedb0404848d1e260f30df26a945bd84e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:14:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a722889dc6f54397b57921941ac70f6f\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b468b6874398434fbb96d1217d400af5 : SUCCESS in 1h 46m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4ef6ef22774c2881bc4c981a066589 : SUCCESS in 13m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f4a3daeb0424bbcb37178076c740e43 : SUCCESS in 24m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/733951ede59e4fe2b966f45da9b2368d : SUCCESS in 21m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385c7b4a461a490bbbb0bef45dd5b2fd : SUCCESS in 11m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f70904d2b22a42c8bedcad997bf9ef63 : SUCCESS in 52m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0986f4c2cee42ca98453dbe23860f5d : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":2},{"id":"15500373ef3a5d594ae02ee01232b45ff24671e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:14:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5c05adaeb7e4a2a61cb3129dcbae69cfb4e15445","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:15:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b8111ad1f9c4bcab6569b690c68553d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47404e51f8704d2b86b0d58404e1e5b8 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7e91a84aac506504090f02ba40c32ee4f0b6f36","revisions":{"88dc695c4ea5dd8855fb0e8f223675fb4a45e5d3":{"kind":"REWORK","_number":1,"created":"2023-10-20 12:40:16.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/39/898939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/898939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/898939/1"}}},"commit":{"parents":[{"commit":"af6726e53f6348dbdc93c7cc5502722ad2d42d7a","subject":"Merge \"[stable-only] Replace cirros image versions not cached in the CI\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af6726e53f6348dbdc93c7cc5502722ad2d42d7a"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@labedz.org","date":"2022-09-28 10:42:38.000000000","tz":0},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-10-20 12:38:57.000000000","tz":120},"subject":"Spread OVN metadata agent heartbeat response in time","message":"Spread OVN metadata agent heartbeat response in time\n\nTo avoid mass response of OVN metadata agents on\nheartbeat update - event on OVN Southbound\nSB_Global table nb_cfg entry increment, this patch postpone\nChassis/Chassis_Private table update for random number\nof seconds in range of ( cfg.CONF.agent_down_time // 2 ).\n\nRelated-Bug: #1991817\nChange-Id: I6373a3c213b24ec957e4d2ea7fc42524517d10d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88dc695c4ea5dd8855fb0e8f223675fb4a45e5d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88dc695c4ea5dd8855fb0e8f223675fb4a45e5d3"}]},"branch":"refs/heads/stable/zed"},"a7e91a84aac506504090f02ba40c32ee4f0b6f36":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-20 12:47:28.000000000","uploader":{"_account_id":35676,"name":"Max","email":"max.lamprecht@digits.schwarz","username":"lamprechtm"},"ref":"refs/changes/39/898939/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/898939/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/898939/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/898939/2"}}},"commit":{"parents":[{"commit":"af6726e53f6348dbdc93c7cc5502722ad2d42d7a","subject":"Merge \"[stable-only] Replace cirros image versions not cached in the CI\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af6726e53f6348dbdc93c7cc5502722ad2d42d7a"}]}],"author":{"name":"labedz","email":"krzysztof.tomaszewski@labedz.org","date":"2022-09-28 10:42:38.000000000","tz":0},"committer":{"name":"Max Lamprecht","email":"max.lamprecht@mail.schwarz","date":"2023-10-20 12:46:36.000000000","tz":120},"subject":"Spread OVN metadata agent heartbeat response in time","message":"Spread OVN metadata agent heartbeat response in time\n\nTo avoid mass response of OVN metadata agents on\nheartbeat update - event on OVN Southbound\nSB_Global table nb_cfg entry increment, this patch postpone\nChassis/Chassis_Private table update for random number\nof seconds in range of ( cfg.CONF.agent_down_time // 2 ).\n\nRelated-Bug: #1991817\nChange-Id: I6373a3c213b24ec957e4d2ea7fc42524517d10d5\n(cherry picked from commit 628442aed7400251f12809a45605bd717f494c4e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7e91a84aac506504090f02ba40c32ee4f0b6f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7e91a84aac506504090f02ba40c32ee4f0b6f36"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
