)]}'
{"id":"openstack%2Fneutron~898944","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I63b0641f6b7ef0e1190f421a90619bb2971d0d44","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2020813","attention_set":{},"removed_from_attention_set":{"19234":{"account":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"last_update":"2023-12-05 16:49:18.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-11-28 14:49:59.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I63b0641f6b7ef0e1190f421a90619bb2971d0d44","subject":"Don\u0027t set port capabilities for OVS HW offloading","status":"MERGED","created":"2023-10-20 14:59:07.000000000","updated":"2023-12-05 16:50:50.000000000","submitted":"2023-12-05 16:49:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898944-bug/2020813","cherry_pick_of_change":898556,"cherry_pick_of_patch_set":2,"meta_rev_id":"6aee49935a4724e53f28d45ddfa6d4ed79746773","_number":898944,"virtual_id_number":898944,"owner":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 16:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-28 14:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-08 21:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-11-28 14:49:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 15:12:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 22:47:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-11-08 15:38:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"1f1537054e749aadf721659b9a3317219ccbc952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-10-20 14:59:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1007eab3369aaa57e385f0643dfd8fb92dc7738a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-10-20 15:00:17.000000000","message":"Topic cherrypick-bug/2020813-06atyb0s0k removed","accounts_in_message":[],"_revision_number":1},{"id":"3926a0fca72104b087add3db56afcaa62f7b38e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-10-20 15:00:24.000000000","message":"Topic set to bug/2020813","accounts_in_message":[],"_revision_number":1},{"id":"096c092e71d466af8089d7bbd91471290116553b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 15:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fbc2ae3d904dd0bd1cb7f3b3fbe488\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee82cee2ded41a1995e828423441cd7 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eff2f6bdbdd4b6ab72b3b9017ab4f26 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8119cd32bfc21dcec44b07feb04f79d05cc52290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-10-24 11:41:47.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c74e58037b02974f19ee9327158552d339600cab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 11:59:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c125cefd66544f648f97fd434501860e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e956a5f315444edba7bf20fe76c12d0 : SUCCESS in 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f02b8ccf054e438274b3801e737ebd : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":2},{"id":"313171b53ab7bb14bd8941b5ec3ed399789299ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 22:47:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fc70bfc77884638ef1c72522240cfaa77c8aadcb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-08 15:38:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95f12c174f4bfed9e8a619ef4e5440b8637772f5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-08 15:42:06.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23fc50eaa89115e5665c90ca9d04b004001a9fc9","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-11-08 20:19:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b696bbb9d1b13c1b15395e99a1d8770ebae6e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-11-08 20:22:00.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6b4a36b16d2a0a07eabe3f94b81f3dda3efdacb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 20:58:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0551ad0d3948359b161617fb15e27d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4982b2e3a6476aa1d5841918489112 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0480fd661c48a0a6daf4c1062915c7 : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":3},{"id":"873e74af3deffa96e956f0196874447beb736684","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-08 21:05:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4f801a628e95ff00c522851bd70076e31a1cbe30","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-28 14:49:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98e5fc457fe7ff650b7cf456c740f45512ab2073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:00:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7b3199778fe8b030042491178b427698060d2326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:49:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3548df165bc74deb99753fff9ed947ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ede6d23bb347bbad2fbc1017bf691d : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d258f4f9e894e2ea6e8da15c71cf60f : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":3},{"id":"80119b7df4e2bd224d0f0bc554fd86b3cf24b4af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:49:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6aee49935a4724e53f28d45ddfa6d4ed79746773","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:50:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ffde699f2e946b08f54c78f597ae7bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccaa9dddbc194868a725eec0c2b87905 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba8335375a28aa51a91a90f3ce855996915bfba5","revisions":{"4d9fd78e4cdd8fee510b9cb66eb939a24e7e33f1":{"kind":"REWORK","_number":1,"created":"2023-10-20 14:59:07.000000000","uploader":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"ref":"refs/changes/44/898944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/898944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/898944/1"}}},"commit":{"parents":[{"commit":"6283440dd1e433bb0a285d5616ce695daad6920a","subject":"Merge \"Add dhcpagentscheduler API extension to the ML2/OVN extensions\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6283440dd1e433bb0a285d5616ce695daad6920a"}]}],"author":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-10-17 14:34:27.000000000","tz":120},"committer":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-10-20 14:59:07.000000000","tz":0},"subject":"Don\u0027t set port capabilities for OVS HW offloading","message":"Don\u0027t set port capabilities for OVS HW offloading\n\nNova will automatically translate VF capabilities to Neutron\nport binding profiles after patch [1] will be merged. Existing\nrecommendations in \"admin/config-ovs-offload.html\" should be\nupdated: there is no need to define capabilities in port\nbinding profiles for new ports anymore.\n\n[1] https://review.opendev.org/c/openstack/nova/+/884439\n\nRelated-bug: #2020813\nDepends-on: https://review.opendev.org/c/openstack/nova/+/884439\nChange-Id: I63b0641f6b7ef0e1190f421a90619bb2971d0d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d9fd78e4cdd8fee510b9cb66eb939a24e7e33f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d9fd78e4cdd8fee510b9cb66eb939a24e7e33f1"}]},"branch":"refs/heads/stable/2023.1"},"0e8c6706406c6475fad2738928ef94f839eeb25d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-24 11:41:47.000000000","uploader":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"ref":"refs/changes/44/898944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/898944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/898944/2"}}},"commit":{"parents":[{"commit":"6283440dd1e433bb0a285d5616ce695daad6920a","subject":"Merge \"Add dhcpagentscheduler API extension to the ML2/OVN extensions\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6283440dd1e433bb0a285d5616ce695daad6920a"}]}],"author":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-10-17 14:34:27.000000000","tz":120},"committer":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-10-24 11:41:47.000000000","tz":0},"subject":"Don\u0027t set port capabilities for OVS HW offloading","message":"Don\u0027t set port capabilities for OVS HW offloading\n\nNova will automatically translate VF capabilities to Neutron\nport binding profiles after patch [1] will be merged. Existing\nrecommendations in \"admin/config-ovs-offload.html\" should be\nupdated: there is no need to define capabilities in port\nbinding profiles for new ports anymore.\n\n[1] https://review.opendev.org/c/openstack/nova/+/884439\n\nRelated-bug: #2020813\nDepends-on: https://review.opendev.org/c/openstack/nova/+/884439\nChange-Id: I63b0641f6b7ef0e1190f421a90619bb2971d0d44\n(cherry picked from commit 3fc8d32383feb325f090af5866cad75e0b564e36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e8c6706406c6475fad2738928ef94f839eeb25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e8c6706406c6475fad2738928ef94f839eeb25d"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"},"ba8335375a28aa51a91a90f3ce855996915bfba5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-08 20:22:00.000000000","uploader":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"ref":"refs/changes/44/898944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/898944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/898944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/898944/3"}}},"commit":{"parents":[{"commit":"6283440dd1e433bb0a285d5616ce695daad6920a","subject":"Merge \"Add dhcpagentscheduler API extension to the ML2/OVN extensions\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6283440dd1e433bb0a285d5616ce695daad6920a"}]}],"author":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-10-17 14:34:27.000000000","tz":120},"committer":{"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","date":"2023-11-08 20:22:00.000000000","tz":0},"subject":"Don\u0027t set port capabilities for OVS HW offloading","message":"Don\u0027t set port capabilities for OVS HW offloading\n\nNova will automatically translate VF capabilities to Neutron\nport binding profiles after patch [1] will be merged. Existing\nrecommendations in \"admin/config-ovs-offload.html\" should be\nupdated: there is no need to define capabilities in port\nbinding profiles for new ports anymore.\n\n[1] https://review.opendev.org/c/openstack/nova/+/884439\n\nRelated-bug: #2020813\nDepends-on: https://review.opendev.org/c/openstack/nova/+/898945\nChange-Id: I63b0641f6b7ef0e1190f421a90619bb2971d0d44\n(cherry picked from commit 3fc8d32383feb325f090af5866cad75e0b564e36)\n(cherry picked from commit 7daf61a0af7327eb46a29805cc93773aa8cd16fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba8335375a28aa51a91a90f3ce855996915bfba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba8335375a28aa51a91a90f3ce855996915bfba5"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
