)]}'
{"id":"openstack%2Fneutron~899052","triplet_id":"openstack%2Fneutron~master~If2875caf9d14f1d13cee7e40b737e573252f0a7a","project":"openstack/neutron","branch":"master","topic":"bug/2015065","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-15 10:24:49.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-01-12 06:33:33.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-01-15 10:24:49.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-01-15 08:40:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-15 10:24:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2875caf9d14f1d13cee7e40b737e573252f0a7a","subject":"Revert \"Disable mysql gather performance in jobs\"","status":"MERGED","created":"2023-10-23 21:02:38.000000000","updated":"2024-01-15 10:29:56.000000000","submitted":"2024-01-15 10:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"revert_of":883648,"submission_id":"899052-bug/2015065","meta_rev_id":"290948f09517071f52074492e805cfd6f5893144","_number":899052,"virtual_id_number":899052,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-15 10:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2024-01-12 06:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-01-15 08:40:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-01-15 08:40:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"date":"2024-01-11 22:43:35.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 21:02:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-23 21:02:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-23 21:02:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-10-23 21:02:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-23 21:02:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f9fb405b1880274a44386648c29b310e66da978","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-23 21:02:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33345f5d70edd02ff1faf35294569e7da35876c7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-23 21:03:08.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1d13222f044ab71345b9d18b7ef213f8d372603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 22:36:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04dbfd787dd94645948ea9acbdcde124\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc96f1473ab149908b57f8cda262270c : SUCCESS in 10m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcbd3b6004f405b9a446850c64ee82c : SUCCESS in 12m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa8ee823312044cdbe0387b9a24e0920 : SUCCESS in 57m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7854ab8744a74af09808b277b888f96e : SUCCESS in 56m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea3c4320ec3b493bae8e40f3ee6975fc : SUCCESS in 42m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6299131306d74dd7bf3bfb58912b8c20 : SUCCESS in 49m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3bd449737111438aa8fb9ac9e620f8c7 : SUCCESS in 1h 13m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/105ae7c18dab47f4b04e482dab042e70 : SUCCESS in 1h 05m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/053ed450f4e04a52b9a91361b53699a9 : SUCCESS in 50m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb08644dcabb4e0f8031ba9affa1e64b : SUCCESS in 1h 32m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/db5f756945fc4cc38c0c5c786f917266 : SUCCESS in 1h 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f0619ce9c721748ecb08a44f0f0967a0cbd6d5e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-24 14:46:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"046ccaa933fede0fe9866a2d4f80454d160253ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 16:33:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa25162029547c9aac5279031b045af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceb9cbf95ab64ee29cd13162a0d81f3b : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e01edfa9d7644edbd94ab824f2979ce : SUCCESS in 12m 13s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9fc611f6de2b46cd97726677bf4cfae7 : SUCCESS in 1h 04m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d4ccb45b19884eaeaabe00d918b2288d : SUCCESS in 1h 08m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d817cfc19e84a12a67f41677394325a : SUCCESS in 42m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6f443c7745844a38d8ea7af40d832a6 : SUCCESS in 51m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea4d27c02b944d6da97b52f7cf11f57f : SUCCESS in 1h 38m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8bb4c18fb104135833bb4af238e12c8 : SUCCESS in 1h 28m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ee4130e42534bf29b8edf07b623b5d5 : SUCCESS in 44m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6b5fc699a7b64390b0755f64a71541c3 : SUCCESS in 1h 44m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf525f9402434f3aa048dc31f1b28abc : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3e90276eb6e3fb9dd30e8873aa88966463434cf8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-24 18:07:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbbe8ea50b7a821fe8b4e93e36ed96e78670d0fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 19:59:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a1da4c8b864936acbf54db77a705fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2d9fe87b5c48cab791785e146980c2 : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9b8f3ff4924ccaa4bca2ecd4674fd6 : SUCCESS in 12m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/20b0de364e1841818f3ef22a6aad2769 : SUCCESS in 1h 06m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7ffb5c37351e4110967667a933216c89 : SUCCESS in 1h 10m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/18a1acc249fc4c95ba93454169bc3dd9 : SUCCESS in 1h 05m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/da3438cefccf46d790b20169d92e40f0 : SUCCESS in 1h 07m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4893da96d8554439bd805829fc2f4336 : SUCCESS in 1h 43m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb4b4a1e6128431c91bea5df74f2a860 : SUCCESS in 1h 32m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bd41414678da4b7f8b12fa0bf5b8c617 : SUCCESS in 1h 01m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/51732e1871e34c409aa4599f3067d0b8 : SUCCESS in 1h 21m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/be0473c1f3c24f2d860ad46bed145184 : SUCCESS in 1h 23m 45s","accounts_in_message":[],"_revision_number":1},{"id":"402e9986947af95174a4fe2e1e139ce351fa5585","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-25 16:09:43.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97a67bfe462d2b8174b74dbf97b607026a80b3c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 17:46:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc6d5ef2fa54989b790177e3c65130f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c03d59223d88425c8743ee5df42db530 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4df4b3e6bd4760b52309d8f16258ab : SUCCESS in 11m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5d66cca1a8a84865b8f292501e509480 : SUCCESS in 58m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/61d122fa52a54aaabe3001111589f364 : SUCCESS in 1h 01m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4b9d435b63b4e6b83d8e60e2b42a058 : SUCCESS in 1h 01m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6319f0e847bc468083ad70cecb6206cd : SUCCESS in 58m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd9b53bf5f1b4de886de496bef4c3349 : SUCCESS in 1h 27m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d219e535ecd4179ad9a5132dd723a8a : SUCCESS in 1h 28m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa8c6387eea041dea200e82881a3afba : SUCCESS in 50m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4378adebaf9746119f5f8574253744a8 : SUCCESS in 1h 28m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84d1250bb0c84141aee411710e7e66fb : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6dc4c421cefeb5e37d1ed8801b516275abef87a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-11 22:43:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea3c58bc8f6117c21a72ddb2bc42a8d71b598538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 00:52:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2404a1b8594f818a78f6e2123bef3d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b4c1f87428f46ea9ea981dc0b0d704f : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb222143093e461dbcf7a9bebb9287a5 : SUCCESS in 10m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e28f16af7ce4441fb45c8c9af811c60e : SUCCESS in 46m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cd92cc1a8f184d26a9b05bc108910f33 : SUCCESS in 51m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e9d396ed22c4caf9436ee459f151fa4 : SUCCESS in 1h 01m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c24af989e5c34bb8a681911234f44b3f : SUCCESS in 48m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dbc43e0f80254615b6647e527f82a722 : SUCCESS in 1h 27m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f3e9457cc5924842903d3e8e2f1f79bc : SUCCESS in 1h 23m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fbf9274c2cf843c281afc841bd5a1629 : SUCCESS in 56m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/93e8fcd95e7e4fc1b7585689abcb8c5c : SUCCESS in 1h 24m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8c1fb62bdab4b9983462675aaab6ca7 : SUCCESS in 2h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"fc1f77751eebb0c231fd8920cb56f0e42f2460ef","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-12 06:33:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c696122939df90cd863c0ebc9ade44d139bf562f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-15 08:40:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"342a9875c3a51c46f990cdb78b0925cd4f8fd5b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 08:40:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe68710ce2be62676c3c30606f0688757598d9ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 10:24:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15fb8bfc2db49ed881061483262fd37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571ab31441db4cd281d3638c7f90af94 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374d7cb5125641d3a39a09598a811a4e : SUCCESS in 11m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3706dccbb940435b828aab1bc91d1467 : SUCCESS in 1h 27m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c07aa00d0cf4c8fb336952b6a01e784 : SUCCESS in 1h 01m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60a3069595f1481e83ec2f9962e7b107 : SUCCESS in 1h 13m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d74e871290fd4b2b83aa024088299db1 : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":1},{"id":"98cbb3dcf57ed93c19aed0cf904c9a4e2541f134","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 10:24:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"290948f09517071f52074492e805cfd6f5893144","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 10:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab71c7fecd440dcad50c5886e07c0e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d86754bc9b44cd81274e5691290704 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f0a35c28f90d7aa8942d5d71dcda5b6055f6125","revisions":{"4f0a35c28f90d7aa8942d5d71dcda5b6055f6125":{"kind":"REWORK","_number":1,"created":"2023-10-23 21:02:38.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/52/899052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/899052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/899052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/899052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/899052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/899052/1"}}},"commit":{"parents":[{"commit":"1d0335810d89ede47cf3b54614382cb25d1986ae","subject":"Disable mysql gather performance in jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d0335810d89ede47cf3b54614382cb25d1986ae"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-23 21:02:38.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-23 21:02:38.000000000","tz":0},"subject":"Revert \"Disable mysql gather performance in jobs\"","message":"Revert \"Disable mysql gather performance in jobs\"\n\nThis reverts commit 1d0335810d89ede47cf3b54614382cb25d1986ae.\n\nReason for revert: So we can look at a neutron DB\nperformance issue with devstack, see vPTG etherpad\nfor more info.\n\nChange-Id: If2875caf9d14f1d13cee7e40b737e573252f0a7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f0a35c28f90d7aa8942d5d71dcda5b6055f6125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f0a35c28f90d7aa8942d5d71dcda5b6055f6125"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
