)]}'
{"id":"openstack%2Fneutron~899069","triplet_id":"openstack%2Fneutron~master~I426ab821e9adb5ef340938df3de14c5aa3301c64","project":"openstack/neutron","branch":"master","topic":"bug/2040166-ovn-sb-db-sync-mode","attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2025-07-15 22:10:32.000000000","reason":"Vote got outdated and was removed: Workflow-1"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2023-10-23 15:25:57.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2023-10-23 15:25:57.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-12-20 22:11:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-07-15 22:10:47.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I426ab821e9adb5ef340938df3de14c5aa3301c64","subject":"[OVN] OVN SB DB sync to account for neutron_sync_mode value","status":"NEW","created":"2023-10-23 14:08:32.000000000","updated":"2025-07-24 20:23:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f146d57aa3c68097cf58c7ae392ae3559545bdf9","_number":899069,"virtual_id_number":899069,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2025-07-24 20:23:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-16 00:02:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 14:25:50.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2023-10-23 14:26:14.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2023-10-23 14:35:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-23 14:38:29.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 15:03:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-23 16:39:57.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-23 17:58:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 18:22:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-12-20 22:11:50.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REMOVED"}],"messages":[{"id":"50240d86ee129686260a1fd5c6ed9514308dc485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 14:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b962f44e3542defa8c89e4b670bda70925be3e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 14:11:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fb42afef1193b96f1aab092467c31c148cdd0f4e","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-10-23 14:25:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e183f743c255b548543bccfb9745cd9a6833be2e","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-10-23 14:26:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25857e16e285a414ddbd2066cdc6c8e34d79075e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-23 14:35:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97a1982dc4ddc064d5427312a610b9cc4fe85679","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 14:37:50.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"437491797a91e01d42e02a28ef27e6284316c1da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-23 14:38:29.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/69/899069/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d6f3e70da9237c65ab777dc48cb254cbdacec273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:03:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b069a221f6a64feba8c494c5676adde6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3146d11088ff46dab1d99c47047b0b0a : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a230280b9e5d479ab9b486fd4c12b0fe : SUCCESS in 50m 27s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f254a291c7741ceb9c06e4fe0f406c9 : SUCCESS in 27m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1efdc4babc7465c803aa7da0a23173f : SUCCESS in 46m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"afb82e87e390e3f88280e5b0b6e664d2bab022f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-23 15:07:58.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/69/899069/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"4f9955a47196535fb3e6cd95d0e2242b9fa39905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 15:25:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c12dc5a0607296a7321ca6a00238d66db0938cae","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 15:26:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f1ea238221740d78d65e78342789c1a181bce53","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-23 15:55:46.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/69/899069/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fc488f999931015b46e07cd0abf33dc95a31c86e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 16:24:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec66d6d92ae745fb99571728a99e0311\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/565c47772b324ff8877452922e4a0cc3 : SUCCESS in 41m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/45c244f7eb804895a66f0f4845f91632 : SUCCESS in 19m 24s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d49242821fda47c88b10ee80f27d4a25 : SUCCESS in 56m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/80ce3a987db640c8a6cde87cbc725b1d : SUCCESS in 18m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b06b20dae886564b302761a3fac9b413692b982b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 17:58:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807f94a13ad946b6af2ccfc8d8ec29ef\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86082c10ac554724ae9ec82f0fd45277 : SUCCESS in 1h 57m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d72f9647d02d47f183924abb09cff0bc : SUCCESS in 42m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f02304da0b5947a18700a9c47284f91a : SUCCESS in 10m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ae21c2f76534cc190a3e06085393d1f : SUCCESS in 25m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d48b6c81d75740f1bf78fa671f6c95e2 : SUCCESS in 19m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bc2867c916f4179b60970b5d071c616 : SUCCESS in 20m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfe3ff565d0f4abe99ff9fd93ccc8108 : SUCCESS in 24m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfbe24bacb14838b795144f64510d04 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e9448982ddf468a804e7bdff7a2ddc2 : SUCCESS in 17m 37s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7145efa0a4db4eb3acc9fdb706b6e9cd : SUCCESS in 30m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/528dbd62e1b548b3a86d3be5f929d335 : SUCCESS in 58m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdb4898e42ea42638abc9dac9eaa376c : SUCCESS in 1h 00m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/139e3937b307412d8893817d850c400b : SUCCESS in 2h 26m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb25b5cdc2254559aca34ff407e2ffed : SUCCESS in 1h 26m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d3a2d1acc51142c39bb27ccafaa16dc9 : SUCCESS in 2h 01m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a87974fac1484c4c9e2a98d4b5dd0d8f : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":3},{"id":"a23c3472c287f117bd467658add861b722e28445","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-23 18:22:21.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a30e235cf3b0d2fd7b357fa7a0245b0d6766e654","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-24 08:19:14.000000000","message":"Patch Set 3: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6a6d46a946823c72065e68105a6f094d4db38332","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-31 21:29:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"76cc66ac9e4cf8ce2ae2f9247a496c0b3b174d09","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-12 21:26:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7293aac3cc12d2a4e1a0f6da919c8f420529f346","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-12-20 22:11:50.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_9656\u003e.","accounts_in_message":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"_revision_number":3},{"id":"d92ec20ab4695028af70a49a46799523d17f975c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-15 22:10:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"12160c6325ec6098b56450b8041857a220c5cb7b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-07-15 22:10:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c174f0b57c4b920a25f34face70d1e73e72917e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-15 22:47:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72502247d2544fc83b69e9eeef501f8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/90810e80544a48d3896f5f99d3a89465 : SUCCESS in 29m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f54e016e2c3d4363a3a199d00ba974de : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebaaa144714d499dbb123683f937f38d : SUCCESS in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c6f06f9e6d05f10a9218a731c38a55071e7478f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 00:02:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/325c58fe7d9a4b19b62e68e08488dda5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de28434716f04ca88169e5027f3b3018 : SUCCESS in 1h 50m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/919448489cac49329aadfe9622272ab7 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d9660da8fe4c0a9e3700da677f4db1 : SUCCESS in 11m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adb632d6f1e6434980d74865dc4b804c : SUCCESS in 22m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f84a651c876e4b51a0f4a4ab062c83fe : SUCCESS in 16m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96a29f5f8244479ebbfcd74b52585051 : SUCCESS in 22m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3031b357226c4a6a89182ad7e1d2855e : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8b078c61d74709a5e8babae67e554f : SUCCESS in 11m 51s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c458f72afbb145379dd9f268fc64fde6 : SUCCESS in 1h 45m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d49fbd638544a89ba9e7e8123aff4f1 : SUCCESS in 1h 09m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfef173fe98b40c5ad79cf53fe2e708b : SUCCESS in 53m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bab79b653cfd482e9dfb13ca2aa3650f : SUCCESS in 1h 33m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e1fef637f5244900a7606b6c901129a8 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":4},{"id":"f146d57aa3c68097cf58c7ae392ae3559545bdf9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2025-07-24 20:23:01.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2025/Arista_Tempest_Testrefs/changes/69/899069/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15a668abebf5bc18ae2acf5a8107bee6aa8a37b5","revisions":{"1b5d6dda3b7f0fdb3e509283d6110a5151de9255":{"kind":"REWORK","_number":1,"created":"2023-10-23 14:08:32.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/69/899069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/899069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/899069/1"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 13:46:20.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 14:08:22.000000000","tz":60},"subject":"[OVN] OVN SB DB sync to account for neutron_sync_mode value","message":"[OVN] OVN SB DB sync to account for neutron_sync_mode value\n\nPrior to this patch, the OVN SB DB sync code was ignoring the value of\nthe neutron_sync_mode configuration option and always making changes to\nthe OVN DB database even if the configuration option was set to \"off\" or\n\"log\" (default value, only log the inconsistencies but do not change the\ndatabase).\n\nChange-Id: I426ab821e9adb5ef340938df3de14c5aa3301c64\nCloses-Bug: #204016\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b5d6dda3b7f0fdb3e509283d6110a5151de9255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1b5d6dda3b7f0fdb3e509283d6110a5151de9255"}]},"branch":"refs/heads/master"},"f727ecebdae5eee867d95bfb07ffb395519dcda8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-23 14:11:01.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/69/899069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/899069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/899069/2"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 13:46:20.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 14:10:43.000000000","tz":60},"subject":"[OVN] OVN SB DB sync to account for neutron_sync_mode value","message":"[OVN] OVN SB DB sync to account for neutron_sync_mode value\n\nPrior to this patch, the OVN SB DB sync code was ignoring the value of\nthe neutron_sync_mode configuration option and always making changes to\nthe OVN DB database even if the configuration option was set to \"off\" or\n\"log\" (default value, only log the inconsistencies but do not change the\ndatabase).\n\nChange-Id: I426ab821e9adb5ef340938df3de14c5aa3301c64\nCloses-Bug: #2040166\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f727ecebdae5eee867d95bfb07ffb395519dcda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f727ecebdae5eee867d95bfb07ffb395519dcda8"}]},"branch":"refs/heads/master"},"165cef40481a85116808a06a37fad3010f687046":{"kind":"REWORK","_number":3,"created":"2023-10-23 15:25:57.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/69/899069/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/899069/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/899069/3"}}},"commit":{"parents":[{"commit":"73cd6dca7442a33ac96fed157d41a4b5ec99915f","subject":"Update OVN db sync code log messages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73cd6dca7442a33ac96fed157d41a4b5ec99915f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 13:46:20.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 15:16:00.000000000","tz":60},"subject":"[OVN] OVN SB DB sync to account for neutron_sync_mode value","message":"[OVN] OVN SB DB sync to account for neutron_sync_mode value\n\nPrior to this patch, the OVN SB DB sync code was ignoring the value of\nthe neutron_sync_mode configuration option and always making changes to\nthe OVN DB database even if the configuration option was set to \"off\" or\n\"log\" (default value, only log the inconsistencies but do not change the\ndatabase).\n\nChange-Id: I426ab821e9adb5ef340938df3de14c5aa3301c64\nCloses-Bug: #2040166\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/165cef40481a85116808a06a37fad3010f687046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/165cef40481a85116808a06a37fad3010f687046"}]},"branch":"refs/heads/master"},"15a668abebf5bc18ae2acf5a8107bee6aa8a37b5":{"kind":"REWORK","_number":4,"created":"2025-07-15 22:10:32.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/69/899069/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/899069/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/899069/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/899069/4"}}},"commit":{"parents":[{"commit":"0b23123a09f0d93fa5bc7c7a5b0c730db53dd9a5","subject":"Merge \"trivial: Autofix low-hanging linter failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b23123a09f0d93fa5bc7c7a5b0c730db53dd9a5"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-23 13:46:20.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-07-15 22:04:25.000000000","tz":-240},"subject":"[OVN] OVN SB DB sync to account for neutron_sync_mode value","message":"[OVN] OVN SB DB sync to account for neutron_sync_mode value\n\nPrior to this patch, the OVN SB DB sync code was ignoring the value of\nthe neutron_sync_mode configuration option and always making changes to\nthe OVN DB database even if the configuration option was set to \"off\" or\n\"log\" (default value, only log the inconsistencies but do not change the\ndatabase).\n\nChange-Id: I426ab821e9adb5ef340938df3de14c5aa3301c64\nCloses-Bug: #2040166\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a668abebf5bc18ae2acf5a8107bee6aa8a37b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a668abebf5bc18ae2acf5a8107bee6aa8a37b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
