)]}'
{"id":"openstack%2Fneutron~899109","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2038413","attention_set":{},"removed_from_attention_set":{"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2024-06-12 13:39:49.000000000","reason":"Change was submitted"},"25468":{"account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"last_update":"2024-06-12 13:39:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc","subject":"[OVN] Update lsp host id when virtual parent moves","status":"MERGED","created":"2023-10-24 15:45:21.000000000","updated":"2024-06-12 13:41:04.000000000","submitted":"2024-06-12 13:39:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899109-bug/2038413","meta_rev_id":"fe52502e0e1afd2491e13d70d5b8f57441a6e4b1","_number":899109,"virtual_id_number":899109,"owner":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-12 13:39:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-06-12 11:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-06-12 09:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-06-12 09:17:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-06-12 11:22:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 08:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-12 09:29:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-06-12 10:18:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 11:22:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d47ab34c73e908c30b21e9699ea770cad2bf3302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 15:45:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9a3491307ff46e2763c59d3c8249a42ed2a5eea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2023-10-24 15:45:58.000000000","message":"Abandoned\n\nduplicate, wrong review.","accounts_in_message":[],"_revision_number":1},{"id":"28fe71f596e3ce30a78aa41ecbd35bd2009acfcf","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 07:50:05.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e5fa0bf16def7866b6653ae881552b3f0e718c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 07:50:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_25468\u003e","accounts_in_message":[{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"_revision_number":2},{"id":"e54a7e5db028b4286fd0f796e49ab8a2e0464cfd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 07:51:26.000000000","message":"Topic tb/stable/2023.2 removed","accounts_in_message":[],"_revision_number":2},{"id":"5374ea5dd51cb42785177d8b01f2109111101be3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 07:51:32.000000000","message":"Topic set to bug/2038413","accounts_in_message":[],"_revision_number":2},{"id":"19ee93bc7c169b102f2523e84f94d117a0dac52c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2024-06-12 08:12:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60edf5d581624e5ea396e260c9d79c25b9c337ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 08:46:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b721f3600149d98606b615ad56d1a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a197c923c854d5a89c975f2a37980b9 : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fd5c6d9f8f947f09d8830902ec6388f : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76e914fd525c4b989b72c92f4244a938 : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d2ac52356dec410e85c4fc7f885daa82 : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"232aa70c7f7121e77db0e61e5b7d14d8a1f9c773","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2024-06-12 09:17:09.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"065f3ddfb6449b34666b866854821a99c2f0ddcf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-06-12 09:29:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e8a6ed7dafa329d13988e46f792bb78b3c78d909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:18:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e12b63d538446197273a2f49f879c2\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/29be3f8222a5442ebf2de2ac63222d9d : SUCCESS in 2h 03m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a24edc773f44553ae6ab03dd80d30ea : SUCCESS in 37m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b76bd5bb1934c87891e7b12bc6a894b : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c31e4d70f6fd4b66902c4a4e3327150f : SUCCESS in 26m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2771b2da1fd40719b4e996ad45ea6c6 : SUCCESS in 27m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04a9fff9305c44dc814ec1fc8fc91aaa : SUCCESS in 24m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cec6aa75fa9f4278b98a9cd1b9dbc78b : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c6b6e433ff45129a23aa164b294768 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/19eabc820c594223a2c71db11910121c : SUCCESS in 47m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b7bd0083642e45269c7a2feff3e2a771 : SUCCESS in 52m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6bcd506057c428cb67a3d25cec8e81b : SUCCESS in 1h 10m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/73a4eb496c0743b991a186a9b7a6cd12 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":3},{"id":"2296e9313622e729f7e75608d29a0d428564f40d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-06-12 11:22:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ee1a3fe2ecf4c67f851cdfe3d4e27fca4da54266","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 11:23:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"066684ea4bc8d99c15d8a2c23ba2c850789da0aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:39:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693db5e2ed664a08b3b9412a8ae51556\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/fa92ff666bd14a928d087f43da446b20 : SUCCESS in 2h 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4920c484bf49420384e24ee7a2362d39 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/543f432263a54c438b45742c1500da5f : SUCCESS in 27m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0f039094919468d87e7abb3188d51c2 : SUCCESS in 29m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9485c72bad834ec29b8c38dd04b6621d : SUCCESS in 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224a772bae3240409a93e4f960de1961 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/649d92c485484267aa735cf2b9ea6759 : SUCCESS in 50m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ab308adb5194f5a889df07975372229 : SUCCESS in 54m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f310187359b41eba26726972f974030 : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":3},{"id":"bbe1b07aa237625c7c5bf065746ceaef8d69b68b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:39:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"fe52502e0e1afd2491e13d70d5b8f57441a6e4b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:41:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c95a2165b164901b22753fb42b2851b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f51d632791b46d8b5344e55f4da08d7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88419c31d74f88cc72c891c90c10f3877f01d3ed","revisions":{"7675ee4dc04525cf8affb5ac9dbb04629d938fd2":{"kind":"REWORK","_number":1,"created":"2023-10-24 15:45:21.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/09/899109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/899109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/899109/1"}}},"commit":{"parents":[{"commit":"037bbe5c540253595bdd64237e804161ec4e7966","subject":"Merge \"\"ebtables-nft\" MAC rule deletion failing\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/037bbe5c540253595bdd64237e804161ec4e7966"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-24 15:45:02.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7675ee4dc04525cf8affb5ac9dbb04629d938fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7675ee4dc04525cf8affb5ac9dbb04629d938fd2"}]},"branch":"refs/heads/stable/2023.2"},"8514867863d7da24c44b88b776c82a6d68d9661d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-12 07:50:16.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/09/899109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/899109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/899109/2"}}},"commit":{"parents":[{"commit":"4c56dfded153b9c6412bafd65bd76d8e5d0bfe2c","subject":"[stable only] Fix KeyError in set_gateway_mtu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c56dfded153b9c6412bafd65bd76d8e5d0bfe2c"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2024-06-12 07:50:16.000000000","tz":0},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\n\nRelated-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8514867863d7da24c44b88b776c82a6d68d9661d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8514867863d7da24c44b88b776c82a6d68d9661d"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"},"88419c31d74f88cc72c891c90c10f3877f01d3ed":{"kind":"REWORK","_number":3,"created":"2024-06-12 08:12:03.000000000","uploader":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"ref":"refs/changes/09/899109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/09/899109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/09/899109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/09/899109/3"}}},"commit":{"parents":[{"commit":"4c56dfded153b9c6412bafd65bd76d8e5d0bfe2c","subject":"[stable only] Fix KeyError in set_gateway_mtu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c56dfded153b9c6412bafd65bd76d8e5d0bfe2c"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2023-10-04 07:31:50.000000000","tz":120},"committer":{"name":"Fernando Royo","email":"froyo@redhat.com","date":"2024-06-12 08:12:00.000000000","tz":120},"subject":"[OVN] Update lsp host id when virtual parent moves","message":"[OVN] Update lsp host id when virtual parent moves\n\nWhen a virtual port is moved from one port to another port the\nPortBindingUpdateVirtualPortsEvent event would only update the binding\nhost id in the neutron database, while it is also usefull to keep the\ninformation in the OVN database up to date with the host information\n\nOther plugins that connect to the OVN database can then also rely on the\ninformation stored in the OVN DB\u0027s\n\nCloses-Bug: #2038413\n\nChange-Id: I59c6c4b2c8b023b9c9c3bab1741d957fa1f738fc\n(cherry picked from commit e68a920c114010a36a667d90866bfb6243148b6d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88419c31d74f88cc72c891c90c10f3877f01d3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88419c31d74f88cc72c891c90c10f3877f01d3ed"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
