)]}'
{"id":"openstack%2Fneutron~899227","triplet_id":"openstack%2Fneutron~master~Ic2bde217a03a884855d299f3142394a4229745bc","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-26 18:36:56.000000000","reason":"Change was submitted"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2023-10-26 18:36:56.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-26 18:27:06.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ic2bde217a03a884855d299f3142394a4229745bc","subject":"doc: Fix wrong sections of ovs-agent config options","status":"MERGED","created":"2023-10-25 07:37:35.000000000","updated":"2023-10-26 18:38:18.000000000","submitted":"2023-10-26 18:36:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899227","meta_rev_id":"d5d5480288d40bac2964de36591d2287924da1af","_number":899227,"virtual_id_number":899227,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 18:36:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-10-26 15:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-26 18:27:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-10-26 18:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 07:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 08:58:18.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2023-10-25 20:31:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-26 15:37:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"b076a065e24d6ce03813dc0f87c398afbbf2807e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-25 07:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"357be977d07ec5df8722a983bd41f0911a57b5af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-25 07:39:35.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8774d9272f6c8c405010ea9189fd0a06047329e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-25 07:42:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a23691220e1a5845d90ae2f50857b4f7b09e845d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-25 07:47:25.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3f21069f50eef5d8d87d559ec8c07903789526a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 07:59:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdaa760ac5549acb1f5d56bddbca73a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14fb0c4d3c4f42de98f43405038ea453 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277b74a562ce4029af7ac31b16692dfd : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":4},{"id":"edd853a952335ebad8bb3b25ff8905f2b4a67029","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-10-25 08:58:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"042c15cb8a662f272d87492f0c2bb06caeeace66","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-25 20:31:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"59fbd6d8c039339dcbc1d839bbdbc30a47b511ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 00:01:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9727a4d2e4f37f8cfc03372cf02b538d57293541","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-26 00:01:45.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a0e58b1a9b79f47c17402b3de69362d5975ffea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 00:14:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797dcc1d3c5b4d50991011b67e6ccd78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2678935e4884807a9bd0d0978cb264f : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0c637edf93a42a79da47016c2298b20 : SUCCESS in 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"4257f50e6ca361e55fdbd5aa04bc3ed67e1143f6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 15:37:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"597b2495a75b850699c50098e3b2ffd09de2b20c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-26 18:27:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7078ecf6ced4ffc461c7970b5266839e390de5da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:27:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c3d33accb1f5909ee939b1c78f4b809a4f5c042f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:36:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5cb29d38d924277b40a876ca9948197\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e41682aa9f401fa862874eba71d075 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264e2554f73d4147b9d7bbb97efcde26 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":5},{"id":"65ff1226057e9acad15ba2788c9e63361394ad50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:36:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d5d5480288d40bac2964de36591d2287924da1af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 18:38:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f64a82ffecd4011b936e0881fe48ae5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98157862d42a4e4babe7747735a939f8 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"34f7a5805e1fe37da899356411b8ff650e598250","revisions":{"54a9060cb1fec660696baf65561f33c972830cc8":{"kind":"REWORK","_number":1,"created":"2023-10-25 07:37:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/899227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/899227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/899227/1"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:36:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:37:03.000000000","tz":540},"subject":"doc: Fix wrong sections of ovs-agent config options","message":"doc: Fix wrong sections of ovs-agent config options\n\nThe sections described in the documentation does not match the actual\nsection names in current neutron.\n\nChange-Id: Ic2bde217a03a884855d299f3142394a4229745bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a9060cb1fec660696baf65561f33c972830cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a9060cb1fec660696baf65561f33c972830cc8"}]},"branch":"refs/heads/master"},"27ef524ed2fe8743f6bcd1232ccea141b818568f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-25 07:39:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/899227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/899227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/899227/2"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:36:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:38:37.000000000","tz":540},"subject":"doc: Fix wrong sections of ovs-agent config options","message":"doc: Fix wrong sections of ovs-agent config options\n\nThe sections described in the documentation does not match the actual\nsection names in current neutron.\n\n * local_ip now belongs to the [ovs] section\n * tunnel_types now belongs to the [agent] section\n * l2_population now belongs to the [agent] section\n\nChange-Id: Ic2bde217a03a884855d299f3142394a4229745bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27ef524ed2fe8743f6bcd1232ccea141b818568f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27ef524ed2fe8743f6bcd1232ccea141b818568f"}]},"branch":"refs/heads/master"},"7de2918fd4e46482030a8a88ee7137f2cdc1a0c2":{"kind":"REWORK","_number":3,"created":"2023-10-25 07:42:42.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/899227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/899227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/899227/3"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:36:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:42:33.000000000","tz":540},"subject":"doc: Fix wrong sections of ovs-agent config options","message":"doc: Fix wrong sections of ovs-agent config options\n\nThe sections described in the documentation does not match the actual\nsection names in current neutron.\n\n * local_ip now belongs to the [ovs] section\n * tunnel_types now belongs to the [agent] section\n * l2_population now belongs to the [agent] section\n\nChange-Id: Ic2bde217a03a884855d299f3142394a4229745bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7de2918fd4e46482030a8a88ee7137f2cdc1a0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7de2918fd4e46482030a8a88ee7137f2cdc1a0c2"}]},"branch":"refs/heads/master"},"b18394a02d03ed715499ccc7e3f6e45aa2bb788d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-25 07:47:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/899227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/899227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/899227/4"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:36:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:46:36.000000000","tz":540},"subject":"doc: Fix wrong sections of ovs-agent config options","message":"doc: Fix wrong sections of ovs-agent config options\n\nThe sections described in the documentation does not match the actual\nsection names in current neutron.\n\n * local_ip now belongs to the [ovs] section\n * tunnel_types now belongs to the [agent] section\n * l2_population now belongs to the [agent] section\n\nAlso the tunnel_types option is not explained in the example snippet.\n\nChange-Id: Ic2bde217a03a884855d299f3142394a4229745bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b18394a02d03ed715499ccc7e3f6e45aa2bb788d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b18394a02d03ed715499ccc7e3f6e45aa2bb788d"}]},"branch":"refs/heads/master"},"34f7a5805e1fe37da899356411b8ff650e598250":{"kind":"REWORK","_number":5,"created":"2023-10-26 00:01:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/899227/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/27/899227/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/27/899227/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/27/899227/5"}}},"commit":{"parents":[{"commit":"dad8c3fecc4e97764a92f558d9df510fa2516877","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dad8c3fecc4e97764a92f558d9df510fa2516877"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-25 07:36:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-26 00:00:56.000000000","tz":540},"subject":"doc: Fix wrong sections of ovs-agent config options","message":"doc: Fix wrong sections of ovs-agent config options\n\nThe sections described in the documentation does not match the actual\nsection names in current neutron.\n\n * local_ip now belongs to the [ovs] section\n * tunnel_types now belongs to the [agent] section\n * l2_population now belongs to the [agent] section\n\nAlso the tunnel_types option is not explained in the example snippet.\n\nChange-Id: Ic2bde217a03a884855d299f3142394a4229745bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34f7a5805e1fe37da899356411b8ff650e598250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34f7a5805e1fe37da899356411b8ff650e598250"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
