)]}'
{"id":"openstack%2Fneutron~899244","triplet_id":"openstack%2Fneutron~master~I931997fe010295120f2400b1ec310dff82497ab6","project":"openstack/neutron","branch":"master","topic":"bug/2039417","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-26 10:17:42.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-26 09:10:06.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-26 08:56:07.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-10-26 10:17:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I931997fe010295120f2400b1ec310dff82497ab6","subject":"[FT] Clear the idl lock in TestMaintenance tests","status":"MERGED","created":"2023-10-25 09:48:50.000000000","updated":"2023-10-26 10:19:34.000000000","submitted":"2023-10-26 10:17:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899244-bug/2039417","meta_rev_id":"982c34aba7be20e64122d8734a0c1b5ecdf2fbad","_number":899244,"virtual_id_number":899244,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-10-25 16:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-26 10:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-10-26 09:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-26 08:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-10-26 09:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 10:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-25 11:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 16:25:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 05:44:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-10-26 05:44:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-26 05:44:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"607811fb6fe2ddbf350b97b65f1ea409d0d9898c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-25 09:48:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c998c8157a0f2e151b85105860cb586160b71e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 10:41:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76a5931e4fa645fa94deafcf7065443c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2286f24cd6a46c4b12fb04a80b071e9 : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e53f80cd16224c13a13d7b113a4ca1ca : SUCCESS in 49m 53s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4dc75f71396a46df82dee79c933de5ef : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09c523d825f34449bc4e363c5d7b05df : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc304e34f70956a6db8bc4225ea3121bf82e2221","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 11:41:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f762bccf9757411f92249b458cd32492\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7551808f55ec4db6a8aa601cc0987e03 : SUCCESS in 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7dc477aa074df3a2a876647b706034 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a003aa0a83974da2ad11c89c85455000 : SUCCESS in 29m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0689b27736f4140921584b5a2d4e9b6 : SUCCESS in 27m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c3f4cfc39044a4baf2aa42889935a8d : SUCCESS in 25m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e21b5ff62334cafaeb41458dcede32e : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75986af9e3b447e4a9b2be01e9166c0f : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44c54aaab3af4086b4a34049ee89f34c : SUCCESS in 16m 29s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a772e2212beb4fc188fe0af8e450ef22 : SUCCESS in 24m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e8863adc2334f158678d9a7f929590a : SUCCESS in 58m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/594f345b68384c838d4e640284fdc0c0 : SUCCESS in 1h 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"310ce5da8213e3ca95ce38686345dbae8ac00c1c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-25 11:50:21.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6528219a7c7bb5421bdb1a22ca91c97089c5004c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-25 15:55:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4867b0bd653882339d361e9a42a63930b1f9e4b2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-25 16:25:34.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/44/899244/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e196fddac075cad9d544a9a6b59f208add3c4b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 17:00:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f661dbe4934337ae9f7de04d710334\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14453291e2ea40e88c5663aadda568ed : SUCCESS in 9m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07b2bd668c5c43f5b04e9507d962c2ef : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7fbdb73537f4e87b0d3e243b7928544 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":2},{"id":"70c98467722b1f9fc6dddfff448f68c555021242","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-26 08:56:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cc772eafc7e6b16a0fbcd934d2d7c8f6dc62a56","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 09:10:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fe485d6b9a39d739913dd7dbca618e4b0dba4775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"336c5fcd5516d652bd44ebae707bd4df44a0676d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 10:17:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e193201ca44437ae7b8a231ec9df56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042db1fa46d2417cbfacbdb7a3e64d69 : SUCCESS in 12m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/332956cb4af442449ed6f54614277b5b : SUCCESS in 24m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebba79ca6fcc4b978747b5fbc9cdd631 : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3100fa32c96431ea65f9131a009680a : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":2},{"id":"e7886a510dd4c1134aa8f91a39d9ad8884ce4899","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 10:17:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"982c34aba7be20e64122d8734a0c1b5ecdf2fbad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 10:19:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31aaca349fbb4a14ab049c54b1b16ef1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771e6fe9b0cd44d7a2ee4a2585816c25 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"39a1f1e7fd60b9b55dcb4b5ea5027e7759fc3b3b","revisions":{"4bc6b72c470540e7c1a1bb8888f61dbd5a306a09":{"kind":"REWORK","_number":1,"created":"2023-10-25 09:48:50.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/44/899244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/899244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/899244/1"}}},"commit":{"parents":[{"commit":"f185379980d8457d5ea8716eadee3ba56e35cb11","subject":"Merge \"Add support for FDB aging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f185379980d8457d5ea8716eadee3ba56e35cb11"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-25 09:40:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-25 09:42:55.000000000","tz":330},"subject":"[FT] Run TestMaintenance tests serially","message":"[FT] Run TestMaintenance tests serially\n\nOne or the other test in TestMaintenance fails\nrandomly with missing DB Lock.\nRunning these tests serially should help in avoiding\nthese random failures.\n\nCloses-Bug: #2039417\nChange-Id: I931997fe010295120f2400b1ec310dff82497ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bc6b72c470540e7c1a1bb8888f61dbd5a306a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4bc6b72c470540e7c1a1bb8888f61dbd5a306a09"}]},"branch":"refs/heads/master"},"39a1f1e7fd60b9b55dcb4b5ea5027e7759fc3b3b":{"kind":"REWORK","_number":2,"created":"2023-10-25 15:55:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/44/899244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/899244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/899244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/899244/2"}}},"commit":{"parents":[{"commit":"f185379980d8457d5ea8716eadee3ba56e35cb11","subject":"Merge \"Add support for FDB aging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f185379980d8457d5ea8716eadee3ba56e35cb11"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-25 09:40:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-25 15:51:54.000000000","tz":330},"subject":"[FT] Clear the idl lock in TestMaintenance tests","message":"[FT] Clear the idl lock in TestMaintenance tests\n\nOne or the other test in TestMaintenance fails\nrandomly with missing idl lock.\n\nDBInconsistenciesPeriodics set\u0027s the idl lock\nbut in TestMaintenance we don\u0027t run it as a seperate\nmaintenance worker and instead call check_for_inconsistencies\nwhen needed so not a requirement to have a dedicated lock.\nThis patch clears the lock for these tests to avoid\nsuch random issues.\n\nCloses-Bug: #2039417\nChange-Id: I931997fe010295120f2400b1ec310dff82497ab6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39a1f1e7fd60b9b55dcb4b5ea5027e7759fc3b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39a1f1e7fd60b9b55dcb4b5ea5027e7759fc3b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
