)]}'
{"id":"openstack%2Fneutron~899349","triplet_id":"openstack%2Fneutron~master~Id49245707d33540a946db6631d125f99dc79c7ee","project":"openstack/neutron","branch":"master","topic":"remove-get_floatingip_by_ips","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-08 15:36:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id49245707d33540a946db6631d125f99dc79c7ee","subject":"Remove deprecated get_floatingip_by_ips()","status":"MERGED","created":"2023-10-26 01:09:17.000000000","updated":"2023-12-08 15:36:33.000000000","submitted":"2023-12-08 15:36:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899349-remove-get_floatingip_by_ips","meta_rev_id":"995127b319d5c6b252df287b44143e0078174e25","_number":899349,"virtual_id_number":899349,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-10-26 01:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 15:36:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-08 09:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-10-26 09:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-30 01:25:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-08 09:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-26 01:38:39.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 01:59:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-26 03:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 09:12:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-30 01:25:48.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-08 09:42:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"9d0d988ecc1ec873986eab2cee96cb075098aa03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-26 01:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5488fcc42b3ffabb7c21f7ae1d70477d4986142","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-26 01:38:39.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/49/899349/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e6e38ce60da5167ec2ff487ed7dbd37d5fcd868f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 01:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529c235668c24e058c283f7dee380317\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a94fa2effb2e4a07a5e7fbc395f80297 : SUCCESS in 18m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7568aec3b2174f21b278417de189e9c1 : SUCCESS in 47m 07s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5e0f7a8395547bcb91f4d50f050816a : SUCCESS in 47m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3932604089344a6cbba11cfa6af86cda : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"370bc32f8635e86c19a084793b1c4fba31186295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 03:26:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b5da658710f4ab2b5f7434a6533c6dd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/58ba117e368d44768dd795173370e8c2 : FAILURE in 1h 48m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afe1ab5d4c664849b17f3fd63ed4cba7 : SUCCESS in 37m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52d53aca9054c658d73850457a8d0a5 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abdbb086031249cb991de4e94b806857 : SUCCESS in 29m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e372100622654237b8477bad67766023 : SUCCESS in 22m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f182c25bdd6540d2a9a7d6de05de3fa4 : SUCCESS in 22m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e2bc3ebcdba46a9a51e22fd951c6c65 : SUCCESS in 25m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b42e8b977c4f42b23945bf8349f693 : SUCCESS in 9m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c8724dc86dbc423ca6cb422acaa09ad5 : SUCCESS in 22m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3f208b23864e4e0db550f33996fac7fc : SUCCESS in 1h 08m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/037a6ee4e9d345ce9a4189ddc987828b : SUCCESS in 59m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7c7ce297fcfe4d53a1eb40e126c65d2c : SUCCESS in 1h 05m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6777c1660c634f43a744019ca1f308dd : SUCCESS in 2h 10m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b667a9509f904e9eb076a035ed99fa07 : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d76b32fe8ecf8eb1d04a19d38816ecd9e435801f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-26 09:12:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e5fcadac6c9237d1b0bded664cadba0997c667d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-26 19:47:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0e6afb64fd827d2c7decb5005a0f4cad213739d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 20:37:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b639e8dacf594068ba1696008562d804\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95f3a6f1af3340888a87d563fa6a6d29 : SUCCESS in 48m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7b5b3cd3cda4417afd1dd2cf09394fd : SUCCESS in 18m 59s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fdd957fdbd4f41769de789834e9e6f96 : SUCCESS in 46m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/48ceb124b3b442849e62f8dbe664871e : SUCCESS in 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3a8223e8e15aefe7df63b458a2a06e2131402e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 21:13:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/173297896aa14d439d409cefbc585a9d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e9f0350f89454040849acc66904d4f56 : SUCCESS in 1h 25m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb0bde28d1a54f5987aa12c5095259f8 : SUCCESS in 39m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aa12c9e662d432cb71a422d8d246eb1 : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/107cce319bff40189b60ac56e7284f94 : SUCCESS in 29m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63bffe55bced40d8a5e0bf7849da25a7 : SUCCESS in 18m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/199b20b8465d470e89d65b1d636f5808 : SUCCESS in 26m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1b07c816e924627a39e41330a285629 : SUCCESS in 25m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88686baf74ed49208d55d9de4d386e27 : SUCCESS in 11m 34s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c57292cb47e24ebaabb7c29327e2d5a8 : SUCCESS in 21m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dbb190cede9949f293dfe61602400bc7 : SUCCESS in 1h 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/453d99a98d7b401c944160891d3d82e4 : SUCCESS in 52m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d0421afe31c420f9930f9580e668109 : SUCCESS in 43m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bf170eece9c146328ffce2ca23f8a33e : SUCCESS in 1h 25m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/adda889c0e3f4ecd86898cbeda023b3e : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3c2a85af95aeb3348a7a640946053dc9a53018a3","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:25:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d0a022131343105c6cb82fe35609a894dacade8c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-08 09:42:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a0573ff11ccfdfafeda27ae5902b77cd882b418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:42:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ba0357b85d41c3bd24f0497759e5df73093ee31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:36:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc8a19823aad49b484a6bc9a76e64af8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7771a7842d404c849f632cfcf8010233 : SUCCESS in 1h 41m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84eb5f7e02554743af692eba46d94fd7 : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f692a20b133b456699f7ed089c9958e4 : SUCCESS in 23m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a56c91f52d9b48919a2454f319597868 : SUCCESS in 22m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b19678a2fe446bbf98ed261a6460d3 : SUCCESS in 11m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae47a8ca346f466487886e1beef9e993 : SUCCESS in 1h 05m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8f3c06c88c114b00950dae83e52bd427 : SUCCESS in 1h 04m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1d3b3cf9fdd94d76a36fcd123e16dbb4 : SUCCESS in 1h 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"995127b319d5c6b252df287b44143e0078174e25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:36:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4d88686518b55b17fe08c121c35291ca6451c54","revisions":{"c4d88686518b55b17fe08c121c35291ca6451c54":{"kind":"REWORK","_number":1,"created":"2023-10-26 01:09:17.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/49/899349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/899349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/899349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/899349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/899349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/899349/1"}}},"commit":{"parents":[{"commit":"f185379980d8457d5ea8716eadee3ba56e35cb11","subject":"Merge \"Add support for FDB aging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f185379980d8457d5ea8716eadee3ba56e35cb11"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-26 01:07:23.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-26 01:07:23.000000000","tz":-240},"subject":"Remove deprecated get_floatingip_by_ips()","message":"Remove deprecated get_floatingip_by_ips()\n\nget_floatingip_by_ips() in the OVSDB API was deprecated\nin Rocky and there currently are no users, remove it.\n\nTrivialFix\n\nChange-Id: Id49245707d33540a946db6631d125f99dc79c7ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4d88686518b55b17fe08c121c35291ca6451c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4d88686518b55b17fe08c121c35291ca6451c54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
