)]}'
{"id":"openstack%2Fneutron~899373","triplet_id":"openstack%2Fneutron~master~I5e4fa6a294988cf43fa031b7d1cc734e6132b199","project":"openstack/neutron","branch":"master","topic":"bug/2040457","attention_set":{},"removed_from_attention_set":{"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2023-12-04 11:01:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e4fa6a294988cf43fa031b7d1cc734e6132b199","subject":"Regeneration new allocation_pools if allocation_pools is empty","status":"ABANDONED","created":"2023-10-26 08:45:07.000000000","updated":"2023-12-04 11:01:59.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8ef6f20b6b410f83ced3a32166362fd9e2650a1a","_number":899373,"virtual_id_number":899373,"owner":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-10-26 09:14:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-26 11:48:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-2,"date":"2023-10-29 22:02:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-26 09:09:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-26 09:14:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 11:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-29 22:02:38.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-12-04 11:01:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"5848236270b20f8bd8e142a2784f689227b9d45c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"date":"2023-10-26 08:45:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2eb9ea77576226c45a641595fa2f005369e7b340","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:09:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29bc3b0cdd0e40b09ea13732673fa03f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8b1843de11846bca6e5ba0dc2b6a6fc : FAILURE in 19m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/668c33f59d364288a64b185c7765ab1b : FAILURE in 19m 04s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0da10c64722475dbbb7f4b77c460a10 : FAILURE in 23m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7377e043a6e64dea97816e16fa5202d8 : FAILURE in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f62a7ad0f82e528fdbe616f81e86c1118a0ee44","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-26 09:14:59.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/73/899373/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"28538092e5b24eeb43f72944633be027e8de9527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 11:48:50.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d51be88810c40acaa0cb27ac2bcf209\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c667a945d03f481f85b3fd16992f6f1e : SUCCESS in 1h 21m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a8d6c7e311f45e89b0ace8e486e7a91 : SUCCESS in 1h 34m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d576bdca08ef433c906fc2ed4f1cec75 : SUCCESS in 1h 27m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6691539357a47e5bcbaac4b6973dd21 : SUCCESS in 2h 01m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ff13cd5dff248c1a3ec8c34f65aa309 : SUCCESS in 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94eda00589624e9ba41b49e3807bf234 : FAILURE in 36m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/932bc7c342f54ba0a48eee2abc11f1e6 : FAILURE in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f974c11e06584c45aee66d4b1378d6a0 : FAILURE in 25m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5433f4e99c8f4dbf93f2d050e2211012 : FAILURE in 29m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87d8f3ada3cb468f917ea3b109aaf927 : FAILURE in 26m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e25845bbdf7e43d6bfaad9a3b1484745 : FAILURE in 26m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de672900053a422da8897dbc270a89bb : SUCCESS in 13m 59s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/32a7240b7d394a639f6455c4193871fa : FAILURE in 23m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/526e99f54d53413784e9fb5c7344a3ec : SUCCESS in 58m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f912f37c26f34214a2fd540b15d2125d : SUCCESS in 1h 07m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8d36781a3f94d39af0feaae2518a916 : SUCCESS in 56m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb723ad2107544a3bc815cddeb36d762 : TIMED_OUT in 3h 02m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cfa3ebb75d9414b874e4ceb6054e975 : SUCCESS in 51m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0b953d42e81c490786fc400a59c7f34c : SUCCESS in 1h 00m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/620313c14c5e4524b7476e7fe7c12a68 : SUCCESS in 1h 32m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e0864013c16944319684e3ee93337e3d : SUCCESS in 1h 15m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/797e53ffd82747e08f0c77294ac1cdb2 : SUCCESS in 55m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0b74e2438dee477387aefba750dde61d : SUCCESS in 1h 17m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/58682f68770348aa83472856e97c5b1c : SUCCESS in 2h 19m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4907d6e6365b4965b79d792a657c7352 : SUCCESS in 1h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"306f553de85e6749ff69f318a74857a477c8092c","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-10-29 22:02:38.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ef6f20b6b410f83ced3a32166362fd9e2650a1a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:01:59.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment and currently blocked by a core reviewer with a -2. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and contacting the reviewer with the -2 on this review to ensure you address their concerns.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63298a53271635508ca672a98266d9ba62428a49","revisions":{"63298a53271635508ca672a98266d9ba62428a49":{"kind":"REWORK","_number":1,"created":"2023-10-26 08:45:07.000000000","uploader":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"ref":"refs/changes/73/899373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/899373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/899373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/899373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/899373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/899373/1"}}},"commit":{"parents":[{"commit":"389bc155f85b77f6dd265d039ac0ea5431cfdb15","subject":"Merge \"doc: Stop configuring authtoken middleware in compute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389bc155f85b77f6dd265d039ac0ea5431cfdb15"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2023-10-26 08:39:30.000000000","tz":480},"committer":{"name":"liushy","email":"liuxie_11@163.com","date":"2023-10-26 08:44:56.000000000","tz":480},"subject":"Regeneration new allocation_pools if allocation_pools is empty","message":"Regeneration new allocation_pools if allocation_pools is empty\n\nIf allocation_pools is empty list when update one subnet, we need\nregeneration new allocation_pools from cidr and gateway_ip.\n\nCloses-Bug: #2040457\nChange-Id: I5e4fa6a294988cf43fa031b7d1cc734e6132b199\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63298a53271635508ca672a98266d9ba62428a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63298a53271635508ca672a98266d9ba62428a49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
