)]}'
{"id":"openstack%2Fneutron~899376","triplet_id":"openstack%2Fneutron~master~I0698aa4621a15c1927ad2c352501cea02e6ee70c","project":"openstack/neutron","branch":"master","topic":"fix-install-docs","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-27 18:20:06.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-27 17:14:53.000000000","reason":"removed on reply"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-27 18:20:06.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-27 18:20:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0698aa4621a15c1927ad2c352501cea02e6ee70c","subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","status":"MERGED","created":"2023-10-26 09:27:19.000000000","updated":"2023-10-27 18:24:07.000000000","submitted":"2023-10-27 18:20:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899376-fix-install-docs","meta_rev_id":"10639a539ded7559a200a903236f971e47202335","_number":899376,"virtual_id_number":899376,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 18:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-10-27 17:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-10-27 16:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-10-27 17:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-26 09:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 10:33:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-10-26 18:13:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-27 14:54:00.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-27 15:39:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-10-27 16:06:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-27 17:14:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"0cbe54bce7aa210eee8b915d6fe0b7d3087478ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 09:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"261b4f1fa9c64fb0151e737f5b86ce82710561bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 09:40:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4a0847c011d4a8ebfcc2940419fa110\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fcda9f21c74a20b8e67a71c75245b9 : SUCCESS in 9m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c715b77010a4428804ebc21544deb4c : FAILURE in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9f33c0e7c9a7792999119542a06d68fcac4f087d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 10:03:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"752592d8687010408247ecbf2ca5a700e00a0d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 10:22:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/105d189876064c1c8b39e301beceadf6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a03aafe2a84dc4b0ec1d6b78ac3e23 : SUCCESS in 12m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ed4cf9fa04f4901859f0deb5425b366 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":2},{"id":"22d63f2e103f2c04fe550c3d23ceba88c1b234e5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-26 10:33:41.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c14ff56e64ed38b3097fde0b5dcd74205f1893b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 11:05:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fa89edf5e7af959e3d3c0f274b961f3a28d59642","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-26 11:05:14.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3871aad4fd976ff643a67eb5e2feac8b1ef8fb66","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-26 11:10:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0d92af1632157ce86d566e5045f55c1871bd71f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 11:21:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/252994a6638d47f3b0b7e45922e1efac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0bf5119a91434cac61e4d7da5ddbdd : SUCCESS in 11m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6c4d36f7914f01a750adac722ef4b2 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f6c143e11e82bd752cb0883947cb4f216103c7f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 14:08:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"36b765825d2c42b88b32594530abd2c334503d63","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 14:08:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab6ee194d7c6498be4340d991dcddefb6841fef7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 14:22:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ebc2e46cfc9465a4be7f228707ac7c755c7ecc83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 14:31:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"07709ef50bfad1c322b0c30a4f49b5fdb0f8ff5a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 14:31:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f76df225713583d181a4980a7449a37295631c29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 14:51:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667da55261654659a01299302ec6a497\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d214c13a6b724b3d9afac16a52635892 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b3dc241b6154494ad698fc4d3020eb8 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"2ab44609fb1debcc43d7ef617f179daa01d0152c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 14:54:00.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d18b0b0b668f7883aa4aeff2297cfdad177d1efb","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 15:39:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a51cab0c03a2e8dc27847e702b3259c70df33126","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 15:57:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1b652ce6754da29f201206219daeb8e4fe9027f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-27 15:58:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"56cf79114404c30e5ed187ef2999a4e012d596ca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-27 16:06:43.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc2b02895a045dfe72db84fb538ae1f3a547e3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 16:09:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2487bd50eb1d48f09ed3ecc2862588ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28eaac7a02cb4260bc243afe7b9904aa : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94f70c118a4545ce9e97080153b88872 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":6},{"id":"0d62017e348f91583fa2e9099ad7cbe543e9a051","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-27 16:09:40.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b95d8b089f7d87feb5325f850eeb8b9563fa2a59","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 17:14:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"05e3f899dc11555591da55d11a02f59434a449c5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-27 17:15:54.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"42ce167e19c6e0edc14ec9aaba536a86ada6e4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 17:16:24.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0ca7268d76208bb193f836fddada4e80803cac69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 18:20:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e140713a692493cb5c51a1c6151f860\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c352ced661a4b3ead68aa9e288cdb1f : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412bd304faf843a48590d2493cb999b4 : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":6},{"id":"a2b93134df3ae9e27bf95f46ab2cc38643bbee71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 18:20:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"10639a539ded7559a200a903236f971e47202335","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 18:24:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4f17a661d544d69e1b58f94e9f227f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396a1145698a4251bca6ff67812e008b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"842f2f8e6cbd39c4eefddc9e1b74c0f7f8ea55ff","revisions":{"639af5d104d0d7cb4db714cbdde6f8dd656076da":{"kind":"REWORK","_number":1,"created":"2023-10-26 09:27:19.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/1"}}},"commit":{"parents":[{"commit":"389bc155f85b77f6dd265d039ac0ea5431cfdb15","subject":"Merge \"doc: Stop configuring authtoken middleware in compute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389bc155f85b77f6dd265d039ac0ea5431cfdb15"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/639af5d104d0d7cb4db714cbdde6f8dd656076da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/639af5d104d0d7cb4db714cbdde6f8dd656076da"}]},"branch":"refs/heads/master"},"54b59dfca008c3846a0b6a4f4d9118dac3cf2e45":{"kind":"REWORK","_number":2,"created":"2023-10-26 10:03:45.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/2"}}},"commit":{"parents":[{"commit":"389bc155f85b77f6dd265d039ac0ea5431cfdb15","subject":"Merge \"doc: Stop configuring authtoken middleware in compute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389bc155f85b77f6dd265d039ac0ea5431cfdb15"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 10:03:38.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54b59dfca008c3846a0b6a4f4d9118dac3cf2e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54b59dfca008c3846a0b6a4f4d9118dac3cf2e45"}]},"branch":"refs/heads/master"},"f9fc94d5921afa66f39158b832d2fd555705bb59":{"kind":"REWORK","_number":3,"created":"2023-10-26 11:05:08.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/3"}}},"commit":{"parents":[{"commit":"389bc155f85b77f6dd265d039ac0ea5431cfdb15","subject":"Merge \"doc: Stop configuring authtoken middleware in compute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/389bc155f85b77f6dd265d039ac0ea5431cfdb15"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 11:04:29.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9fc94d5921afa66f39158b832d2fd555705bb59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9fc94d5921afa66f39158b832d2fd555705bb59"}]},"branch":"refs/heads/master"},"57db4d97dceef8fc606d8dc4ab6a82b6e972b656":{"kind":"REWORK","_number":4,"created":"2023-10-27 14:08:27.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/4"}}},"commit":{"parents":[{"commit":"53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7","subject":"Merge \"doc: Fix wrong sections of ovs-agent config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-27 14:07:24.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57db4d97dceef8fc606d8dc4ab6a82b6e972b656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/57db4d97dceef8fc606d8dc4ab6a82b6e972b656"}]},"branch":"refs/heads/master"},"a45b7026d1a57b4e6c6b1bcbe78d82c62e56c79e":{"kind":"REWORK","_number":5,"created":"2023-10-27 14:31:28.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/5"}}},"commit":{"parents":[{"commit":"53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7","subject":"Merge \"doc: Fix wrong sections of ovs-agent config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-27 14:31:12.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a45b7026d1a57b4e6c6b1bcbe78d82c62e56c79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a45b7026d1a57b4e6c6b1bcbe78d82c62e56c79e"}]},"branch":"refs/heads/master"},"842f2f8e6cbd39c4eefddc9e1b74c0f7f8ea55ff":{"kind":"REWORK","_number":6,"created":"2023-10-27 15:58:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/76/899376/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/899376/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/899376/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/899376/6"}}},"commit":{"parents":[{"commit":"53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7","subject":"Merge \"doc: Fix wrong sections of ovs-agent config options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53f4fd6b9fcb4f8ba907bfbace342bf902fc55f7"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-26 09:25:53.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-10-27 15:57:41.000000000","tz":120},"subject":"[Docs] Fix info about provider bridge name in the bridge_mappings","message":"[Docs] Fix info about provider bridge name in the bridge_mappings\n\nPreviously in all our install guides there were info that name of the\nphysical interface should be put in the bridge_mappings config option in\nths OVS agent\u0027s config. This wasn\u0027t correct as bridge_mappings expects\nthere bridge name instead.\n\nChange-Id: I0698aa4621a15c1927ad2c352501cea02e6ee70c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/842f2f8e6cbd39c4eefddc9e1b74c0f7f8ea55ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/842f2f8e6cbd39c4eefddc9e1b74c0f7f8ea55ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
