)]}'
{"id":"openstack%2Fneutron~900415","triplet_id":"openstack%2Fneutron~master~I1f77dfdb91706d7b02576798d64a29f87f161d5c","project":"openstack/neutron","branch":"master","topic":"list-routes-only-fetch-needed-devices","attention_set":{"35825":{"account":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"last_update":"2024-01-10 18:50:43.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2023-11-15 10:09:10.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-02-21 20:44:43.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1f77dfdb91706d7b02576798d64a29f87f161d5c","subject":"Only fetch needed devices when listing routes","status":"NEW","created":"2023-11-08 14:23:33.000000000","updated":"2024-02-21 20:44:43.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":16,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"ed7842122a3623fc4496e95203d43c22101e07b5","_number":900415,"virtual_id_number":900415,"owner":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-11-15 10:56:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-15 12:19:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2024-02-21 17:40:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 14:36:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-11-08 14:57:48.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 15:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-08 17:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 17:07:37.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-11-15 01:23:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-02-21 17:40:23.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"acaa629a953335d2d3b2818a6f4727bd84b4b6cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"date":"2023-11-08 14:23:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d399c0bee6774fd7531a9c5f40d932a20e6930f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-08 14:57:48.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/15/900415/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"fea67d52c36498a91a2c8fc58bb4d0eaeff8f3e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 15:14:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4064450287d84b1ea262d15e0687c20f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c72c1255368d444f8352afaac1cef6c4 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/113b01734f7845bfbd4f70844b333a3a : SUCCESS in 48m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a259bd87f7d4165bad3b8e9406440df : SUCCESS in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afc7ddbde0e2333ff98d41220144fe82cc3ce5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:20:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3426fab2e64041f39e60226126b7e172\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/82aadaa3dfda4314b2041f071825393b : SUCCESS in 1h 40m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8d20248ffecc4c13a011f736f87c2458 : SUCCESS in 1h 42m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/695ed987363149698a7894b0b23e44c4 : SUCCESS in 1h 38m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c045fbf53b6e478d9499aa7955216ca2 : SUCCESS in 1h 52m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3122c37540c480ebb3ae69d2311020e : SUCCESS in 38m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085b81454cf641fb8475bfa0490f40a9 : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a17ca5dcbf84e42b28e4eeb9b9da124 : SUCCESS in 30m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/396eb625ae614417a45688d4bbcda6d3 : SUCCESS in 31m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f517a24d0c214712b9ab01348150ea06 : SUCCESS in 11m 41s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3e575b7f53834b6f84c37f37103d1cf9 : FAILURE in 6m 50s (non-voting)\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3bae85501cb24a1ca1e6948d6b46d4f6 : SUCCESS in 51m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/003668218a3d45e8aa6621cab985b034 : SUCCESS in 56m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27cc5525c0da4d9796884203ff84766a : SUCCESS in 1h 11m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/36d28454d9434b138b742c6efe10bb33 : SUCCESS in 2h 15m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3f334871d374aef87b379d25ca2896b : SUCCESS in 1h 07m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58f95fee4efb44f09be47692117f3dbb : SUCCESS in 49m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1dda0f0f2cb54eda9794e41941896b04 : SUCCESS in 1h 27m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/47292c491eec440189093a51d3735be8 : SUCCESS in 1h 38m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32cbea5231864b4fb8c61566849bfe3f : SUCCESS in 1h 03m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65661ad1bd6441ef93e38b06b7fdd3f1 : SUCCESS in 2h 20m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9bc326c391614defbd30a7bae5e2e5f7 : SUCCESS in 1h 42m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed8060c8776b4c45bfa6874e4aad806d : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":1},{"id":"beffddcb6b34e3cbba562aa90b61ed596900d227","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-11-13 17:07:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea516dca7878ae1c856e2f81a7db4a310319e6cc","author":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"date":"2023-11-13 17:16:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc5c1e671a17f2946095d7a173a865fc3b21c0de","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-11-13 17:23:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0315f0b67fa311af2563f9028e2bea8bf940324","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-15 01:23:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2160f3297ca5a1aad4851d8cb8f6e1abb241c79a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 02:04:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25dbd77be76d4823b9475175f0530351\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/012e9dc1aa284ded9e3198d5c5792a2a : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bcf7f27d2594d63b9b7c490b24b6935 : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4452b099470b440d9a03b63ebc32b828 : SUCCESS in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8e89a814d79f7ca148645223bbdea8f719f30b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 04:26:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00a287bf9d4640fe961e50558311f8d4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7c8e3e9625c5450d923b7b63249a8ff7 : SUCCESS in 1h 40m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/236186a6694f4b68872549258be18f9b : SUCCESS in 2h 07m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4106bf67c6a3466ab6494763bf9acea0 : SUCCESS in 1h 40m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e6ec28fd88484305a0cdbb4393054292 : SUCCESS in 1h 30m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d18816324a84dd79da8fcef391b79ec : SUCCESS in 31m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d476f2acdd2474c8fc1c364e5dd7685 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44e1e736bb44478aa4f2da211e011c06 : SUCCESS in 23m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2eef06f253ef425fb81538944321a30c : SUCCESS in 25m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e935db79cf104762b5ba00f9b2b3cc8c : SUCCESS in 12m 17s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d2e71caafa404e6b96e04193571940ce : SUCCESS in 28m 52s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a7534734634f495b949c5bfc8f549a0f : SUCCESS in 46m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/abff8e406dc648b8ad685ba9b0a66540 : SUCCESS in 1h 14m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3f8daef52b744edeb2e0082db489c6b2 : SUCCESS in 58m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d588b6f6acd4379b490cd728c31d9b1 : TIMED_OUT in 3h 01m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb59dfda614c4f20821da53eaf53b382 : FAILURE in 49m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0a342bdd755149578b45b79c1333c61c : SUCCESS in 1h 06m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/296a647795a94bab9affe1ca9d44e2ad : SUCCESS in 1h 19m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/103e614c197b42c7b927c1a128f33a35 : SUCCESS in 1h 07m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/039c1ef076e64f4baf70945b9a135767 : SUCCESS in 52m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9d66f2e506a24654b2d0b217a64eb019 : SUCCESS in 1h 33m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/18ac645186b04963807ce8a587888592 : SUCCESS in 2h 01m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3a9a6cc6250848999f65bb6061554081 : SUCCESS in 1h 23m 20s","accounts_in_message":[],"_revision_number":1},{"id":"041b550fa6ed10f6ee75ccb3858050fbd1c2299b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"date":"2023-11-15 10:09:10.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5f7dcf43f2183ff67d838bf21d53d1dcbb4cbdd6","author":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"date":"2023-11-15 10:23:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3119ecf92ae3d4cc06c9fd5005442a5cba763b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:36:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c75294504e8a4622bb555362b4a96d19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8970440b2fee40178a1ee46d4742117b : SUCCESS in 25m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62390bfddc6e4ba4b842c27d7210154a : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/583d3a5a2a294f52a3c30e12e8f11b9b : SUCCESS in 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c057f3bccdc21db1c74b0cbc0ae26d66958269a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-15 10:56:04.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/15/900415/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8c6ebc536bcaba1501e571c18ee3ec406af62878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 12:19:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38a059b2d23a416e9e341e1bbb3fea52\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b273060094284ef48867e26066d9e82c : SUCCESS in 2h 01m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae3fbbf10f814853b759d9b2f35c0120 : SUCCESS in 1h 57m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/12724b029ca543b5bd21a749fa806028 : SUCCESS in 2h 06m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef0c0123e0bd42d187b0bf8f6905c7af : SUCCESS in 2h 08m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/720a96c9ce5142ee856db235b192bad6 : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b354a4222a34ca5b647852aedaa19b7 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5c03734575d49a5935dd1d35120554d : SUCCESS in 27m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0799347fc2214690b8224616d9bfb0ec : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9989defa86754cd6abfc253beb78fc10 : SUCCESS in 10m 04s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f9492bc65cf047808f3befb9c4f04cc7 : SUCCESS in 27m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c00edf8c81964f7db186adadffd23b79 : SUCCESS in 55m 12s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/038ab17cd3754d6394620ebc6ebe657b : SUCCESS in 1h 03m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e12fdfe5e6ff41cda21e98644211303a : SUCCESS in 1h 15m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b514e518e8c4fc79d54ae2322b25215 : SUCCESS in 1h 56m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/357b80f2109c4ad08e9b07000dc53c3f : SUCCESS in 1h 07m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/626707f97f294c6d9002eda54ca13d07 : SUCCESS in 1h 13m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79d97e9214fc48b49cfa58bc0f00e9f9 : SUCCESS in 1h 33m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0101d41e37134ffc849682fa4a25015c : SUCCESS in 1h 18m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30f7bbc455a844f7830a52e3434e6b7d : SUCCESS in 1h 04m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/97142f7af6cd4182845a9a70978a6178 : SUCCESS in 1h 37m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f7a7e8b62cf94116b55fdd8f377bb242 : SUCCESS in 1h 59m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3968f37353684047919794d37a50c811 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":2},{"id":"902730cede76ffcbcf89581363304924b0dc27e9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-10 18:50:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8f9b4a9df61ddc364767b755e877e391d39c787","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 17:40:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"061be10b4ad03cecbeb410321af5b95129bcea3e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 20:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5347d80797ac8003ab48d3e1048c39706819b441","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 20:36:06.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0ea75c84262fb844206659b4cacd7e80d4d715f5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 20:36:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5bea224c3b8fdd88f6237972ac715d90ca502de","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 20:42:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed7842122a3623fc4496e95203d43c22101e07b5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-02-21 20:44:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a5e40ae6ae583c65f7d4d184475211783f4f2a1","revisions":{"087195179f090b837e7c648f601e96fd88abec06":{"kind":"REWORK","_number":1,"created":"2023-11-08 14:23:33.000000000","uploader":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"ref":"refs/changes/15/900415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/900415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/900415/1"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Adam Oswick","email":"adam@adamoswick.co.uk","date":"2023-11-08 14:14:03.000000000","tz":0},"committer":{"name":"Adam Oswick","email":"adam@adamoswick.co.uk","date":"2023-11-08 14:14:03.000000000","tz":0},"subject":"Only fetch needed devices when listing routes","message":"Only fetch needed devices when listing routes\n\nWhen listing routes, we fetch all devices and filter out only the ones\nwe need. Due to privsep, all of the interface information needs to be\nserialized, sent to the unprivileged process, and then deserialized.\n\nIn environments with lots of interfaces, this logic can be quite slow.\nBy fetching only the necessary interfaces, we can achieve a noticable\nreduction in time taken for the agent to perform a router update.\n\nChange-Id: I1f77dfdb91706d7b02576798d64a29f87f161d5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/087195179f090b837e7c648f601e96fd88abec06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/087195179f090b837e7c648f601e96fd88abec06"}]},"branch":"refs/heads/master"},"1a5e40ae6ae583c65f7d4d184475211783f4f2a1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-15 10:09:10.000000000","uploader":{"_account_id":35825,"name":"Adam Oswick","email":"adam@adamoswick.co.uk","username":"adamoswick","status":"GoDaddy"},"ref":"refs/changes/15/900415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/900415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/900415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/900415/2"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Adam Oswick","email":"adam@adamoswick.co.uk","date":"2023-11-08 14:14:03.000000000","tz":0},"committer":{"name":"Adam Oswick","email":"adam@adamoswick.co.uk","date":"2023-11-15 10:08:45.000000000","tz":0},"subject":"Only fetch needed devices when listing routes","message":"Only fetch needed devices when listing routes\n\nWhen listing routes, we fetch all devices and filter out only the ones\nwe need. Due to privsep, all of the interface information needs to be\nserialized, sent to the unprivileged process, and then deserialized.\n\nIn environments with lots of interfaces, this logic can be quite slow.\nBy fetching only the necessary interfaces, we can achieve a noticable\nreduction in time taken for the agent to perform a router update.\n\nCloses-Bug: #2043564\nChange-Id: I1f77dfdb91706d7b02576798d64a29f87f161d5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5e40ae6ae583c65f7d4d184475211783f4f2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a5e40ae6ae583c65f7d4d184475211783f4f2a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
