)]}'
{"id":"openstack%2Fneutron~900574","triplet_id":"openstack%2Fneutron~master~Iea375aa6c3e48ed55cf49118885438b2be5c2b1b","project":"openstack/neutron","branch":"master","topic":"bug/2042943","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-12-04 04:59:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iea375aa6c3e48ed55cf49118885438b2be5c2b1b","subject":"[DNM] Reproduce issue","status":"ABANDONED","created":"2023-11-10 06:19:59.000000000","updated":"2023-12-04 04:59:06.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3f0ef7f0fdcc0b8cc7deeabbbae7eee1db519fe9","_number":900574,"virtual_id_number":900574,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-15 08:27:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 07:14:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b6327a09290e6b3d5ea7c90115431f102f2dfce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-10 06:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a277156e399eb5ccd73504c297ec0e9e1d2650b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 07:14:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c09efc94f7447f4aea61866485071d2\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7cd4a810ee654a5abcf6d5dabc99b055 : SUCCESS in 46m 49s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2a079d0ddb9d4dbd9ba82fcefd71f4c1 : FAILURE in 29m 34s\n- devstack-1 https://zuul.opendev.org/t/openstack/build/a2a576e80e1e4755b3edf853e98ce09d : SUCCESS in 31m 40s\n- devstack-2 https://zuul.opendev.org/t/openstack/build/735c274f4b6e4aa98f2cbe0fbe4a7d85 : SUCCESS in 31m 36s\n- devstack-3 https://zuul.opendev.org/t/openstack/build/40c9616c458b4bcfa888558b2e245cc8 : SUCCESS in 16m 18s\n- devstack-4 https://zuul.opendev.org/t/openstack/build/27b50f77abc34441ad21981221a005ab : SUCCESS in 31m 23s\n- devstack-5 https://zuul.opendev.org/t/openstack/build/a54c6c2107b74d90ae20ce7ab4cdd0a8 : SUCCESS in 16m 59s\n- devstack-6 https://zuul.opendev.org/t/openstack/build/0626bcf91cb6414ea4491d1d568f90a2 : SUCCESS in 16m 11s\n- devstack-7 https://zuul.opendev.org/t/openstack/build/308319c37f7d47369bb97168221a344e : SUCCESS in 30m 03s\n- devstack-8 https://zuul.opendev.org/t/openstack/build/b7ebb27e44ca4dd996f01466cde1e8d3 : SUCCESS in 17m 01s\n- devstack-9 https://zuul.opendev.org/t/openstack/build/e4b7aac95f79426b943badc0f58eee33 : SUCCESS in 30m 12s\n- devstack-10 https://zuul.opendev.org/t/openstack/build/3fa4d8c605f546a79c3330503588170d : SUCCESS in 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"68aa5ce548ad8d2b91e6a7a260c2dd05f3f88bde","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-10 09:34:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81cfe9275b2695143a575b801fa994589ef1accc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 11:16:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbaf18e2bff94039845fa9a6741bbbb1\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e63cda9d566b4b95baa784391f4d0242 : SUCCESS in 35m 04s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8305c790df4f4f63a041259fa1ea4a4a : SUCCESS in 1h 31m 12s\n- devstack-1 https://zuul.opendev.org/t/openstack/build/4fda4a1bf7104af5bfe3e7bac97bbab3 : SUCCESS in 36m 14s\n- devstack-2 https://zuul.opendev.org/t/openstack/build/71da826cbb1f4d2e85466c11fab7c3d4 : SUCCESS in 31m 58s\n- devstack-3 https://zuul.opendev.org/t/openstack/build/c86497b79f81455d8c0cf6dcbd474dd8 : SUCCESS in 20m 01s\n- devstack-4 https://zuul.opendev.org/t/openstack/build/0b744694b6f54822a41ecb1ce585a631 : SUCCESS in 14m 33s\n- devstack-5 https://zuul.opendev.org/t/openstack/build/98fb5e18cd674a57b7979f704aaa79ab : SUCCESS in 31m 33s\n- devstack-6 https://zuul.opendev.org/t/openstack/build/5e83918bfd5042e4a7fc78e46b5031f2 : SUCCESS in 20m 28s\n- devstack-7 https://zuul.opendev.org/t/openstack/build/a693b47880974e98b356e431cf7bc9ca : SUCCESS in 30m 17s\n- devstack-8 https://zuul.opendev.org/t/openstack/build/4b39d7faee2e4f26adf5208353784758 : SUCCESS in 16m 40s\n- devstack-9 https://zuul.opendev.org/t/openstack/build/1dcc28631dc841c781ba5ed3f08826d8 : SUCCESS in 20m 01s\n- devstack-10 https://zuul.opendev.org/t/openstack/build/6f78cb9bab524736a76309b6ebdce813 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"47ac85e6f47013f194c46843eb26e39d273990f2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-10 11:33:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91de8a1186a703abe8a34560b2ab1e85b63f80c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:10:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6262e6c6aa994355984cdef716b57191\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a650738d039843b29821cbb6286ee31b : SUCCESS in 40m 02s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/61c0688e1b464214952af930882444ed : FAILURE in 2h 30m 21s\n- devstack-1 https://zuul.opendev.org/t/openstack/build/91e8e169375d4738867ef5dd405a5a7c : SUCCESS in 35m 59s\n- devstack-2 https://zuul.opendev.org/t/openstack/build/787dc0b28139411da50b5aa42410066a : SUCCESS in 20m 37s\n- devstack-3 https://zuul.opendev.org/t/openstack/build/0e3f6091450c48f6b750d68b98dbff03 : SUCCESS in 19m 50s\n- devstack-4 https://zuul.opendev.org/t/openstack/build/a6e51b5cc92142fa87d8db18cbbb4ed2 : SUCCESS in 23m 16s\n- devstack-5 https://zuul.opendev.org/t/openstack/build/6e9e489f07274f2d85981a86d2f5d4d5 : SUCCESS in 22m 34s\n- devstack-6 https://zuul.opendev.org/t/openstack/build/0c8ec71ad26c42769c5c6bb499ddbd54 : SUCCESS in 23m 07s\n- devstack-7 https://zuul.opendev.org/t/openstack/build/992826d1834b403f8131f2a8635b48f2 : SUCCESS in 18m 04s\n- devstack-8 https://zuul.opendev.org/t/openstack/build/4fc76694213a4967895683904e4c682c : SUCCESS in 20m 14s\n- devstack-9 https://zuul.opendev.org/t/openstack/build/4d020f88087d4ca9b69a40f34ea7a379 : SUCCESS in 28m 57s\n- devstack-10 https://zuul.opendev.org/t/openstack/build/32e34619ec9440bdb7d23034847e44d5 : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3a797b3ef3ec414d780ca1b2ff3205000c418274","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-10 14:11:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4349efc19c83bfe25ccce7cb6643dae1ab72761f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:16:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c078c2619c044f92997cf026dd077c22\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/096924d0c4ea4f3ba08667c2f0be2b1f : SUCCESS in 57m 24s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a24b1ad503904d2181b5bd8cef58ce4f : SUCCESS in 1h 59m 00s\n- devstack-1 https://zuul.opendev.org/t/openstack/build/1c411970db844f6a994db628821c77cc : SUCCESS in 36m 06s\n- devstack-2 https://zuul.opendev.org/t/openstack/build/1f86715b54934ddbbfe0a36b279443ba : SUCCESS in 14m 45s\n- devstack-3 https://zuul.opendev.org/t/openstack/build/91a78888535241d48b4f7dc8066b6c8a : SUCCESS in 18m 50s\n- devstack-4 https://zuul.opendev.org/t/openstack/build/88956c21f5cc45dc9c8d4692174fa01b : SUCCESS in 30m 42s\n- devstack-5 https://zuul.opendev.org/t/openstack/build/9d82a4120498448f95f243da1e2011d9 : SUCCESS in 17m 27s\n- devstack-6 https://zuul.opendev.org/t/openstack/build/82dce5975ed54b00bf0f170906f8ddeb : SUCCESS in 22m 36s\n- devstack-7 https://zuul.opendev.org/t/openstack/build/89fa4af250704af3b1de30d408bfb8b4 : SUCCESS in 28m 43s\n- devstack-8 https://zuul.opendev.org/t/openstack/build/493bb2854f774b1a8d35a1eecf91c239 : SUCCESS in 24m 34s\n- devstack-9 https://zuul.opendev.org/t/openstack/build/31f122d13334468587eea18ed74cac7e : SUCCESS in 30m 46s\n- devstack-10 https://zuul.opendev.org/t/openstack/build/0d912faeca0940189396b0bfdec964cf : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"822180f3b6089e8e968cd952fd48bbcf3577f410","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-10 17:49:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bdf97a1c0c1eca283ebd0bd975cef3d7c28765c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:41:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74970139d2f84e79ae0edcb41ef3e197\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7fcd26b4f49c4f098902000aedc423ae : SUCCESS in 48m 30s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5747841953f64402a5f0f07cce978485 : SUCCESS in 1h 46m 26s\n- devstack-1 https://zuul.opendev.org/t/openstack/build/f4ed8de6638d4aa49598cdafe180f1e4 : SUCCESS in 29m 15s\n- devstack-2 https://zuul.opendev.org/t/openstack/build/238d8441b4f34aa5b84624cfe8670a55 : SUCCESS in 18m 26s\n- devstack-3 https://zuul.opendev.org/t/openstack/build/a3a990f3a4c34a4ea196d61f89cd92fe : SUCCESS in 31m 56s\n- devstack-4 https://zuul.opendev.org/t/openstack/build/5909bdd307ba4c2ebbfc47668de5cf96 : SUCCESS in 30m 57s\n- devstack-5 https://zuul.opendev.org/t/openstack/build/7b90d73180f2428397072c31c8aaa4ba : SUCCESS in 17m 23s\n- devstack-6 https://zuul.opendev.org/t/openstack/build/3168e536efa54de38595a812ba8fe8d8 : SUCCESS in 22m 04s\n- devstack-7 https://zuul.opendev.org/t/openstack/build/07190c5378e541a08ee772b0cdab5469 : SUCCESS in 32m 30s\n- devstack-8 https://zuul.opendev.org/t/openstack/build/046e68e2b193407da0173dd8305ad72e : SUCCESS in 27m 56s\n- devstack-9 https://zuul.opendev.org/t/openstack/build/2e450c4e0cc045639448abcea7ddc84b : SUCCESS in 30m 54s\n- devstack-10 https://zuul.opendev.org/t/openstack/build/5d416be03aba491dad0a55c148a1232b : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":1},{"id":"df8acd45784778de5f42a96dca389e23ed19d9e2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-15 03:47:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd47da2a201976e3e859cc4898a27d5f4a5de160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-15 06:05:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"32391b7480ce70ae39df520832704a572eef937b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 08:27:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/79020062269543f8aae083d1b30049a2\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/21ccd2dd024243cb876dbff410993ec5 : SUCCESS in 37m 09s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/41ea99563a994dd6867bc0c6e612c3a5 : SUCCESS in 1h 29m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-1 https://zuul.opendev.org/t/openstack/build/f2903765d6834a8bb56c6f8ee081ae9c : SUCCESS in 2h 03m 06s\n- cinder-tempest-plugin-lvm-lio-barbican-2 https://zuul.opendev.org/t/openstack/build/8a14e52658034df89c253b46bcd4f5f8 : SUCCESS in 1h 59m 37s\n- cinder-tempest-plugin-lvm-lio-barbican-3 https://zuul.opendev.org/t/openstack/build/64a5cac22a4a4442abe9a5dad1a36438 : SUCCESS in 1h 20m 17s\n- cinder-tempest-plugin-lvm-lio-barbican-4 https://zuul.opendev.org/t/openstack/build/5c91cd75ce4c409caa197989dab692c0 : SUCCESS in 2h 13m 49s\n- cinder-tempest-plugin-lvm-lio-barbican-5 https://zuul.opendev.org/t/openstack/build/2c054a52c4a148c3b73c09ff790d8d09 : SUCCESS in 1h 54m 00s\n- cinder-tempest-plugin-lvm-lio-barbican-6 https://zuul.opendev.org/t/openstack/build/3f549b26dc0a4a70876dca4039c53d6a : SUCCESS in 1h 31m 47s\n- cinder-tempest-plugin-lvm-lio-barbican-7 https://zuul.opendev.org/t/openstack/build/e1b68dc73ae448ad8a1fc16241d6b48b : SUCCESS in 1h 55m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-8 https://zuul.opendev.org/t/openstack/build/0e7e77315a224a5ab0dadca0925907bf : FAILURE in 2h 08m 21s\n- cinder-tempest-plugin-lvm-lio-barbican-9 https://zuul.opendev.org/t/openstack/build/b0726294c09d48d090d08fd3d92b0c72 : SUCCESS in 1h 41m 12s\n- cinder-tempest-plugin-lvm-lio-barbican-10 https://zuul.opendev.org/t/openstack/build/85c1b70f0e0141b18f977b3930969cd9 : FAILURE in 1h 53m 49s","accounts_in_message":[],"_revision_number":2},{"id":"99a87aa3e00439e0f0e60a1875faf489bbeb89f4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-15 08:27:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b28ef6a770af3fb09ce2ea14a8691ce6aa40c057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:55:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14a71b8411c0454fba81b90df7bf6b04\n\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5122fdaea2414abdb8ecb48ba516542a : SUCCESS in 48m 54s\n- cinder-tempest-plugin-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/94273ce9df434d0cb20e8abc2da4e2c2 : FAILURE in 1h 50m 54s\n- cinder-tempest-plugin-lvm-lio-barbican-1 https://zuul.opendev.org/t/openstack/build/9cfe092bfc884b49823e72908be469e8 : SUCCESS in 1h 29m 01s\n- cinder-tempest-plugin-lvm-lio-barbican-2 https://zuul.opendev.org/t/openstack/build/1c30f84fe56643ba90bc9f5a4dc22094 : SUCCESS in 1h 46m 10s\n- cinder-tempest-plugin-lvm-lio-barbican-3 https://zuul.opendev.org/t/openstack/build/1b7ef2f69aba486491c861d21e57ee5f : FAILURE in 2h 19m 03s\n- cinder-tempest-plugin-lvm-lio-barbican-4 https://zuul.opendev.org/t/openstack/build/6495476279c94b819570050a49f862e3 : FAILURE in 2h 24m 48s\n- cinder-tempest-plugin-lvm-lio-barbican-5 https://zuul.opendev.org/t/openstack/build/012cd6650bfc493e9c970243a838eb30 : SUCCESS in 1h 28m 30s\n- cinder-tempest-plugin-lvm-lio-barbican-6 https://zuul.opendev.org/t/openstack/build/c27d2c836faa4566b1472be5006eb488 : FAILURE in 1h 46m 11s\n- cinder-tempest-plugin-lvm-lio-barbican-7 https://zuul.opendev.org/t/openstack/build/7961b88b2f80455aa55e2809cfa6884e : SUCCESS in 1h 50m 40s\n- cinder-tempest-plugin-lvm-lio-barbican-8 https://zuul.opendev.org/t/openstack/build/a7e56099c7c64641920beba62afa7c85 : SUCCESS in 1h 53m 07s\n- cinder-tempest-plugin-lvm-lio-barbican-9 https://zuul.opendev.org/t/openstack/build/3ae65be51a3548c69ab5ccb611d65245 : SUCCESS in 1h 50m 28s\n- cinder-tempest-plugin-lvm-lio-barbican-10 https://zuul.opendev.org/t/openstack/build/c1cf568a9fae4759889361b9a56f4146 : FAILURE in 18m 02s","accounts_in_message":[],"_revision_number":2},{"id":"3f0ef7f0fdcc0b8cc7deeabbbae7eee1db519fe9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 04:59:06.000000000","message":"Abandoned\n\njust for test, related patch was reverted","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99b08f87e6306e3d8f0c8759a1214f4cb18721b0","revisions":{"7566308750c010cbe9a6b9c0d91aa4e30b954c98":{"kind":"REWORK","_number":1,"created":"2023-11-10 06:19:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/74/900574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/900574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/900574/1"}}},"commit":{"parents":[{"commit":"26f173423986ca6c110ee517bae8d5db844d6ad6","subject":"Merge \"ovn: Spawn metadata resources on additional chassis event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26f173423986ca6c110ee517bae8d5db844d6ad6"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-11-10 06:19:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-11-10 06:19:05.000000000","tz":330},"subject":"[DNM] Reproduce issue","message":"[DNM] Reproduce issue\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/900395\nChange-Id: Iea375aa6c3e48ed55cf49118885438b2be5c2b1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7566308750c010cbe9a6b9c0d91aa4e30b954c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7566308750c010cbe9a6b9c0d91aa4e30b954c98"}]},"branch":"refs/heads/master"},"99b08f87e6306e3d8f0c8759a1214f4cb18721b0":{"kind":"REWORK","_number":2,"created":"2023-11-15 06:05:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/74/900574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/900574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/900574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/900574/2"}}},"commit":{"parents":[{"commit":"5f2777f1374d92a0fe421e51c3e9359198e5f6f2","subject":"Merge \"Get default value for BFD/ECMP extra attributes from config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f2777f1374d92a0fe421e51c3e9359198e5f6f2"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-11-10 06:19:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-11-15 06:04:35.000000000","tz":330},"subject":"[DNM] Reproduce issue","message":"[DNM] Reproduce issue\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/900395\nChange-Id: Iea375aa6c3e48ed55cf49118885438b2be5c2b1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99b08f87e6306e3d8f0c8759a1214f4cb18721b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/99b08f87e6306e3d8f0c8759a1214f4cb18721b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
