)]}'
{"id":"openstack%2Fneutron~900665","triplet_id":"openstack%2Fneutron~stable%2Fzed~I709c477f66ada037e47ec9abc7fbed1d1a238d20","project":"openstack/neutron","branch":"stable/zed","topic":"cleanup-fullstack-tests-stable/2023.2-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-13 16:39:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I709c477f66ada037e47ec9abc7fbed1d1a238d20","subject":"[Fullstack] Drop all linuxbridge scenarios from fullstack tests","status":"MERGED","created":"2023-11-11 17:49:02.000000000","updated":"2023-11-13 16:40:58.000000000","submitted":"2023-11-13 16:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900665-cleanup-fullstack-tests-stable/2023.2-stable/2023.1-stable/zed","cherry_pick_of_change":900664,"cherry_pick_of_patch_set":1,"meta_rev_id":"6e5fb7447fe8649783e14197fe1ecd1be038b2ba","_number":900665,"virtual_id_number":900665,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 16:39:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-13 14:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-13 13:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-11-13 14:23:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-11 19:30:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 13:25:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-13 14:23:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9359f8ed9b0a0e791b8cbf4bb65b7e0aacc8cb95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-11 17:49:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"3e1de051e6a05b403c5be15e1c8f6323ee8273c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-11 19:30:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f00f5a10380843f78c495cf649962581\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d3b573e375c4324a7cd3de37d21ec40 : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d48d41c70c44c7da1ec08f38541167d : SUCCESS in 10m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fcbfe8d4dbe242149dafeb2e18a4684a : SUCCESS in 1h 35m 43s","accounts_in_message":[],"_revision_number":1},{"id":"2786986e884b51a4d099383893445893eb918072","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-13 13:25:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c23457b0d4f30c963ecd519c60feebe591afa19c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-13 14:23:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"131a5df4918b1265d05ec25af7de823a5db31a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 14:23:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca61a0191e32269c7f1f2cace58b9d6daf6c0365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:39:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348b3e2942f8495facc0fa596ba7aeaf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f51a899b6c45cba51df72aef2d231d : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b17a95f0ea604fd39721bdf8e4f8971c : SUCCESS in 11m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d2c6de192d046feb2095a169719bf58 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"59540bc0085becea8502a8fe8eadde9fd8634aa5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:39:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e5fb7447fe8649783e14197fe1ecd1be038b2ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:40:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac16e11be7eb4ff0a1bbc76fca5b1a63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec099da672241eaa2a6c91bd2365d30 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e38f2ab9a6b439ac1b739a39ec52b4e6771f631","revisions":{"0e38f2ab9a6b439ac1b739a39ec52b4e6771f631":{"kind":"REWORK","_number":1,"created":"2023-11-11 17:49:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/65/900665/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/900665/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/900665/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/900665/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/900665/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/900665/1"}}},"commit":{"parents":[{"commit":"028495984fdfa1419ec73d313d37046c2a878c62","subject":"Merge \"Reduce lock contention on subnets\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/028495984fdfa1419ec73d313d37046c2a878c62"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-11-09 09:15:51.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-11-11 17:49:02.000000000","tz":0},"subject":"[Fullstack] Drop all linuxbridge scenarios from fullstack tests","message":"[Fullstack] Drop all linuxbridge scenarios from fullstack tests\n\nAs we discussed on the Neutron CI meeting on 7.11.2023 (it was video\nmeeting, no strict logs stored) [1], to save some time in the fullstack\njob execution we want to stop running all the Linuxbridge related\ntests/scenarios in fullstack test suite.\nLinuxbridge is experimental feature now, not really fully supported by\nthe Neutron team so we are allowed to do so in this case.\nThis patch is doing exactly that. There are no any Linuxbridge agent\nrelated tests/scenarios.\n\nSupport for Linuxbridge in the Fullstack framework wasn\u0027t however\nremoved completely. It is like that because of 2 reasons:\n\n* there is one DHCP agent related test which tests different segments\n  and is using Linuxbridge agent on one of the fake nodes,\n* maybe there will be at some point need to have some Linuxbridge agent\n  test there so the infrastructure for that is still available, it won\u0027t\n  cause any issues while it\u0027s (almost) not used currently.\n\n[1] https://meetings.opendev.org/meetings/neutron_ci/2023/neutron_ci.2023-11-07-15.00.html\n\nChange-Id: I709c477f66ada037e47ec9abc7fbed1d1a238d20\n(cherry picked from commit 688c5e2a911410260a34817a7e15a88bfd465fef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e38f2ab9a6b439ac1b739a39ec52b4e6771f631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e38f2ab9a6b439ac1b739a39ec52b4e6771f631"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
