)]}'
{"id":"openstack%2Fneutron~900883","triplet_id":"openstack%2Fneutron~stable%2Fzed~I6203009750a4e589eeb808f842cb522d61476179","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-11-14 15:59:57.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-15 00:05:59.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-11-15 00:05:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6203009750a4e589eeb808f842cb522d61476179","subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","status":"MERGED","created":"2023-11-14 11:13:46.000000000","updated":"2023-11-15 00:07:21.000000000","submitted":"2023-11-15 00:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900883","cherry_pick_of_change":900647,"cherry_pick_of_patch_set":1,"meta_rev_id":"fed811dead43277a1a35baaa2afbe2f4f8d654ae","_number":900883,"virtual_id_number":900883,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 00:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2023-11-14 15:59:57.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-11-14 19:46:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-14 11:54:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-11-14 19:46:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 11:13:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"CC"},{"updated":"2023-11-14 11:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-14 11:54:58.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-11-14 13:16:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 13:34:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-11-14 13:34:25.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-11-14 19:46:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"32ca7a90d26be304f994b2daa573b99fa309387b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-14 11:13:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py","accounts_in_message":[],"_revision_number":1},{"id":"ced83d08687cee925380c7005a828180eedae1ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:20:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4793dba5620149529cb196f273416b5a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99b741fc2ef469e9955bed91ace4898 : FAILURE in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03be53e0dc654524818bddc7c3ca8df5 : FAILURE in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07c9c2e744a0cb8308832080f5e99c71c15fc698","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-14 11:25:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a8185c5a7247b1f736860c88c3f19cb46b4ad578","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-11-14 11:27:38.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"e7b95db16cbb74313216d6072b68539383a56f59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:45:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f4a1a91b08a416a94061a3f826b6e56\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2533b403faf4c35a526a41e5ded3d00 : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e951b7a09df54d89ba119cce72bc95b4 : SUCCESS in 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e5ddba171e3d7fc79b340e9c747adaafbdcb7e71","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-11-14 11:54:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2265b8ef5b162ab4ca7e9e802cc88d0019e16e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 13:16:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7802256e01f949dabe7affdffe14e361\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/cdbdb9a3587d4451bfe65bcf9a45b43a : SUCCESS in 1h 48m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9668e1eb19da40c9ad6966fd597e7599 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3113093a0c8f469eb3f8717f4ef49743 : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef9b6b5d4b7047b88d59658d912dca08 : SUCCESS in 24m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1538a769c41472881921a8fae31fc8e : SUCCESS in 22m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f367547d5b5242a3840702c07d1fdd8a : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8040586339284cc8a32a67e4f8edc308 : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a528e68029634b23bf718aaf726c9f8c : SUCCESS in 42m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ab8b98818ed41a28b3edeb519941992 : SUCCESS in 44m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6838bf59a2464092a0fc66b31b65ae77 : SUCCESS in 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"700eaf27316ea7caaf80bf912888e64d5d99b403","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-14 15:59:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9cdc4f3ef8a36345224a075cfd5f91a957cde7d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-14 19:46:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"90914a79655b39f5840bb642f60ce49e03e93084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:48:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cbc4ad545515a98ab61d6f5c43329b4ec6882b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:05:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abdae246e5754c3297384e1078abac33\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/510dd2a315744562b45c276e3fa6325a : SUCCESS in 2h 00m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628e53472bba4512a770551bd3c89a7d : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb34e1dffd9d400481abe7ad17b7e4e6 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8df7eebb5ecd4aefbe028753b611de5d : SUCCESS in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e806063483974232b291876252cbd45b : SUCCESS in 11m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b8fcdfbb7e54abf9776d88d05b36caf : SUCCESS in 41m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/526518a8878c40d9a35808c03e7da463 : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c3551fe2b2aa1a6a893cac9b210405e6e50d6c38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:05:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fed811dead43277a1a35baaa2afbe2f4f8d654ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:07:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771aee8e3c5e4be7881ced221de676f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2866f61f9a4d5bbd74ded04d4ab15c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15d5db0d6e0a563511f13bc6b014bf07c7c8db38","revisions":{"2241a2ae7c5f1579038704c00bf3e2e32e9f0624":{"kind":"REWORK","_number":1,"created":"2023-11-14 11:13:46.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/83/900883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/900883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/900883/1"}}},"commit":{"parents":[{"commit":"0e38f2ab9a6b439ac1b739a39ec52b4e6771f631","subject":"[Fullstack] Drop all linuxbridge scenarios from fullstack tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e38f2ab9a6b439ac1b739a39ec52b4e6771f631"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-11-13 15:42:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-14 11:13:46.000000000","tz":0},"subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","message":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\n\nWhen neutron server restarts the mac address for NAT entries related\nto ovn-lb FIPs gets re-added, distributing the traffic that should\nbe centralized and therefore breaking the connectivity. This happens\ndue to the port being down. This patch is ensuring the MAC entry\nis only being readded in case the port is UP\n\nCloses-Bug: #2042938\nChange-Id: I6203009750a4e589eeb808f842cb522d61476179\n(cherry picked from commit f2a3020cf0a46dbd896c5f7b4b4f6643d32a6b4a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2241a2ae7c5f1579038704c00bf3e2e32e9f0624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2241a2ae7c5f1579038704c00bf3e2e32e9f0624"}]},"branch":"refs/heads/stable/zed"},"15d5db0d6e0a563511f13bc6b014bf07c7c8db38":{"kind":"REWORK","_number":2,"created":"2023-11-14 11:25:43.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/83/900883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/900883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/900883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/900883/2"}}},"commit":{"parents":[{"commit":"0e38f2ab9a6b439ac1b739a39ec52b4e6771f631","subject":"[Fullstack] Drop all linuxbridge scenarios from fullstack tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e38f2ab9a6b439ac1b739a39ec52b4e6771f631"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-11-13 15:42:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-11-14 11:25:43.000000000","tz":0},"subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","message":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\n\nWhen neutron server restarts the mac address for NAT entries related\nto ovn-lb FIPs gets re-added, distributing the traffic that should\nbe centralized and therefore breaking the connectivity. This happens\ndue to the port being down. This patch is ensuring the MAC entry\nis only being readded in case the port is UP\n\nCloses-Bug: #2042938\nChange-Id: I6203009750a4e589eeb808f842cb522d61476179\n(cherry picked from commit f2a3020cf0a46dbd896c5f7b4b4f6643d32a6b4a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15d5db0d6e0a563511f13bc6b014bf07c7c8db38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15d5db0d6e0a563511f13bc6b014bf07c7c8db38"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
