)]}'
{"id":"openstack%2Fneutron~900968","triplet_id":"openstack%2Fneutron~master~Iec784889ace233b9b4383cb79eae33ae3166129b","project":"openstack/neutron","branch":"master","topic":"distributed_metadata_data_path","hashtags":[],"change_id":"Iec784889ace233b9b4383cb79eae33ae3166129b","subject":"[tempest] New case for distributed metadata data path","status":"ABANDONED","created":"2023-11-15 01:32:54.000000000","updated":"2023-11-17 03:39:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba36782f39bb0d0222dd16fd9e4b9e3fb03294d7","_number":900968,"virtual_id_number":900968,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-15 10:49:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 04:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3961306c8fe60807bf56b36a801acae6e546cde9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-11-15 01:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c371fa9a43d640d0c07ce5388051d919ca796d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 04:08:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5125328104344c76a928e08b452e5f60\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5b43a76aba147f3bf3d6c291cad734e : SUCCESS in 1h 38m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d735196e9a78417fa7ecec3ded16f8ec : SUCCESS in 1h 46m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c3bb9bfff97b4cf19aa35a7b6fb5b936 : SUCCESS in 1h 20m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/851161c556a745d0bd5da49858ace3c3 : SUCCESS in 1h 45m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ab904aba6ad438183011f1f81041f30 : SUCCESS in 41m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57822e3e35a34785a71d653176844353 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bcfa2a013f743a1a225850226c906aa : SUCCESS in 12m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b6a590881e754b83a9918e0815482746 : SUCCESS in 1h 09m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3073ddedd63a49009362bba01fdc4a9d : SUCCESS in 57m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1606e457d2924f5ab39ac170e1c8dc9c : SUCCESS in 57m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8769826f769946d9bd447c97f4a121dd : SUCCESS in 1h 52m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0c99b305776419ca98c49da877ea04c : SUCCESS in 1h 15m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eee2ed8ea4244cdabe225163ebc75aec : SUCCESS in 54m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ab207ba11e654836aab642326120b7ac : SUCCESS in 1h 36m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53e31689e8ae4e968519cf7bb2e20640 : SUCCESS in 1h 23m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4a83e165eef343e8b89cd882bde3bed4 : SUCCESS in 54m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/331b0081bb854d23bf52e4c2a0a1cd16 : SUCCESS in 1h 58m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df30dab81ac24044936390bf7c58b1b8 : SUCCESS in 2h 29m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9d0651a0bbfb49ec9521665725751e03 : SUCCESS in 1h 36m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a094b74fe64657658692f095ecc49025f14eb5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-11-15 06:54:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1ecb774f9468a91cf0eb4ce133ac994e951d8797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-11-15 08:36:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee984a5098f08c9b0583af7b9d4443385b03382f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:49:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1ee4fa20f164516a8487e4bd19b7d27\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/095cce8870314072a504393b0dd98b39 : SUCCESS in 1h 39m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de099faa834b4e1a86302b8699e17624 : SUCCESS in 2h 04m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/bbc32bfa86cc4ecc84cb581dd99a4752 : SUCCESS in 2h 10m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6e2beda528b414098046649ad387696 : SUCCESS in 1h 50m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb6e175fb2b04f2a9278a1ffe23faba1 : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c93238a36d454056b448f500e4678d10 : SUCCESS in 9m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d60b122211e4ce7b5701e34ddb6b381 : SUCCESS in 11m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9391315832164d54a99ca51bfd82d339 : SUCCESS in 55m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/22163f23080345c98727cc85339a517b : SUCCESS in 1h 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a65c8526764c4408b0720841182f378d : SUCCESS in 56m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba2376f3197e499caa4ab63b60b96dcb : SUCCESS in 1h 47m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74876de83e0f46058e9516234944e4cb : SUCCESS in 1h 01m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4063728b3bc84046a257aee96a61e0c2 : SUCCESS in 49m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/281315a6b3074f58a20c8317876f4591 : SUCCESS in 1h 19m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ebb48c492a0048b282a79aa07e0f5ed3 : SUCCESS in 1h 20m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f9fda82bb4246d2a29f3fc43fc25d32 : SUCCESS in 1h 02m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d291a3b13292416b968156639c1b59a0 : SUCCESS in 1h 53m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dac3ee15e404414a8420e29361ee7071 : SUCCESS in 1h 51m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5570c0ad53aa437c92c572f98a3ff8c1 : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":3},{"id":"ba36782f39bb0d0222dd16fd9e4b9e3fb03294d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-11-17 03:39:15.000000000","message":"Abandoned\n\nzuul\u0027s openstack env is all-in-one, so 0.0.0.0:80 is used. Host haproxy can not start.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8dc0468a26aef8e001f34c4dec00a677d361dc1d","revisions":{"635350f707249c7153117b5973d16c1711b318aa":{"kind":"REWORK","_number":1,"created":"2023-11-15 01:32:54.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/900968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/900968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/900968/1"}}},"commit":{"parents":[{"commit":"15a34d0dc515cf17eeaeb09fbf372340ccb899c5","subject":"Add metadata path extension openflows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a34d0dc515cf17eeaeb09fbf372340ccb899c5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 01:05:14.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 01:14:21.000000000","tz":480},"subject":"[tempest] New case for distributed metadata data path","message":"[tempest] New case for distributed metadata data path\n\nAdd config option ``neutron-distributed-metadata-data-path`` for\nneutron-tempest-plugin.\n\nChange-Id: Iec784889ace233b9b4383cb79eae33ae3166129b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/635350f707249c7153117b5973d16c1711b318aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/635350f707249c7153117b5973d16c1711b318aa"}]},"branch":"refs/heads/master"},"9b60dc2123f84538359369f52528739af7e03e5c":{"kind":"REWORK","_number":2,"created":"2023-11-15 06:54:51.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/900968/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/900968/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/900968/2"}}},"commit":{"parents":[{"commit":"15a34d0dc515cf17eeaeb09fbf372340ccb899c5","subject":"Add metadata path extension openflows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a34d0dc515cf17eeaeb09fbf372340ccb899c5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 01:05:14.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 06:50:54.000000000","tz":480},"subject":"[tempest] New case for distributed metadata data path","message":"[tempest] New case for distributed metadata data path\n\nAdd config option ``neutron-distributed-metadata-data-path`` for\nneutron-tempest-plugin.\n\nChange-Id: Iec784889ace233b9b4383cb79eae33ae3166129b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b60dc2123f84538359369f52528739af7e03e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b60dc2123f84538359369f52528739af7e03e5c"}]},"branch":"refs/heads/master"},"8dc0468a26aef8e001f34c4dec00a677d361dc1d":{"kind":"REWORK","_number":3,"created":"2023-11-15 08:36:19.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/68/900968/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/900968/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/900968/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/900968/3"}}},"commit":{"parents":[{"commit":"15a34d0dc515cf17eeaeb09fbf372340ccb899c5","subject":"Add metadata path extension openflows","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15a34d0dc515cf17eeaeb09fbf372340ccb899c5"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 01:05:14.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2023-11-15 08:35:49.000000000","tz":480},"subject":"[tempest] New case for distributed metadata data path","message":"[tempest] New case for distributed metadata data path\n\nAdd config option ``neutron-distributed-metadata-data-path`` for\nneutron-tempest-plugin.\n\nChange-Id: Iec784889ace233b9b4383cb79eae33ae3166129b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8dc0468a26aef8e001f34c4dec00a677d361dc1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8dc0468a26aef8e001f34c4dec00a677d361dc1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
