)]}'
{"id":"openstack%2Fneutron~900982","triplet_id":"openstack%2Fneutron~stable%2Fxena~I6203009750a4e589eeb808f842cb522d61476179","project":"openstack/neutron","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-11-16 09:09:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6203009750a4e589eeb808f842cb522d61476179","subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","status":"MERGED","created":"2023-11-15 06:40:36.000000000","updated":"2023-11-16 09:10:48.000000000","submitted":"2023-11-16 09:09:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900982","cherry_pick_of_change":900885,"cherry_pick_of_patch_set":1,"meta_rev_id":"5e01bf0bb6a10563465060edf63e907dcde40ca0","_number":900982,"virtual_id_number":900982,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 09:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-16 08:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-15 14:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-11-16 08:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 07:24:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-15 09:15:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-15 14:19:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-11-16 08:01:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0ad281e3cf6360d968ae8cfe0f4c078f0614f8c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-11-15 06:40:36.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"c79bc12e787bdcd7ea5d3adc81d4949e83146adc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 07:24:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/979312c27b894ef6bc27bfa926c1fb7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/afe07008ebdc4d10b041dfee8a370abc : SUCCESS in 38m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b11198426b2b49c7bc31c108b1ed45b6 : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"067195aadce388a135ab4c036cc2a21468b7d659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 09:15:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ebc2b93a7ee493b867f27f01dcbb8a0\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/9f9a11a454e14979a13ed55b757a3bb8 : SUCCESS in 51m 02s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/cd20b9f0804d46caa479d64dd4873085 : SUCCESS in 2h 27m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03025f8a8e844c75ac1464752f663a6e : SUCCESS in 37m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3880969cd4304cceaecc29a6d6d75a9b : SUCCESS in 10m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af701e239c7740b9977badd5c7f04a80 : SUCCESS in 21m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b846058d5e844274b5fe60c0763bde90 : FAILURE in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e0b1e383da84ed3936b8649b322f0f7 : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23fa99602d564820928b6335aa5000ea : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1316ac462654d48824bbaa03543c61b : SUCCESS in 37m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f772da3be8b4e1b940eefe31de0c8bf : SUCCESS in 50m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c55e21ca26246338da329ed138e1b49 : SUCCESS in 54m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80f74743098f929fe5703d2482bc4e4800d64425","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-11-15 09:38:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbdc77acd8a2387b998f977b843dfc6287068fb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:02:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fd5ec658044b3981b609e1497092fc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b541e0ce4283435b8de8be8a13cae53c : SUCCESS in 21m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9c6ff32709b6471eaaa62e3ec9c733a6 : SUCCESS in 23m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7572e775d178a1dff095a825055787d872a3ac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 11:40:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eec37959bbe643f6ac744357198873bb\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/43dcab65ef564cdcbc7b853ca44d3d04 : SUCCESS in 48m 43s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/caa73205d92946e18abc231f9974cff5 : SUCCESS in 1h 48m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60d81114f77248f685c215aa76591a54 : SUCCESS in 36m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953402c99b834669b1fc8b2b8650c2d3 : SUCCESS in 12m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28daff8bec0e4d71a57f1efdaac24a86 : SUCCESS in 26m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dee07407e3a449fbb35464412ca5108 : SUCCESS in 27m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c1efc6aab34c38a433f9a8478759fd : SUCCESS in 28m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f100ee69575a4a2aac25d98b5436d62f : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/924260fc0f5f4dc08fd30509571f7c28 : SUCCESS in 50m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/065f957f0e3b4c9a8a0a0fca7e69d742 : SUCCESS in 47m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eee69b61ec2d4f0ca3297322b04eff79 : SUCCESS in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f09cf4cfb25806d483e8f49396519516dd6f2dfa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-15 14:19:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7d05efa0852b3d660e9667c599c0e78ea28d8cf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-16 08:01:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"72c54ddf5330cbdfbf1fe7ae54502edbd7e5398b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 08:01:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"748fdff80cdadbd779fe8cc6448066a66e8d20a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:09:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081b3044f6f04d76a25b0c641ead60e1\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/ae434b24d81845e4bbeb1a9f42c28cbe : SUCCESS in 1h 00m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28df68aead314414947c4dfa832fa647 : SUCCESS in 14m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca92f92940f0478291dca6d143db05f4 : SUCCESS in 27m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40fa5bf9ecfa45e0a3285c5943c1f3ce : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c1ac38bd534e3382c881b0a28d2f33 : SUCCESS in 8m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7380044c3d064dfc873785b1cd2d7562 : SUCCESS in 47m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfc5f0469a2f4ce4a535d2ddf813f42e : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9a16f31ef92a1c252c0cb9f5f4cbe02c0a9e656a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:09:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5e01bf0bb6a10563465060edf63e907dcde40ca0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f93918b6f84098ace488704c8c7268\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6797a9c414bf414ab94bfc006530a15a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86aa8a3a1031cbc6476399308a8f6344e9383652","revisions":{"86aa8a3a1031cbc6476399308a8f6344e9383652":{"kind":"REWORK","_number":1,"created":"2023-11-15 06:40:36.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/82/900982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/900982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/900982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/900982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/900982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/900982/1"}}},"commit":{"parents":[{"commit":"d266c08b86fa3487efd5cb585917135cb04cc0c3","subject":"Merge \"[DHCP agent] Add route to OVN metadata port if exists\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d266c08b86fa3487efd5cb585917135cb04cc0c3"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-11-13 15:42:51.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-11-15 06:40:36.000000000","tz":0},"subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","message":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\n\nWhen neutron server restarts the mac address for NAT entries related\nto ovn-lb FIPs gets re-added, distributing the traffic that should\nbe centralized and therefore breaking the connectivity. This happens\ndue to the port being down. This patch is ensuring the MAC entry\nis only being readded in case the port is UP\n\nCloses-Bug: #2042938\nChange-Id: I6203009750a4e589eeb808f842cb522d61476179\n(cherry picked from commit f2a3020cf0a46dbd896c5f7b4b4f6643d32a6b4a)\n(cherry picked from commit bef1f7a4861a1d19e3a9003ffed303dd47b70707)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86aa8a3a1031cbc6476399308a8f6344e9383652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86aa8a3a1031cbc6476399308a8f6344e9383652"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
