)]}'
{"id":"openstack%2Fneutron~901073","triplet_id":"openstack%2Fneutron~stable%2Fzed~I083649ab1b9a9057ee276a7f3ba069eb667db870","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2030804-stable/zed","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-27 22:36:36.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I083649ab1b9a9057ee276a7f3ba069eb667db870","subject":"Catch non-existent entry failures better in ip_lib","status":"MERGED","created":"2023-11-16 15:20:16.000000000","updated":"2023-11-28 02:18:15.000000000","submitted":"2023-11-28 02:15:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901073-bug/2030804-stable/zed","cherry_pick_of_change":891236,"cherry_pick_of_patch_set":6,"meta_rev_id":"4f1b978988b5645eef0788183659368fbb9a96ce","_number":901073,"virtual_id_number":901073,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 02:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-21 15:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-22 08:18:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-11-22 08:18:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 15:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-16 16:25:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 15:23:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-11-22 08:18:12.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"ac2a97803ef90ee9a468a2146f6bfbc517563ffb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 15:20:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* neutron/privileged/agent/linux/ip_lib.py","accounts_in_message":[],"_revision_number":1},{"id":"2b2be1e2245d1ab33fd1ec6a8e7df2a86406cc38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:37:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/248b43a9de3f4ab789332faa817577fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ad3bfcc63014380a5e373077670bbd5 : FAILURE in 5m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddf2efac05994c30b4736e8c31692b69 : FAILURE in 14m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02d415060347e318a620269883fc50c8d422f7bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 16:25:57.000000000","message":"Patch Set 1: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c63cf1cf90574bc2991f66b89416a285\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/ec2b3036e074499fb8093c351ad7f93a : FAILURE in 17m 16s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/825148cb84d44c418a1680c4680e6347 : FAILURE in 16m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/bf38c2e2599a43c09e70de70583527f9 : FAILURE in 16m 57s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/4195b716c02e4b27bbf0c1d4c08e39f8 : FAILURE in 20m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55a9983a8257427fab9714ba521faa40 : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8466aa2570a4e83a30040d9941ff4d8 : FAILURE in 8m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee2c326130c4b44a49ddc785c452f2a : FAILURE in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/606002faf0ac4beda2f998f7e53d2312 : FAILURE in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9ae4c939cc64f25a60cb9a4ca1f0051 : FAILURE in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f78983fb91241d095837a801432123d : FAILURE in 8m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89a35e92cd08449e8e4df1805c2d8ca3 : FAILURE in 16m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5efed1bc8a84106bd1386c717369804 : FAILURE in 12m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a8b534c17a4b48159239e5a13897f0ac : FAILURE in 17m 22s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/76301bc156054f139d878752fead3551 : FAILURE in 56m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6cdcb205ec24cb2a7f2ce0f4888ecbb : FAILURE in 50m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9ed6d5ce1ee64cc0891b55b2656ec0e5 : FAILURE in 32m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2d9eab33a58f426380e52f1a4e5780d6 : FAILURE in 17m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa19c25dd554023a0f088a25020bd3e : FAILURE in 14m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7c4a21d5d784ae9b75f6679469824ac : FAILURE in 19m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"461415a43056ef4b061fdff3c0c2213c1d19bd19","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 17:21:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"33e373d526510f0ba0507df7dd66c4d28776090d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 17:21:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"62ef4ae59fee73ce1a648e74998ebce5741e2f6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d41aeb94f32a4902a6acab42c14ada34\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eebfdfd724c8415099150a8433cc849f : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/293374b194d245b3b76df7aea2a52a9b : SUCCESS in 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ed090cc796e46816a886b97d7dcf21f74e1554c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:37:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e52ee6e13c224a35ae30bc05200f5cc5\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/363b3069f2e7412ba95860bdd34664fd : FAILURE in 2h 10m 12s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/002993bef5dc40a1854f1b97328237a5 : SUCCESS in 1h 43m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/16d7c853bf1b44fa862736debf6b4eb1 : SUCCESS in 2h 09m 28s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e4b3e1cda3a740369ff3c6e876b68ff7 : SUCCESS in 1h 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a634a30ce9c34efb8b7496e23156f08e : SUCCESS in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2270768b44ab48c9b4e12929f6a28c41 : SUCCESS in 13m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed05920d8e7438bbe73d8dc083e1cef : SUCCESS in 24m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1668640b377347238ca6cebc9c60df5e : SUCCESS in 22m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac8289a3b99341419796d591793304f4 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256680a7fbd3402e94b0399ea494745d : SUCCESS in 11m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dbf57c7e6194dd4aeb770066ab79676 : SUCCESS in 51m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/68ee148a0e014a88935c15f1a20c2567 : SUCCESS in 1h 45m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f9b740bf5b8941b9b4a3f9470081f5aa : SUCCESS in 52m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/062ddd3c197f4a19ba4aa6f18999cce2 : SUCCESS in 46m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1506ed90ba824cfb84a5cf650791ae99 : SUCCESS in 47m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3db7aa76ff1143b6b8be810d3ff37cac : SUCCESS in 1h 03m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3e2787ec6f7f435e917f6d31df41bf49 : SUCCESS in 1h 15m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99676fa95a0d49bca888cfcb6fe21d6c : SUCCESS in 42m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5cb80513f6934bf8ba7b4afc3948575c : SUCCESS in 55m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf0a973d0e81c0b2a37504589b969975f8c58c13","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-16 19:51:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0228882352a2b05eca967fbee91fbec5a92e8daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:31:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be2a7af907944401a508e650e273924e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dc3c804dfb042348ea8665d2984daca : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0a9399339df466db1224bbad4342508 : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89b83dee54f6c6e99af3effa51b2e473ad37dabc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:33:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3518d7616ea04d1d90c92a1394abe1ed\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/db9cc4cc50944dc293acecbdbaafff6b : SUCCESS in 1h 33m 32s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/73eb3c23bb3c419bbaedce760ff4c8f6 : SUCCESS in 1h 15m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/6aad9229ca6c43ce97228fe19bb4c379 : SUCCESS in 1h 36m 09s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/971989d283ae49a393e7cc2443af9f47 : SUCCESS in 1h 25m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc244b41bd4c4a3d9c19dd4c4444f34f : SUCCESS in 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7424b94d50824428a0e1d93207fd2cae : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a607ce99a81a4abdacb30007048f0f31 : SUCCESS in 26m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f01be2b6e7241009825b7e66d68e772 : SUCCESS in 22m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01a0c08ca2134d01b82df3267c80c4f8 : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f71789366e74f148b3b2668b369d4e1 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/59ec6741bd104dceba203a12b7ba4157 : SUCCESS in 49m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c23b0123d5e4022acf52c01ead784dd : SUCCESS in 1h 40m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7d120dab80cf48af8b8a83bcd71aa2bb : SUCCESS in 48m 35s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3deeb95a69e14c6294b6f9955de4b45b : SUCCESS in 46m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/81e44467d7a844d8a400b9082d26f5fb : SUCCESS in 51m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3d425c06f56d4cd2adaf932c378d1ec4 : SUCCESS in 52m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ff67ccb2cc4453dad0038536e73d38e : SUCCESS in 55m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8edfd0634fb43f884fe9e2471351452 : SUCCESS in 1h 09m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c8a941ab85bc49799753547f3d9de1b3 : SUCCESS in 1h 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18aacc00682feb6b61aac21fd12efb4a876aeeea","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-21 15:23:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74434af4c3377b4764a3849128f39418a5bb1b30","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-22 08:18:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6be5b8ca9f642319c5fdaf08189a4fdadc8c70cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 08:18:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aec75f9791ecf8d935a1a2c8649594b1ece18b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 10:01:08.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c6afa5265e241ce9284e68456a5acda\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/3ca51396469c4068833d26a330536adb : SUCCESS in 1h 37m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9a4e4dac2443e5bf42a847710097f7 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/123ab91764f849ea8db3a5b4e2d5631d : SUCCESS in 23m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/196306cb38574f178656ad7efc795279 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46682a4003ca43d89aad326f8dba221b : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/38377e3f6b3a4e13b00683b86fe89f2f : FAILURE in 52m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/73442aaabf4144fb9101425a0056bb97 : SUCCESS in 1h 37m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2fedfd426d124b03bd32257de2ef3c2b : SUCCESS in 52m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd9f8802f2034b4aac8eaa4f3a30b42d : SUCCESS in 1h 00m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d238fae7323d409dbe499094f4e58dc1 : SUCCESS in 1h 07m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd3816f0ec54457ba50f5c844f620dc9 : SUCCESS in 40m 35s","accounts_in_message":[],"_revision_number":2},{"id":"374eca3122df80df83913cf54047f8cf7449a865","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-27 22:36:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af3e385797b5e62cc873f245c081ba6612ab2671","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:16:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd988aaef53464bafcd6315af6e554d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/466f90e14edb4f3ca2f8597e3f521350 : SUCCESS in 37m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f291f10c54f48959756e1102cdd3ce4 : SUCCESS in 18m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf0c96affeb8e77ffba74f0672742273c738ffc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:27:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c4d47e98b244473b2966b809bb53fa3\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/fe4cd27c4e644d6b8ea07be77ab878c1 : SUCCESS in 1h 50m 05s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/79ff3c69bed44fc593cc3bd28f456a63 : SUCCESS in 1h 11m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/f5177e4c74b24990b484097c39da97e9 : SUCCESS in 1h 21m 51s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e37c2fc28add427aac6038e9506c4e7c : SUCCESS in 1h 37m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ba4ce32193b4e419280710c12659b8e : SUCCESS in 31m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc696ecb4bce4d8984392988b90ade26 : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b2208d971694e56be4bbe0c1d2ae0ed : SUCCESS in 22m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81047e51ffb74fe589355520df9344a4 : SUCCESS in 24m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/562be1662eb446d3b57bc106b6c2c8ff : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce9487333cab43b488bcc18ed0ec59d1 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0556e19cd3284b2c8535c5e2527179cf : SUCCESS in 52m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4d1e61a92954c359698ef18521c0af8 : SUCCESS in 1h 24m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3ad7692d40e24a5cb0fe80a5f5a7e8e7 : SUCCESS in 1h 12m 46s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6822f90ee2a24a1c9d5337f5a03b5cf2 : SUCCESS in 59m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f4f5868fc404ad2a26e9224809108ea : SUCCESS in 48m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2242d264e16a4027875e0c51e8d661d0 : SUCCESS in 1h 20m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8b0022110554cffb3cd832e2a700c32 : SUCCESS in 1h 07m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f91777ae9e74e85a4c0bab15c1c69da : SUCCESS in 58m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43b7704643ec495b891b42f32f65b3e1 : SUCCESS in 1h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f32ca7f55691a335a5cb16f8f198c83fc636fb7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:28:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"775b1583674b9523cf5c8db96e65796223f8c7fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:15:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/198b740d49744d03aa2f435f562c40be\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/373c47a37c954eee92694a0aedd33d15 : SUCCESS in 1h 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bebd8f33c144cceb379f1ad8670d577 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d68369494ca24909b12ab5273f7b540b : SUCCESS in 17m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/136a693ea4074db09693515481579327 : SUCCESS in 24m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e474da1bdb403ab117bcc9080497f5 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bffe11ebe59a45d4b0b2b337ffa91e68 : SUCCESS in 1h 02m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/166f005747f14bf38a001a17f771ad3c : SUCCESS in 1h 41m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b45b156e632b4dff8a3727e556fc22ab : SUCCESS in 1h 13m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5026f130edcf4d7ebe1f93993d1fe663 : SUCCESS in 41m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3bc310444e1a4ce4846b19447a8dcb21 : SUCCESS in 48m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ab2671f0cc44c839d08211b76c133f0 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":2},{"id":"94e349f9a2a7bb2bb43f6c9bdc5715a172a0e5cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:15:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f1b978988b5645eef0788183659368fbb9a96ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:18:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e758450cec764dc7946db5c655cbe7f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2e01a6cd4a43d1a48557ea49215629 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"61ad633a331bf72e4081e0a12abb4463478efa7c","revisions":{"f2e8488559c90bc3401caab21d3401a82959df1a":{"kind":"REWORK","_number":1,"created":"2023-11-16 15:20:16.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/73/901073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/901073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/901073/1"}}},"commit":{"parents":[{"commit":"3ab7d6e6b3c7328ce860144efba3f85b5538eb86","subject":"Merge \"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab7d6e6b3c7328ce860144efba3f85b5538eb86"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-11 21:05:49.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-16 15:20:16.000000000","tz":0},"subject":"Catch non-existent entry failures better in ip_lib","message":"Catch non-existent entry failures better in ip_lib\n\nThe privileged/agent/linux/ip_lib.py code was not always\ncatching \"entry does not exist\" type errors when deleting\nentries, and most of the callers were not catching it either,\nwhich could lead to random failures.\n\nAdd code in the IP route, rule and bridge fdb code to catch\nthese errors and not raise on them, other exceptions will\nstill be raised.\n\nAlso fixed delete_neigh_entry() to not raise when the\ngiven namespace does not exist to make it like all the\nother calls in the file.\n\nAdded or modified functional tests for above cases.\n\nChange-Id: I083649ab1b9a9057ee276a7f3ba069eb667db870\nCloses-bug: #2030804\n(cherry picked from commit 16875b5f92731a9cf2d7e819d406bfcc442339f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2e8488559c90bc3401caab21d3401a82959df1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2e8488559c90bc3401caab21d3401a82959df1a"}]},"branch":"refs/heads/stable/zed"},"61ad633a331bf72e4081e0a12abb4463478efa7c":{"kind":"REWORK","_number":2,"created":"2023-11-16 17:21:00.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/73/901073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/901073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/901073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/901073/2"}}},"commit":{"parents":[{"commit":"3ab7d6e6b3c7328ce860144efba3f85b5538eb86","subject":"Merge \"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab7d6e6b3c7328ce860144efba3f85b5538eb86"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-11 21:05:49.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-16 17:20:06.000000000","tz":-300},"subject":"Catch non-existent entry failures better in ip_lib","message":"Catch non-existent entry failures better in ip_lib\n\nThe privileged/agent/linux/ip_lib.py code was not always\ncatching \"entry does not exist\" type errors when deleting\nentries, and most of the callers were not catching it either,\nwhich could lead to random failures.\n\nAdd code in the IP route, rule and bridge fdb code to catch\nthese errors and not raise on them, other exceptions will\nstill be raised.\n\nAlso fixed delete_neigh_entry() to not raise when the\ngiven namespace does not exist to make it like all the\nother calls in the file.\n\nAdded or modified functional tests for above cases.\n\nConflicts:\n  neutron/privileged/agent/linux/ip_lib.py\n  neutron/tests/unit/privileged/agent/linux/test_ip_lib.py\n\nChange-Id: I083649ab1b9a9057ee276a7f3ba069eb667db870\nCloses-bug: #2030804\n(cherry picked from commit 16875b5f92731a9cf2d7e819d406bfcc442339f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ad633a331bf72e4081e0a12abb4463478efa7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ad633a331bf72e4081e0a12abb4463478efa7c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
