)]}'
{"id":"openstack%2Fneutron~901474","triplet_id":"openstack%2Fneutron~master~I81bc5f3d98f752d926a243cbd17b8b894f2bdf58","project":"openstack/neutron","branch":"master","topic":"remove-ovs_integration_bridge","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-11-28 06:47:36.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-01 18:20:23.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-01 13:29:52.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-12-01 18:20:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81bc5f3d98f752d926a243cbd17b8b894f2bdf58","subject":"Remove ovs_integration_bridge configuration option","status":"MERGED","created":"2023-11-20 18:29:55.000000000","updated":"2023-12-01 18:21:55.000000000","submitted":"2023-12-01 18:20:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901474-remove-ovs_integration_bridge","meta_rev_id":"805837afcb7597e340891ca819e9077972bc68fb","_number":901474,"virtual_id_number":901474,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-11-23 07:38:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 18:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-28 06:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-11-23 07:07:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-11-23 09:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-27 07:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-11-27 08:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-28 06:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 19:03:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 19:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 21:09:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 02:45:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-11-22 05:22:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2023-11-22 05:26:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-11-22 08:13:00.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-11-22 20:02:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 09:18:31.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-27 07:57:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-27 08:09:17.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"2930cbf7747ea4d3627f1bb2c2b014dfd0707e51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-20 18:29:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50fdd1d5b9bea7629d84117ebbf9ac23c5f9d598","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-20 19:03:07.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/74/901474/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1c98c116b118599836f15512822d150ebd91fbc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:29:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82e646e083a44fb8bbcd7e36ff120479\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ac7aa0b4d484462b1aa54f1b6303a47 : SUCCESS in 56m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf56e5e8725048778c06bc2cbaec609f : SUCCESS in 55m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d4d1f60ec3940c68fae185b659841e7 : SUCCESS in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98f6327ad5eaadf8577b10c083c62ebb0fbe9b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 21:09:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1723e66649af44a0837cdec713215662\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eb3c95e96492445c8b332c257b16d6f1 : SUCCESS in 1h 56m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9344cb2f61554792b9f732651506bb57 : SUCCESS in 1h 54m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6b1a672d500b4fdeba5af46298297ef7 : SUCCESS in 1h 31m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b667b65cfe644e9e8e70e3cc8f5f0b5e : SUCCESS in 2h 17m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/079d09b8e4f84419a4e118fd577f1f9b : SUCCESS in 39m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ccd3f358cdd4b78821b492afc7770e7 : SUCCESS in 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6209f213ccf24af29a52b634ac573e71 : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bfa744d65a84a3e83a878d745b70c19 : SUCCESS in 27m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bfe4ccbd7b34e8e9de1b910a39ef588 : SUCCESS in 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8943819df8e94a6ea9c1b33a295a3c82 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d65385f62f524114b70b7a3677189988 : SUCCESS in 19m 27s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7b152078a6b24397b1480bfe41e2f011 : FAILURE in 28m 50s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4afdb552afa94e3690803f8efd967f98 : SUCCESS in 1h 03m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0bca5b85375e4435a2e61509ae917844 : SUCCESS in 1h 13m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/279b2f6fc49f427d9ac03540ff60b068 : SUCCESS in 56m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6685f9bff2774030bdd78d4d4e157986 : SUCCESS in 2h 36m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/84b1e197f6ad4eb082958115933178b1 : SUCCESS in 54m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae1676be9ba74fdb8c82c2128eee4802 : SUCCESS in 1h 05m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a839a6fdf8e747df95df55e2137b636a : SUCCESS in 1h 36m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3a8294e914764a5198b7be88103fc57f : SUCCESS in 1h 18m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e725e5239c444e14854fc774d840ac4b : SUCCESS in 57m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/93cb508b0cb74d57a7ca20b6669e4a42 : SUCCESS in 1h 25m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8aff821855f14375b6433be5d313f196 : SUCCESS in 1h 54m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/748f08bb39764bc584c2f33976d13883 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d72e04b5025585b3f1ca98ed5c693c0d31fe87fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-20 21:28:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1208f865ca189a439729d03a8c7738ddfce4fd52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 21:48:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/044b15a88a374fa9b5d45ad5bc06aaec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7882d24e6d7b4f389d7beb0972c4021f : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d314ded7d1a74771b7d604a737b51a81 : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1582ee17249445c9bebb8e62de39a5b : SUCCESS in 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc511a15d1b8e789b4c3353b78b10e3eca12bbf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 00:29:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1eca03ca8894adc9cbcdd7e515d461c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f4712c16811f4e4abe3e84c656a9894f : SUCCESS in 1h 19m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6132bfecf774efa9631b616556990fc : SUCCESS in 1h 51m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/95a91bb271114b2289abc3f2e2f938d9 : SUCCESS in 1h 43m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6a64c31d7e1d43d593c4a3fe51a7b0c5 : SUCCESS in 2h 19m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a00e2f4910ba4a76a1b14a8ed35a36cf : SUCCESS in 43m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08c5acf9f5f74e33922a02d3afefd5eb : SUCCESS in 52m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31024f67e8214ca8a8242003bc1974ea : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e0771e3d2b4d92a96d420bfde1d034 : SUCCESS in 29m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/145a2aeecf6442a2b8625de9f1d56f19 : SUCCESS in 24m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d4498d56e3f49cca984e305605a01b8 : SUCCESS in 11m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dea6966fa4bb44a495eaa56e5e4642c6 : SUCCESS in 11m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3036cb37599b49b3af16bba1bb04e9a1 : SUCCESS in 29m 31s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/643b506dc2dd449a9cbda2f6d5c2f8c4 : SUCCESS in 1h 04m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/57e061a8090e434da7fb8298fb59867d : SUCCESS in 1h 08m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4cd8729ddcea42edaa87b6cf67836e89 : SUCCESS in 1h 14m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b16eca84abb41b09c7934fb76d5ab8c : SUCCESS in 3h 00m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/279e3f0a955644cd96dc0216a29953c4 : SUCCESS in 1h 00m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fc3a8d46ccd34a11acc959d6621b57b6 : SUCCESS in 1h 09m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/00b035f08ad74f19933e540c98bcaeda : SUCCESS in 1h 27m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/546d2ceba4914afab3ef04ef46033b6e : SUCCESS in 1h 13m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/68d677b8a8d4456090ed3580ed4435ba : SUCCESS in 37m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57fa6d182e6b42859ff2f9f847a6860b : SUCCESS in 1h 32m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3838f22f21eb40d98dbc5a1b82489336 : SUCCESS in 2h 05m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d85c7c0c9a4b4db492274efc76c8da28 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":1},{"id":"72a2c01d511632bf43cb0075b17f3b83e621bf3d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-21 02:45:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"81230261c3703497a37d2e6037ba533bad6bab33","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:31:44.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"924e7c701c769ac7213266503c987e34b8496b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:32:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ffbda262685c89b595ceccebaed163c5e9958b41","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-21 22:05:33.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/74/901474/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"19da1d5bfe1f5ceb057df2c180d09d76e8ec9a26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 22:21:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7f769ca19844ef985cc51db63bb7a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f1df1cf901b46b4bc004cc474dcdfae : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/faaec16e30d1481cb6dc88dcb543959c : SUCCESS in 47m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af291f418275462ea55800a8ca6d88e9 : SUCCESS in 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57b0fa65b9f42dae09ab5f863803954772e263b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 00:28:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b1780df77344da8c05ae77efff41fe\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/efc23e5686e84f47810f23d4706ec403 : SUCCESS in 1h 36m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0219bad7d71a49a79de431b7cc43b643 : SUCCESS in 1h 51m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/db9c735276fa4db0b8e9834bcb0d7067 : SUCCESS in 1h 24m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aca89e7a2c124e42b102cf886bd93828 : SUCCESS in 1h 43m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/766291f2845345a29508cc9456d23d9b : SUCCESS in 35m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a1a0e930394153aa047cf81a2d567b : SUCCESS in 38m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbcab60de67437bb271480303d0285f : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfda8955eb38444982326cba2dd56a31 : SUCCESS in 24m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab1132c5453f43649c94e66e89d1867f : SUCCESS in 22m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278124238a184a629813c5c72f518e70 : SUCCESS in 10m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8de255b56741c889254710a94b4f48 : SUCCESS in 17m 39s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/21d3803565b34b00a5b7ad34ab0db788 : SUCCESS in 30m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ed63fa17662d4f44b71fcedcfb350a5f : SUCCESS in 59m 44s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2f0d90609e194060b142aae30b1bb428 : SUCCESS in 1h 06m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed307f6f3c944cd383263ccca46dc4ac : SUCCESS in 57m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a48be029da4484fa2fa2eb3f85e901b : SUCCESS in 2h 54m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1470b59cb72b4402988598901219bc23 : SUCCESS in 59m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/70a438699bc14bbda9701ad701da7c0f : SUCCESS in 1h 03m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f81790dab6ec414fa7dc6289b02da8c4 : SUCCESS in 1h 25m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2fcbeb0533484ee6b8c11728c293981b : SUCCESS in 1h 13m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d593b3eaaf354e92b647f42ebcb162f9 : SUCCESS in 59m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2a7588f1db6a4349b14c669bb66d61eb : SUCCESS in 1h 32m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/437f724cdb8a4274b2b9e33dfab61c89 : SUCCESS in 2h 12m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11e989d79ab44d0c9bebe97da9d1e826 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6f83ac9ca7caccbec586ac11a60dca4a26da5af4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-22 05:22:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb446d95eda1ec17ea3a979e894c0de72bd640a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-22 05:26:10.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fad554fcbaf10b5dc27e149eb0357295224a5a51","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-22 05:26:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"853739af267e919ee9d3b3bfecd772f0e22f4cc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 06:10:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/000f2d370c5742548f1acdd27b00ed05\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7548ac3488ab4803b1c37d0ecbd22f4a : SUCCESS in 16m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b9bc6b7f2454b62860fae4e4ea143cc : SUCCESS in 42m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb3cd7fa8d0143a8aa54c5f998eaf7f9 : SUCCESS in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55ce018a4b219d4379fab577d00551e868bc80af","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-22 08:13:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8610c13a591075b48cde801c2b024aeec594c8c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 08:29:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1adc1a12d434ea09e316a37a0b6b4b1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b08a53e528f8403497a4c7bf99afb482 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d4a4d453e514dc3b175b8bcd1007530 : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fb288858b1e14d4893bc3710abca38d8 : FAILURE in 1h 34m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9a0e20f6591d4e6983ccbc1a507351d3 : SUCCESS in 1h 33m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4caaaf1364d64f81bdddc05499213cec : SUCCESS in 33m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ab5cb4c8e1445e6b83007e8fa233ca0 : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2be667dc1c24c89bb431ec09f1d1f38 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dbe51376b48453ca1fce00131518777 : SUCCESS in 19m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa8efcd27d9b494b98f0280059e57e41 : SUCCESS in 18m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662c847256a94c5cac973576922332bd : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e994b013a1743ab8dbb56d2ea915017 : SUCCESS in 15m 17s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/41607c2996ac4d249166892c7f846c57 : SUCCESS in 20m 10s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8b09529f5ec45e0bad054d0b315f5cc : SUCCESS in 59m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b004dd125e53443c9e554303a8b9a64b : SUCCESS in 45m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d0e0f68009ce4422af4dfbc64a59c70d : SUCCESS in 55m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a758cecc8ea4d278a4f6c20a4b1825c : FAILURE in 3h 01m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/544dddc08dcb44a1953f683a0b891ed0 : SUCCESS in 39m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c42c44ff5b7943fc8a909f9f01d7cd87 : SUCCESS in 1h 04m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf843a515389436faa46f97511bb05c1 : SUCCESS in 1h 29m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0353df1cad79425f86aae933f8a12abc : SUCCESS in 1h 41m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7be65fce4f3c4b8796cedebd80c694ef : SUCCESS in 48m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5786594180c0415aa36682d6b3e9481e : SUCCESS in 1h 27m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1b53ae88d9224451b02fefb554fb06c2 : SUCCESS in 1h 32m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6090f4d385c5414a9a1c69badcf115c0 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":3},{"id":"c2da003637b05420da1fbdf31b667fe48b6a214f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-22 20:02:08.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/74/901474/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"3a85d8a3543d4e4c64e672ab1704c06087caeb79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-23 07:06:53.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d9daf034f6727babfd93aa7dcfb7164c5715db71","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-23 07:07:17.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b57a34a2ff28b110fc103a704c7fde4e1f92a8c9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-23 07:38:09.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/74/901474/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f98208cbb58e48f925fb2393c99c7815da0aa81e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 07:41:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bab9b892edf47fab342b45d5c370a6a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fcf391e68734ec9be389fd867054a32 : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/011a5628ec914344862469c254609f5e : SUCCESS in 20m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6088f9f98a95411fa5710e0708b4fb34 : SUCCESS in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7bc7ea9cf2a38b636cfd6ce94a410cd83f0d9ab5","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-23 09:18:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8e01052ed35065e92111fa11c617da109b2762a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 09:21:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6502702a2f094b83bd371ef0f985ffe7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e5a7b0563988429b995a8316e643c53f : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74235ae6179e41dda8845d96dc4f9766 : SUCCESS in 1h 39m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/569e72f44e8f48fe88c3a3b0dbebfaad : SUCCESS in 1h 18m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9c902fd7ed74c23b7dca1cca1d23ba3 : SUCCESS in 1h 24m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28be4b9b6ea74d23bfe370b01c1aa4ee : SUCCESS in 36m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47d91e803c524c708aeab54b0ada2b63 : SUCCESS in 39m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/113a7c6cf2c2498cabfce4e87d75c106 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df8721ac341a4644ada5f44719aa6e2a : SUCCESS in 26m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76e8e7afdcb7404088941cb9a9311152 : SUCCESS in 25m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19d57f5af554560a0744cdfff6e6dff : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d89a128ab34410923c1daa631b6e64 : SUCCESS in 24m 03s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e07eedb1e75b45ba9ea950e537945320 : SUCCESS in 28m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f55276f967274769803c45f334f961b7 : SUCCESS in 59m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c91acb2cc1b24848ad0c2a2efab36d55 : SUCCESS in 1h 07m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45737c9a63524c0a9fc0cb430a3bf228 : SUCCESS in 51m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4fc45444a8d4a59b1e4cd6c0b8aeb45 : SUCCESS in 2h 06m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e723c0c4eca487f9fde2fccadc263ce : SUCCESS in 1h 11m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ab0e8d182224559930c8e270b096169 : SUCCESS in 50m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecd21cab8f7f445d98ccc6097e9fae85 : SUCCESS in 1h 25m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f87f0a4bbd347da84022a16e7ee1b33 : SUCCESS in 1h 21m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f6b4e1e79c641278150ae0f13888fe5 : SUCCESS in 46m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa82845a83374024a5d4b9a4b899b8fc : SUCCESS in 1h 50m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5f7792f7901497f9b8d6ce2954fd082 : SUCCESS in 1h 59m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e6e148876585465ab9334941e1820122 : SUCCESS in 1h 28m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3120f03600041c83a357e6a9e75bc2582c3bb65b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-27 07:57:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4fb061f584c17d8a8398069691b9aff743163cac","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-11-27 08:09:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"92abf9392c4378e491d322a2682fe743b36bd571","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-28 06:47:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"485881abe2857d664e38e43e9374294bea8a8460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 06:48:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c0ffdf406838b026805d048eae6ac55bb0e89dbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:29:02.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e54461cd46d46de8786833900bbb761\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5837b85e8f2146ceb8b5406e984039e7 : SUCCESS in 1h 55m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95cfb407a73041dba04cfb29f7e0ef5d : SUCCESS in 2h 26m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a148d969b4a74775b79be0dc16d41632 : SUCCESS in 2h 34m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/586235c4d4354698b8d3a49b57dc6480 : SUCCESS in 2h 00m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670f55ede07b45049f232ee8abe69f5a : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1210d56477a4d438ce69b79b251e148 : SUCCESS in 29m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cde338f663c34c3cb9a6661d26b46350 : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a61ea04232749f79c3b2bb80ec636f4 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d32e4eedc4d40fdb1127bc240c92759 : SUCCESS in 18m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab98547eac174203916cc3e2f0e8fe86 : SUCCESS in 57m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/65f60aa743c94c9aa530d2a75422a129 : TIMED_OUT in 3h 02m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c792360ed5ca493085e530a863b5c0e4 : SUCCESS in 1h 15m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a606d6e4b9df4184b955ced7f8b75d1b : SUCCESS in 53m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/772e0d4bc2fe4c928d4bb1edcac74dc8 : SUCCESS in 1h 00m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e957d4a746b7443997778ac078b3e1cc : SUCCESS in 1h 04m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/017b95cebb644199bcec1af71e406f63 : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":4},{"id":"e7523f0fab0f964885a29f43af66903c45e95cf2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 17:45:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8159620e2f33a9dbdc596f21336c6cbb8edea78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 19:35:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e961ccc172fd4e2fa017e3479070c8c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc5cb2f2937c45bb8be89e4e4084722d : SUCCESS in 50m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/edcb64f4ebb54cc5bf0d2b665cdfba86 : SUCCESS in 44m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f90a27d3bf64622849e7021f186d27c : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a271e7d13578dd938b9eaa91b3c9e18ac8e5129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 20:49:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1327e614d66748d9902f690352653191\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/677ef033eebf461b905472e4f7b58d1b : FAILURE in 15m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8bff9f3ee93846509fc71a1b3ad636e8 : SUCCESS in 1h 25m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4d85aec74c4f4f2e986103c855dd2a81 : SUCCESS in 1h 41m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/99593f7a65c94938b662857c40c38d95 : SUCCESS in 1h 27m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a87bdb5905594c7ea86f491332395752 : SUCCESS in 32m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c4fcad153c84d21b707e6e914285881 : SUCCESS in 42m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6789f383ed3f4b61a1ce39426f443bf8 : SUCCESS in 10m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f39858dea1e440acac1d34ecbee46e66 : SUCCESS in 25m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5a4030339f74202b1a0ee1e10fd3d9c : SUCCESS in 27m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0cca20bda84f1b86e91a73ff591913 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e67e1cb454f74186b56336fc38a86b70 : SUCCESS in 20m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dda4e4aacc034fdd9282ba93ac65b463 : SUCCESS in 31m 16s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/04c006c6e0704038ab527387cc6f7d43 : SUCCESS in 1h 05m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fc9e179e8441406fb6849d11a01befc5 : SUCCESS in 1h 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e41ab0953fd42dda7abe3505cc778ce : SUCCESS in 1h 01m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/170e82f66cb942e5b57878d4cf01362b : TIMED_OUT in 3h 02m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4b135b13daf44048d42379c656c8885 : SUCCESS in 1h 09m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/813bdc12e1c8434ca82705295be39c88 : SUCCESS in 1h 04m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6ab3ab5cfdd48a0ac2bda8e62902a60 : SUCCESS in 1h 20m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bee36d9e59b543c78fd7344c07ae5035 : SUCCESS in 1h 11m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/307a5961edd84722b79e3a659fffa275 : SUCCESS in 54m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7b381dbf15f94f8f96405a4d5c979b9a : SUCCESS in 1h 39m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1f01a1120c2a41088dd6b88733041f47 : SUCCESS in 1h 44m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a6ab043c16b643358a1d8adb97ea4d95 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":4},{"id":"48b9be2f2a1445f9e5c19ee7671aef834611d976","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-01 07:43:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bab8d8e7193832ba6c91e129942632e51ffef185","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 08:06:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b2c3be4714b4fb196581776c7144789\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64151dc266174204a862f28021df6f49 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc4e7b9b19e8403d9d96e9a2d84c7736 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8537d9a84b9841fb9a651f8eb4ab9a35 : SUCCESS in 19m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2666beee59f1e50d6af9d04395a9b428c4381213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 09:57:07.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08cbb5b186e341399c1fdff247ca4e1d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4efdf5cd085f4c0ba754b8b59ff73ce4 : FAILURE in 17m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3dd44dfa22d7467893e0afacccd686f2 : SUCCESS in 2h 01m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/52979bac8f454d829dd91c7d1d0a9552 : SUCCESS in 1h 50m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/717687e0544a433fb86947d1e1622132 : SUCCESS in 2h 11m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/590a683b938d4220a5c578fa71d07d9b : SUCCESS in 41m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/801b91f6567d42f6beb5351a443f8d48 : SUCCESS in 28m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab330d6c3744a17bec44e8d8c154a10 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d23fde5a0b5848fc9f004149b6ca315a : SUCCESS in 21m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b07d74583d514b7f9cc9955520ad95bb : SUCCESS in 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11a10c31480246779cdbb441010e9022 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/888220bcc8254843aa76c4de3c1b4d37 : SUCCESS in 11m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3fe62ca677724c5b963a5202d215958e : SUCCESS in 28m 05s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8c483d4b382c469aa5909fd0fc03e597 : SUCCESS in 56m 18s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/90db25b629154e37a9d5bf140a85c5e1 : SUCCESS in 1h 14m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d11465ed65f48aa9edcbe16e076aeb4 : FAILURE in 45m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c31f753c3514d84b581bafea0ca2f24 : SUCCESS in 1h 49m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3277885f1a61494e9cca3ef4cfbe2772 : SUCCESS in 49m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9f19c5b143441478d8f46919381c158 : SUCCESS in 1h 03m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/03ca9369bdb346008552b16ee411e175 : FAILURE in 1h 11m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2ce4d6ede544456eab0ea7418cd6cabd : SUCCESS in 1h 23m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/64090711350c4dd4a32a545dd39437d5 : SUCCESS in 1h 09m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9b35ac1c63b04cde8e526416dddc6402 : SUCCESS in 1h 21m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/113c270c6acb49bb85c511fe65eaab85 : SUCCESS in 2h 02m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f5c80323d06f45c493b9f6a050e36304 : SUCCESS in 1h 09m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c3c8e5e3339970117b41b4516dc71cfdc2ca3f7f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 13:29:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f0808c530b0b97bef19c70f5fce7809bc3ed69c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:19:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d14460e9b3476e8aeb46219e355cd8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72bdb6578c4c4df49a880afdcc82a480 : SUCCESS in 19m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8c09d118c7c44f088191b28f731a6d5 : SUCCESS in 46m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df1bb5d9a6574984a5b679aa48d41f91 : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4e019cdbafd1762d2586ba65ca1601f078b621c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 15:39:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b8828a4c6e24ed0916e14eadbe8e3cb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2528964b5264489ab044d3ce3630b05b : SUCCESS in 1h 33m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/294d0af52c604526a876b35477cff0f5 : SUCCESS in 1h 26m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7e08c27a57bb45be934f281e476c7f51 : SUCCESS in 1h 34m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c69f62e74fb54c81aea101136e9cecdc : SUCCESS in 1h 35m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1902ddae69c24ee0a278977b9da55e3c : SUCCESS in 34m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9de0fb70c39b404fbf93b859b78832f0 : SUCCESS in 50m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4acc44a78ff34a6f96973cb346fa44ab : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d1581e9ea7746fdb0b8bb537370c5b6 : SUCCESS in 26m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74d08b64f8ac4817b1993a7f8e8e1e97 : SUCCESS in 23m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d436c0998e304d289be62eff3253ae55 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/551c7bd5fff14005b7df330d45ad3321 : SUCCESS in 18m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/59ae28662cfd43edac1344067eb9358a : SUCCESS in 27m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8a17484408b24751b2c92afb9ac7578c : SUCCESS in 58m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c8d064c98c4a43af9cfb0eba4d00b2ae : SUCCESS in 56m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6ffe22344114f19bdbacda0c684910b : SUCCESS in 50m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/748925369722418ea5eb379a9d000022 : SUCCESS in 1h 38m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdef7c342fbc47fb825ce5e6ed47b836 : SUCCESS in 51m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c3c9430ad02495bac1a3040fa809708 : SUCCESS in 1h 01m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e7aded22ee6f43859dc0e288e49175ae : SUCCESS in 1h 23m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/087b15c5767947969f2f6ab158a90285 : SUCCESS in 1h 09m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/773d2eeac6de4101b1493822f2359fd1 : SUCCESS in 39m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/504120ea4c494a15a79016ab1381a90d : SUCCESS in 1h 57m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a38de2be2f724d66be19e1e98e5da403 : SUCCESS in 2h 03m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8e2a8af73e274c1c96b5b90bed79c7d5 : SUCCESS in 1h 32m 59s","accounts_in_message":[],"_revision_number":4},{"id":"60f3f78ba0328c6decfcf5c751838a78aefc53bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 15:40:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fc5915caa275c30d604e986afd99a8f471b478d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:20:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93eeb57ea56348d5842a4c636720b4ff\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/26f37e6029234606aa1acb23a924d95a : SUCCESS in 1h 43m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0246e783f88d4f45a80108f9e1fd992e : SUCCESS in 1h 37m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/15e71b07a8814fa4864f08b8873ece2e : SUCCESS in 1h 58m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d90d09cd5a02413ba963b2cd337aba99 : SUCCESS in 1h 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b696f33cba6947ba8c7b912570602fa9 : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c143c6113cd0409988d001be27746e06 : SUCCESS in 26m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdf46903722348349ef35555d2b4b900 : SUCCESS in 22m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b8490aefb74e2d8f89ab0558af611d : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81ae31b96c38458985085527f997e5a5 : SUCCESS in 18m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/99e79d4ebae0416398e574f869026ffe : SUCCESS in 52m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8f2a45e72744948993e5835c6ae62c9 : SUCCESS in 1h 59m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5006b95ed369458988d1b851609955d8 : SUCCESS in 1h 07m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e961dfcb02404280ab91a562634390cb : SUCCESS in 1h 00m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e9ab9e329684df6a1028f1c17256ae9 : SUCCESS in 43m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2719964f58ea43d6b7e1dc055c4a0c55 : SUCCESS in 1h 09m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e1a73047fc9b41d6a96ec77a5b7cae8a : SUCCESS in 1h 07m 59s","accounts_in_message":[],"_revision_number":4},{"id":"6effe46d3d02e071160371c062358c9852c59f23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:20:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"805837afcb7597e340891ca819e9077972bc68fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:21:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32bd86137a704174a4d5f345027c713a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ce133ad8ab4fba9213f8d9db250fa7 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/431a5c9449c44ac4808fa1d543de544a : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1c13fb03d1d9040014f6927be43a9ddcabcb7f5a","revisions":{"b6a3038316b71ae915e410052ce059fcaa0893a8":{"kind":"REWORK","_number":1,"created":"2023-11-20 18:29:55.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/74/901474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/901474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/901474/1"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:28:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:28:03.000000000","tz":-300},"subject":"Remove ovs_integration_bridge configuration option","message":"Remove ovs_integration_bridge configuration option\n\nThis option has been marked deprecated since Wallaby\nas it is a duplicate of OVS:integration_bridge, let\u0027s\nremove it.\n\nTrivialFix\n\nChange-Id: I81bc5f3d98f752d926a243cbd17b8b894f2bdf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6a3038316b71ae915e410052ce059fcaa0893a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6a3038316b71ae915e410052ce059fcaa0893a8"}]},"branch":"refs/heads/master"},"4f7d3ef99c0f2c4349b700bff0f098365b3ceeb9":{"kind":"REWORK","_number":2,"created":"2023-11-21 21:32:03.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/74/901474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/901474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/901474/2"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:28:03.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-21 21:31:53.000000000","tz":-300},"subject":"Remove ovs_integration_bridge configuration option","message":"Remove ovs_integration_bridge configuration option\n\nThis option has been marked deprecated since Wallaby\nas it is a duplicate of OVS:integration_bridge, let\u0027s\nremove it.\n\nTrivialFix\n\nChange-Id: I81bc5f3d98f752d926a243cbd17b8b894f2bdf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f7d3ef99c0f2c4349b700bff0f098365b3ceeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f7d3ef99c0f2c4349b700bff0f098365b3ceeb9"}]},"branch":"refs/heads/master"},"e0dfa2ea41a335f42e287b5f159847de22e2aa8d":{"kind":"REWORK","_number":3,"created":"2023-11-22 05:26:10.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/74/901474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/901474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/901474/3"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:28:03.000000000","tz":-300},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-11-22 05:26:10.000000000","tz":0},"subject":"Remove ovs_integration_bridge configuration option","message":"Remove ovs_integration_bridge configuration option\n\nThis option has been marked deprecated since Wallaby\nas it is a duplicate of OVS:integration_bridge, let\u0027s\nremove it.\n\nTrivialFix\n\nChange-Id: I81bc5f3d98f752d926a243cbd17b8b894f2bdf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0dfa2ea41a335f42e287b5f159847de22e2aa8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e0dfa2ea41a335f42e287b5f159847de22e2aa8d"}]},"branch":"refs/heads/master"},"1c13fb03d1d9040014f6927be43a9ddcabcb7f5a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-23 07:06:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/901474/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/901474/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/901474/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/901474/4"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:28:03.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-23 07:06:53.000000000","tz":0},"subject":"Remove ovs_integration_bridge configuration option","message":"Remove ovs_integration_bridge configuration option\n\nThis option has been marked deprecated since Ussuri\nas it is a duplicate of OVS:integration_bridge, let\u0027s\nremove it.\n\nTrivialFix\n\nChange-Id: I81bc5f3d98f752d926a243cbd17b8b894f2bdf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c13fb03d1d9040014f6927be43a9ddcabcb7f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c13fb03d1d9040014f6927be43a9ddcabcb7f5a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
