)]}'
{"id":"openstack%2Fneutron~901478","triplet_id":"openstack%2Fneutron~master~Ifbe9f4cd0c291ab61e0adb93cfde866001086345","project":"openstack/neutron","branch":"master","topic":"remove-vnic_type_blacklist","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-12-04 04:51:43.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-13 02:49:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ifbe9f4cd0c291ab61e0adb93cfde866001086345","subject":"Remove vnic_type_blacklist deprecated name","status":"MERGED","created":"2023-11-20 18:40:50.000000000","updated":"2023-12-13 07:49:16.000000000","submitted":"2023-12-13 07:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901478-remove-vnic_type_blacklist","meta_rev_id":"54727cf60f03e847d9701cbecd42cd37b96aa3ed","_number":901478,"virtual_id_number":901478,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-11-28 12:12:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 07:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-12-04 04:51:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-30 23:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-12-12 15:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2023-11-28 10:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-11-29 06:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-28 10:18:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-30 23:34:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 19:37:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 20:07:33.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 21:28:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 07:53:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-27 08:11:14.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-11-28 06:45:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-11-28 12:12:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-29 06:08:01.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-11-30 23:34:44.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-12-12 15:30:50.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"6d3364e1badf8d81ad042f7c0c970f940fc3e62d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-20 18:40:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae4072e9c427ab3891960d86db2e47baae131876","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b60273f8e1e14d2090205409a99effc8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f39bd00b0f164e44962e7093266a041d : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2a7539eecfc486581d4ec6dda736630 : SUCCESS in 53m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12a61813932844109bc684037b757265 : SUCCESS in 18m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73801fe6b844f125572daafa959d8e151472d7c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-20 20:07:33.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/78/901478/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9bc464d9668beb074e85ca145716a768e27d1dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 21:28:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/269a9778be3c4b07b0f748e01c711d9d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e057812e0b84192b459ffd983568a59 : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/62fa8b5a828a4d6cb0b537e297d972de : SUCCESS in 1h 59m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/38bfc317a5694ad59647e6f3d6061c32 : SUCCESS in 2h 08m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/04d829433a044f8e91377cd620a86a14 : SUCCESS in 1h 58m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44bb49d2b7bc45609d32afaa3ae43a3b : SUCCESS in 47m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a77cb87f089746ea9cbaf46aeb481342 : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a868dd88a71478ebbfc649126ead43b : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e50a432377cf4f689b5399d5c459d61e : SUCCESS in 27m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/322cf430db344e3ab5a4b3ecabe68c8e : SUCCESS in 22m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f88336c7cb954083b3c28a4771e8ed9a : SUCCESS in 12m 10s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/88429eaac2f24b5ca06cf4631f7a9c61 : FAILURE in 26m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1f1e69944d904e25b557852448c31785 : SUCCESS in 1h 07m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3abd7d1239f1414393b02841c5cfdc8b : SUCCESS in 1h 04m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0ec12663fa5409884cc5c24c0dc784c : SUCCESS in 1h 00m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/984714701e7648509a5e6aa112e46787 : SUCCESS in 2h 45m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/172a4b5b2e2e42419ec6c8c0e0ef61af : SUCCESS in 1h 00m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2a5dced39f4647b9a4481e7b84b96e59 : SUCCESS in 1h 05m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b7af9bdff66409f8902ccf887bbd482 : SUCCESS in 1h 28m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2cd73e2b077d4d9893bddcfc22964123 : SUCCESS in 1h 23m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2074ce102c404f57805c6df6e89469d5 : SUCCESS in 58m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/65382db117794c87bd899c132d596258 : SUCCESS in 1h 27m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/28eeb01ac7494e9d9128b696dd06e4fb : SUCCESS in 2h 14m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/80640de926d2417398d957919a5e795f : SUCCESS in 1h 18m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f8765b6424fa230787c5b07fc9266909dea7697b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-20 21:29:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5c9959b92696d2de2ecee1bc00bd4e40415f721","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 22:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75235192356493f84e47a6674e21471\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4048b37e65a400189aca37f0aadf861 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11d9eca2f3254085bb4e76d2c9eac3b6 : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ced946f1f7041db82721dcdacc20e6d : SUCCESS in 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e2f96ffaf01541e8738053bf34f570b490f9f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 00:31:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bb1ae559974f0ea227dd8bc9a9ec05\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9fd70f52656c4c02bcd34e68b2d5d16f : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a422eb6f2ac46739d05c9a81db86906 : SUCCESS in 1h 49m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/afbae655bf7b4171825c8caed94d3e63 : SUCCESS in 1h 19m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee15dec670cf4341a3df13b14e38c1de : SUCCESS in 1h 24m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dbd88b15f9904d19bc87793485f97824 : SUCCESS in 33m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0eea25f79124bb28e61fbf3620abb78 : SUCCESS in 35m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d8702b41594446a8cd6fcd779632ac : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c9194dc58cc49e9905fc8a184482df4 : SUCCESS in 23m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4bc55f11bad43059e16b441c7458652 : SUCCESS in 23m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7558e8e997994ffe8574ce045324b148 : SUCCESS in 11m 14s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/01f34b1ba7a841b9b00346906a7c44bb : SUCCESS in 23m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/04ad3b214d0b4e3686ada9c7690cdd5d : SUCCESS in 1h 03m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b87f307cc99248ad8cd8c06509c5d526 : SUCCESS in 55m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dec4a34101244524811be24cde222329 : SUCCESS in 1h 13m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63e5eb6a45d9474c973d1ee11fa01e02 : SUCCESS in 3h 00m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e8bb0eb3e7fd4ae7bccaedb0921e673b : SUCCESS in 36m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a8211d24aa8046a5b602739e12772baa : SUCCESS in 1h 03m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2aad7f3908874a758e11468eab9cddbe : SUCCESS in 1h 30m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/07d54569f4414128a1f5934a9ee230d7 : SUCCESS in 1h 47m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d88fd65ee8bf4e97ac01bfb417ec102c : SUCCESS in 58m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5aab116c0029490183bf15639800cd71 : SUCCESS in 1h 35m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92f61738a35842b98c5a5a7cc171289c : SUCCESS in 1h 57m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/277adddd273a40f79c775ffd72421943 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2cbb9d453246e3f7f761082635cae55edf01a886","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-27 07:53:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01ccbbe3b7dca65d0b3624ee9aee2e709619eb24","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-11-27 08:11:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"59f5839800d58d00a4f3a944dc53d859eae3d75a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-28 06:45:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"726ed3d9b386408167bdc34f08042fd45e39a269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 10:18:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_1131\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"_revision_number":2},{"id":"04d9a734a899638aa0f51586449f8c9040a1c92c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:03:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2738014dc39b45ed9e29e94430d6b57a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4f8faa2f3cf46db94a8e66a50feb4dd : SUCCESS in 20m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c83113356ebe482b83e34ecf44a86cea : SUCCESS in 43m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6b726bf787a4e0e9c262dfb655f3965 : SUCCESS in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d219a88599e2e36eaef4cdf2d886f4acde045f7a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-28 12:12:58.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/78/901478/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5afb741263d7d288acb8ff71784d91a1ba485b03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 12:49:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/272d6ea15466465ebd68654eb6df5328\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52b05254b6024ac59dd2a0c1822e055c : SUCCESS in 1h 21m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f5eed09aed0c4757b8f0d22da36fa7bd : SUCCESS in 2h 05m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9d9b6d5a7f8d4796a0c4de836cac750c : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0dd748fa707845d3bf6c2544634625bc : SUCCESS in 1h 38m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d90057882e54403886aeaa3787a88bc : SUCCESS in 37m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c735a4f04464a18a99242e1b9e54889 : SUCCESS in 35m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b70e2c273b495c8b70b900ee3ffc6f : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd6654872c34b78994a7d42faa9f798 : SUCCESS in 27m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/341e414309d44b0eb4669eab9f6af6d9 : SUCCESS in 23m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa500ea2ae3483486583893f7bd1312 : SUCCESS in 13m 09s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/80958032ab814800be8c9277cd10a7fb : SUCCESS in 22m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12e50f9fa306479caf9c05a177b518f5 : SUCCESS in 1h 00m 33s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d779997259c543719317c86e8764fde2 : SUCCESS in 54m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fbfb687124594effbe667b44b9e730f6 : SUCCESS in 1h 03m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0020466f8784f068765487aa576fd74 : SUCCESS in 1h 56m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f187a3a2bf54eeeb09eb7edaa396947 : SUCCESS in 1h 12m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/960a0ad5997f4e11be27b9c7aa7c1c4c : SUCCESS in 1h 06m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/59e0573d04e447c790d6f11df6abfb29 : SUCCESS in 1h 31m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/19a09123939547498252465117eb2b75 : SUCCESS in 1h 27m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a48286560bc8468a91a9fde0c7debed9 : SUCCESS in 59m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/22b598abf7314439b6f1a6a5be864444 : SUCCESS in 2h 23m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5a7ce9f8a1c4bd1a7ae9b51dba8717a : SUCCESS in 2h 07m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ab5dfeb169dd47b49b396c6ec7807ff7 : SUCCESS in 1h 30m 26s","accounts_in_message":[],"_revision_number":2},{"id":"95ca3c12c4d8d64ba45a1e9a39f5ade444b297d0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:08:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"505aee62acc2e6965822ebc1446e0dd5b99d50b7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-11-30 23:34:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"834e8ab63dd87eee89a688454cc4a1f2f7f7cd58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 01:45:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c144a8407f84970c9a70ea8104fd3732afdc8f2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 03:51:03.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/901476 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"b8487b5306e95b1a0274649a92ea42ffa4c28eb4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 13:28:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77c7b7f36b27f5556a47f0b89a9ee025a7935919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 14:15:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e219c53189b744729d598ae15170c2c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e06d88023a5f428d94ba4e209550f68c : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d946d95b6a064d6c81cda3fd97c56050 : SUCCESS in 44m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a7a1c9e26f5a4a38aa5e02883bfdd1ee : SUCCESS in 17m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9551c2e0b87eae5c77f47593c96da07f59c5f8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 15:31:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/450b863ef759408d856c376e8a479db7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d5d07242a9d44b1f91bbcdd58fa58ddd : SUCCESS in 1h 18m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2b2d7fdda5ff4dd385182ea6cadfd548 : SUCCESS in 1h 28m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2041ddb512814603a8b18b6c52d9fe0c : SUCCESS in 1h 21m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db16adb6538e42a89670e6e3905b8708 : SUCCESS in 1h 39m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09e12ba996084ecf8d62dcd0d489bcc2 : SUCCESS in 38m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fb2a7e4fa9b48cea1399da84090176d : SUCCESS in 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f0b778c99a4e9bbba66cb942f99f22 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d1887b6c0824f958189c0382106e858 : SUCCESS in 25m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/643d12c685a3491285c041f1d94b0f5c : SUCCESS in 25m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e10889b0294beaa20ca49668afad53 : SUCCESS in 10m 24s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9664feb8ba6941f3a5ee2632e2f89976 : SUCCESS in 23m 04s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3fa4a3ed9ba84cc795446bf3e17e4003 : SUCCESS in 57m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6db1b66bedee44edbe6d66d89ac97b42 : SUCCESS in 55m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee198453d00e4c6e8dc5c0e60dc844c5 : SUCCESS in 58m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24eabaa5c1524f5297a5cf19297a01d4 : SUCCESS in 1h 44m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/533b3ced96f944fabc76d7e21b3742b2 : SUCCESS in 47m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/63f791c56ec7465e81277443b3c67778 : SUCCESS in 1h 02m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2dc9a8aea3574f84a9a4ee64658dd877 : SUCCESS in 1h 19m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6131c38341da4767b1ea829c74ee5e92 : SUCCESS in 1h 09m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/991cfeba9fea4542bf977755dc6b2094 : SUCCESS in 1h 08m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ac14e7784be845d298df7f12277f2b67 : SUCCESS in 1h 57m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/090d42b51ac84780bcf19bfb1ac1980a : SUCCESS in 1h 39m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/221ade4d227a4fbcbf4ff4eb66a3a0c9 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":2},{"id":"fd1c2be3c236647547b87d0d6a5e91072722af0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 15:32:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"72525150cbefcf92786a5d3c0becdb847b80884b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 18:45:26.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cd2789d514a4616b4fd212bc91af2b0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/063f0df98320403b952bb7f80669c62d : SUCCESS in 2h 08m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1f64412e517d4279985c1519edd5eb7f : FAILURE in 22m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/245a84e106d84d17a8deef3bfb54bc9b : SUCCESS in 1h 28m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/784202d216e4406fbed909808ecd8dd1 : SUCCESS in 1h 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779a46bae59d4db5a954aa765be2f49d : SUCCESS in 12m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4aedd643264327a5862e6995f03e42 : SUCCESS in 25m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ced28bed29a4e37a5b7828f90dcd50d : SUCCESS in 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4128dc0f94284e1f9aaac82c4a10c34b : SUCCESS in 11m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d7369cb4a2e426ab84d50a7832a1d2c : SUCCESS in 1h 00m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b08392b29174b869b0a0a2d2f6eb308 : SUCCESS in 2h 49m 48s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e661940a1c814a7a8f1ebd762254ad54 : FAILURE in 54m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9897737399d34b119c101a2dde465480 : SUCCESS in 54m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9f24efb6697d499793681ba3d6ebab3b : SUCCESS in 41m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dd9edaefc079400a817c2f8a4c25ac09 : SUCCESS in 49m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c8968aedd404fa1b78117cf47c31d96 : SUCCESS in 1h 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d1c503792975f17b2a6377e183c58f7e269c1049","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 20:21:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"838e7fe5a403ed5572b8d72f9c7773d0371fe04e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 20:45:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6fcf70b7c4459d88d8fe790a8ba236\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2202ed9ad04641bc9959b1474383e41a : SUCCESS in 23m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19ea9e338d924f048b22950912cc4300 : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f470504d1fe644c9a7fed9726282767d : SUCCESS in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"805d5afdd9eee32e5fb12bb41e8cceb36aaf076d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 22:40:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b93afb2e81641758158eb3ad303cc13\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/809bbd7f42cf4b7a92872cb9e61f8ab7 : SUCCESS in 1h 47m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e173a3f4bab744ec877972e0d3adf719 : SUCCESS in 2h 08m 45s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/85178c87b41740e4b551cfd5375083ea : SUCCESS in 1h 42m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9eeb857a56474696bc84baf0950242d1 : SUCCESS in 2h 04m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e97590d98359453c993f0a8a5eeb03af : SUCCESS in 41m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47e5e2aaccf4454da35453a391dd6549 : SUCCESS in 27m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29780ad0f6654e8b9172e7656c2fd34a : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2b56021330742feaf06d86923bc3c9d : SUCCESS in 25m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d8becf89d0442fca770653b7068ec85 : SUCCESS in 22m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bbaeac22c54bbdb34e49464c5eb581 : SUCCESS in 10m 46s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3c78a409eac0454eb12a2ae00a16fa61 : SUCCESS in 25m 06s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6c4c0c27e5024648968f73beca726667 : SUCCESS in 50m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c5830c15fdc74d599168bde01a9519b7 : SUCCESS in 1h 13m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dd12b0b370041b097a3b543808590e4 : SUCCESS in 1h 01m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d658a973eb66476c9ddc9a3b6639c1af : SUCCESS in 1h 49m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f12268c17c549a388dc0e624049da76 : SUCCESS in 41m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8bd95b5d0eee433b8429d610ac8c0078 : SUCCESS in 1h 24m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/218b0112e387423e94c9dbe75798c7b1 : SUCCESS in 1h 07m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c02276e1dcb14e43b8d5dc8dde662ab7 : SUCCESS in 1h 28m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e312a87a28364422876306c0b9543c06 : SUCCESS in 56m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6c562f91cb1244c2bbc67ca424f6f176 : SUCCESS in 1h 15m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c2e8f554fe464424b8e71c2f8fd7caba : SUCCESS in 2h 12m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/61a0dd1d3eff4fb9b92c0783f0cdca10 : SUCCESS in 1h 13m 45s","accounts_in_message":[],"_revision_number":2},{"id":"050a1ea15f9298353389335b7c64a20c45ccad9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 22:40:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"617a2a868d31cb1e34664715022d0540b79971fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 01:18:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19af11757db24e1fa49fa49c9e5a32a8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e983b548f70a4f969c5d082cec6f793e : SUCCESS in 1h 18m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d5e369076e0944acb12d86d6280cc731 : FAILURE in 18m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4f9b672060e34c00bd7152dbb12b3904 : SUCCESS in 1h 31m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/faf807ef2b554d648077bbb04d287816 : SUCCESS in 1h 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1b6c7d57494991942c42074fe405a7 : SUCCESS in 9m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/186c70f8c0a44a21a6c5fae5355ebdf7 : SUCCESS in 25m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b45faedd3d34688b5c78338e28eaff1 : SUCCESS in 20m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5b90db1fec4fc5b0f1c67d842cc6ea : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/557fdee049354288ac32681dc36144e9 : SUCCESS in 58m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6fcb360e4ea4c558f5b0e24869ab6a9 : SUCCESS in 2h 13m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ca860e1c1df4de39f807cb592c5c8ee : SUCCESS in 1h 15m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4118478c7de04a1e8e7bb6a5ab78b412 : SUCCESS in 1h 05m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25e967ac57084b2eb3df413d19ccad69 : SUCCESS in 1h 07m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4b0875884c534cd28c6f7c5767112f84 : SUCCESS in 52m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b0bc7bc7f85d4cc0b433e86c47135419 : SUCCESS in 58m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7a6bf03a06e189dbe5a53ba8a38e3451a9975ccf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-02 16:02:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7754fcfb69e1dc3e05a3e6b31adbaa962205e514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 16:47:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa0cf8cfcdc4c3f8e6ad58e63cf8551\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65311af1b3754857b1b284403339bccf : SUCCESS in 19m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6003cf6637f2425fa183449a34e1e7be : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ddcd40969a843c0bf90b8c9b30ebd80 : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e959a65bba480f075110bdb2bbee3f4f0385c116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 18:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1a351de80164876b2f3ba55ef10dc09\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e2d0e8d1f54f482ca5977474e32f16da : SUCCESS in 1h 20m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/581948d5abb844e09ad6185f646d8b13 : SUCCESS in 1h 31m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9d30494ffb9f41c394081e12246c2fb7 : SUCCESS in 1h 19m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40c82122aa49468cbb181f91627e0eb6 : SUCCESS in 1h 21m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/603db09e10c64dd5a652ba0ff17b7381 : SUCCESS in 34m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7055b0724a2484ea9dcb143184caa06 : SUCCESS in 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b96ce4d06745b0a60e3d04454d9587 : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58be7af8d0f04c23b99c0926fa55e1c9 : SUCCESS in 26m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bad0a17233147b0bb811861cef43776 : SUCCESS in 27m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2261e869e14b5283a8ebdbb60c8df8 : SUCCESS in 12m 15s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/335786ff2547487db5c6366429ee20e8 : SUCCESS in 31m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/377bd0c8c0ce4a47be5a798cc82f1ed2 : SUCCESS in 57m 43s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/95dcec47a65f43acb65f5c4e3d641bc4 : SUCCESS in 1h 06m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/184e030d68564012baa3e6b7ab3b7c23 : SUCCESS in 1h 00m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/85b8430fb6c048e3be4159bfc913bf89 : SUCCESS in 2h 21m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c83a35012f344948912980fc0c41a101 : SUCCESS in 1h 04m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f15c6236dacc4575979b8f91ab306bd6 : SUCCESS in 1h 00m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff2bf31324834e4c8b52999371a714cf : SUCCESS in 1h 20m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ca74912958cb4ca9a508be8401ddc4bc : SUCCESS in 1h 11m 33s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3ec178e096f4ac5b72231c7737c5899 : SUCCESS in 57m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7f8f733f6e94be7a60eace5c3df7204 : SUCCESS in 1h 25m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f43561367edd431fa62c6bcc13461313 : SUCCESS in 1h 33m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a68278cbf2d44e4cb41a9152f589c066 : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"836a531e6115c8155424e396545e1a52eaa47c80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 18:25:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"21f8b70d9364fd744f70259721786c68dd09e87e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 20:38:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96fdace9b0f64ea493d5d6105799f4ec\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5cf53fc2ed924451819116683bf49264 : SUCCESS in 1h 22m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c253847300ce4ee9a5526d1e40da1d9d : SUCCESS in 1h 33m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6c5cccb93174c9cbd5c3966d4c07d3f : FAILURE in 14m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/ebb161dd5aa24f588c61051ad6161318 : FAILURE in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a19aa0069d847ce963dc254da1c2187 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4655ee37f4104a04af640bae0aa668bd : SUCCESS in 26m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/925ecc5dbfd84c8c8b886c6aa27bb10f : SUCCESS in 16m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e798b8435d394819800a5a780e46b54e : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a06a2d6f4ef74e33abc170542a3431e0 : SUCCESS in 45m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/59fe73c1fcc2485cb4eaebf526d4eb77 : SUCCESS in 2h 12m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/391d6a9a9ac345259e367b108e6bd18f : SUCCESS in 1h 30m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5b52739c61f4c329a20d9e58f839317 : SUCCESS in 1h 07m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a563976b13e6440693c13230e313537e : SUCCESS in 1h 01m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6254b42a24324f82a76544115167513f : SUCCESS in 57m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a4f166ec7969470caa4d0fddaf7bb651 : SUCCESS in 1h 16m 07s","accounts_in_message":[],"_revision_number":2},{"id":"54f7fc665931f5240f60185e9961113957a2cc5a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-03 15:46:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76cecc7f8255ee52dba3c4c00cb0e4b4fe9aba07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-03 16:07:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162addea99774aa7b3753cf3104a1930\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7f18daeeaee4d518de85da7a74fde88 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/554b75d22ad742c186b4b25b5e848cc0 : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa060c0ad42e4817906f4f3f3c46e5cd : SUCCESS in 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13adeb1322aca208c502c6863edc1c26ddb4c4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-03 17:57:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f9eca35dea3491b887336bfd0c8f884\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/14756df9e6704bf696405c1a6e8a5860 : SUCCESS in 1h 28m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e2954fcfdfbf422d95d8744ddca9f3a4 : SUCCESS in 1h 25m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/7a772e114617423e960b4051f04b2707 : SUCCESS in 1h 26m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7baedcc3c55947808f4f63d721ca1bb5 : SUCCESS in 1h 22m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2bb34823b6b400ca79eb322c9c82c8f : SUCCESS in 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fd9db25ddb84cd0b1039af6f736b3b3 : SUCCESS in 37m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8033aa7b3b6646a182c1c62487b3d290 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/141d5d16d113469eb2db4b60a01b50a7 : SUCCESS in 18m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18edd5bb763540a594de9e12b5036bf3 : SUCCESS in 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46474f3f65fc4c548a87806723618c05 : SUCCESS in 10m 30s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f0f49fc586d04f59b1d0edc467d6f395 : SUCCESS in 19m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e11f910fa23d4a2b9fd3daa8d0301715 : SUCCESS in 37m 38s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3e110e5801154bafaeb0902fa3fa162c : SUCCESS in 44m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/561625b0f5f7431ea3406c86f2a01159 : SUCCESS in 46m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9c3309bf3964a739423dd9437e64f47 : SUCCESS in 2h 09m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f74e4e6cb04c4876b1d08c2e5b03bfa7 : SUCCESS in 40m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32a52c34e9d74d3f86873a23f6c26fc8 : FAILURE in 1h 02m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c85ddff0ea1b42cea48c00a045078877 : SUCCESS in 1h 31m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5629504d4ce440e5be27b3c20ba68d04 : SUCCESS in 1h 28m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0b7480b109a4a1ebf325545e5a3388f : SUCCESS in 1h 04m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a2efaabe8013417c88148b323b70b6c0 : SUCCESS in 2h 00m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8603c78b96ab499186d36abe726ee2a5 : SUCCESS in 2h 01m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4f0aa06d031c4f528c6aa7239ee4b5f5 : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":2},{"id":"277d25ca15b7cf60373182b9b3a64034ce287899","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 04:51:43.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b622850da33e1273f94c1ddfad926de209cc3527","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 05:29:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5906ec7e8ae7413ba81cd6d1d37c4a91\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61243fb4b3b34c7cb7d3aa3692339eb0 : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48316f59b8534569924f6f6f1109a4ae : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad3cf85f1b824779a2a2150fc1e72273 : SUCCESS in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e6bf3ac24902d36f166c6866b99bfaf0132553a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:10:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba49df9429384f74bd682c231e950d0b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6f0c8634953f47d19311b2e0779a10d4 : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a4e9e7b0c25e45a081a502b2fde001a5 : SUCCESS in 1h 27m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d3f7777a66df4f568908817dd16a8699 : SUCCESS in 1h 14m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f5640c3da164721bcdd94bfd4dd11f4 : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/80176da86c4342bdb715431b9f9d0be1 : SUCCESS in 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ef3a910b7d34ddabccad95f28cae153 : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a32ac030e6c14a22a6ff2f87f6d1e978 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c634ca7586c0484db86bfc5ae6211309 : SUCCESS in 25m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62cae7e3c48b4a38b4d4ff0c79608a3f : SUCCESS in 23m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f1a0b790b54dd98e22fee2c894ef59 : SUCCESS in 11m 40s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8690417134054d3eaff3e34011a5cf06 : SUCCESS in 30m 44s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e696d0d1ca2642c6bc84c934f93df760 : SUCCESS in 55m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0832b24f34904e92b83878d2d32c1e43 : SUCCESS in 1h 09m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a1f2afc13624601b08335c93b018c78 : SUCCESS in 58m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/303ef70359764b7295dc416d32366b0a : SUCCESS in 2h 17m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c064790676bd46b8994630291a6235df : SUCCESS in 1h 07m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9558ea4eae82413abcf1b47b5b9d540c : SUCCESS in 41m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/aefd14f1b4154ae19a24883b40c90f0e : SUCCESS in 1h 35m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/518569b78bd04b86b953799c02451a9a : SUCCESS in 58m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/02d80fb65b8d449b83e41ab27f0f5a66 : SUCCESS in 1h 04m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b50c8684233c4f758620f22369c8c429 : SUCCESS in 1h 14m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3c67ce6b546f49d29dde7b114a7e3eee : SUCCESS in 2h 09m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5347efa9d2eb42fdb5b792548403a62f : FAILURE in 43m 01s","accounts_in_message":[],"_revision_number":2},{"id":"78125e6a9a565506e7c705bc012c5551e2da6044","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-08 06:51:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"adeb248babebbf896e01a582dfa09d51aa146ee4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 07:17:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a82d9f4ad8841bda368c72ce3895739\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe46428640774cbabd1b5833abe52c97 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/637445c959e14b32ac69d24108461847 : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b707c5f15b9d494a873477bf04729a1b : SUCCESS in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1160d521d978e874927d0d059eaa49c680853bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:01:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5412858e75434cf5b162724b9f3e5ac2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0b59d3587ad49b38ea97ac5491b36be : SUCCESS in 1h 54m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e03888e5ad24fdeab16738015ec11a1 : SUCCESS in 2h 07m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3774640481d84504baa3ee90c7f450e7 : SUCCESS in 2h 05m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2239a1763e643b1a31aa6b51be04fc1 : SUCCESS in 1h 50m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64ea2bcbe6d340bba2a2ce3babd0bf72 : SUCCESS in 41m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1b56d6ceea84ea4b8b8f0a29f99a0ca : SUCCESS in 41m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b716055cf24a4a10b310cb000724254d : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aa2cd99b45a4badbae532896917b90c : SUCCESS in 26m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/effe4a1f0be44359b3dd644f1535ad9c : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c513e2093c5e4ed98b6b2baaadf1a5ff : SUCCESS in 11m 33s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b13aa57cc7104b95ad8f3b71513d5bc8 : SUCCESS in 29m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/22c59a8b2ecf4c19adc1226076461aa2 : SUCCESS in 52m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8ccc3e7d0fd445838da4390906fd1bd2 : SUCCESS in 1h 09m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3bcc0e3519c44d79879899e76454c713 : SUCCESS in 1h 02m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e3e7e08f04c402ea73a8ba323ec4139 : SUCCESS in 1h 53m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c93ce0256d064a459cc2b79b261c6d70 : SUCCESS in 54m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e879e24130f94df19ff50bfea24fe2fb : SUCCESS in 55m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ecb78914aba49bd8b84bf820646f221 : SUCCESS in 1h 22m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d5d9a2920cf14e8dbf1f3dde592a4786 : SUCCESS in 1h 14m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e722a6192f174affb31bc39a228a0e34 : SUCCESS in 1h 13m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8142101bbe4547c7829091f6a5ce830c : SUCCESS in 2h 00m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/354d853f55254f21b21b7740e4c9d25a : SUCCESS in 1h 31m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/db9d0a30f7df4d3d8e52d300114ea854 : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d6bc80112692470c360de7494c934daf41d49eb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:02:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bf3f5327d9623dfea1310a4d231fbceb41565ef3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 13:17:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9322e3017b7f4fe4afb651392a21ade3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2b87a09df82949b29e2bbe3a55b326d4 : SUCCESS in 1h 41m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f9159cee019476fb4d8d8dc0d40abc8 : SUCCESS in 2h 13m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/71f344ecbd0a4078aff69c45ef1b0546 : SUCCESS in 2h 16m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/dd556644ead14186bf9b0116314c4ae3 : SUCCESS in 1h 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5486101530d4eaba5b0fe77ca97f15b : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0021d7985c054521a5d5738241a8a17c : SUCCESS in 21m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a50e46e37f74fff84d5cee1fc5aa245 : SUCCESS in 19m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e4ba415de148d4b66da815bd4a0ca3 : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff72aa98544240aca18aeb34b2898fa1 : SUCCESS in 58m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/914aaa06759146f0970b2067797035d1 : TIMED_OUT in 3h 02m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5dd1e840898940768a6d83516d34ee87 : SUCCESS in 1h 24m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74afd7a1606f4f8393e073fc9dc29fd4 : SUCCESS in 54m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb0e0ed1e21f48b5a00de26358fbe137 : SUCCESS in 1h 02m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e0f9c86a5be94b1eb7cd0df20c9121ed : SUCCESS in 1h 03m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e916d579f7242fea0008a1796b44b74 : SUCCESS in 1h 41m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e7cb25eb26b4ee4e0a3218b1402fa471e6caa95e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-11 19:10:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18e38af7cdd0d099549c19e7bfd25512ba65c846","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 19:37:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2836a6e9682c48b7b760b5f806da4579\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73f3b1844ca24b0eae7a34e3cefc139d : NODE_FAILURE Node request 300-0022963727 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4fbefa100b0849b5a109c56ffbf1d3f6 : NODE_FAILURE Node request 300-0022963728 failed in 0s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/621f1c349cda40e6b72a2a9900393649 : SUCCESS in 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2ed9397adf74337b938f6edea5f9489b89578fb","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-12-12 15:30:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da7cef4e5c1f93ad3f93ec519b0dc9cb097d11ed","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-13 02:49:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9a00e22d09b68417f9f42d8dfd233e5b18743a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 03:13:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308472bc816b4c2081e85762939ababb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7984df3092d4854a833fc75827e218c : SUCCESS in 20m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68078f6235834ec0b96965d6e1ffa4e4 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44082a13c3d54e13918e2441e270bfd5 : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d2f9c178c199fae04706cc57cd83c587cb6d5f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 05:42:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2facf91051844fb0abc9f7357e1b06a5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b698d751657f48e088ff1545845a9a26 : SUCCESS in 1h 17m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e8cc5757b5f3408791849806062107ca : SUCCESS in 1h 21m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/d199bf8c28514ddd8ab541020cee6514 : SUCCESS in 1h 30m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc052882fdcf44b1932154dd6fafbb29 : SUCCESS in 1h 36m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7cc061031e694797b6be0d4f209f7d52 : SUCCESS in 30m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97b1f956395c492fa8ba236ddb909a68 : SUCCESS in 34m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b265455d2a9044a1a0a627ce4e4492eb : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ac629223894933a876d4a230602227 : SUCCESS in 28m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8622d69459584a04ada42d3842297080 : SUCCESS in 29m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f0ac9cc17b4a06856263736b9110cd : SUCCESS in 10m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f75568d976334f9382755d561352b620 : SUCCESS in 30m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cf1c9771105549eca90550670d6c884d : SUCCESS in 53m 25s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1a75c89557e24cc3a06ecad5c9b2e6bd : SUCCESS in 1h 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be951d9c2c744adea6202ab51a3cff6c : SUCCESS in 1h 06m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f60d9d3037a461094012894c9cb23ad : SUCCESS in 2h 00m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/01e982a515fc40188dae36cfc7ce7fcf : SUCCESS in 56m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfb837f2665e4781b50b87cd3f0a67df : SUCCESS in 1h 06m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9f3e7b4be184811803eb696abb2e298 : SUCCESS in 1h 26m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/56027f754412450dbe574c7a193f5ecf : SUCCESS in 1h 13m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5357ecf68154d5fab6e691714ea3805 : SUCCESS in 1h 00m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fa295d5300904c5c87c562c1d9e6dcda : SUCCESS in 2h 04m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/385b35a980584a75bb0a3f57d55623e4 : SUCCESS in 2h 11m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/96df0e4feb1b4ad091f84a117adfd93f : SUCCESS in 1h 18m 20s","accounts_in_message":[],"_revision_number":2},{"id":"6b319e203c4a3fc27e21a939471371c9099fcf41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 05:42:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bdf7733742067c4f24957681f9cb2684a9126e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 07:48:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7060f1ab0e5f49bc9f7e15282a65e8c5\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cbe40ff4626b4b99a6ce8eb67a63e909 : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7f68718738a4e69a5fc4f350191e8b5 : SUCCESS in 1h 54m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42fb7f762f2242af84f1c40c78958013 : SUCCESS in 1h 45m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9e0c291a080846708dfc830b3c78085f : SUCCESS in 1h 54m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e12e94472e14ae5996ec8685e391ec2 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47f28715fada45c7ad6ac4e1c0601f93 : SUCCESS in 27m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0134480e31024f6592d13c0b57d99e97 : SUCCESS in 34m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caccb7d6a4fb4a968b9ac45e385b1e26 : SUCCESS in 12m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9d1db948145430ea7eed83aab381ae1 : SUCCESS in 1h 04m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/57d4576515434d259c70e464b4acd704 : SUCCESS in 1h 57m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd0bf2b805cf485b9ce11ec219d69ea4 : SUCCESS in 1h 27m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f8b1403381e849b9a84bbbe4551fba71 : SUCCESS in 1h 05m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/82c7bb3eb6b4403aa70a836e6cae9ce9 : SUCCESS in 1h 06m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/64bdb8fb73a844a28495169775b0da41 : SUCCESS in 39m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f42799e7e68746e0b3badfe70604a61e : SUCCESS in 1h 11m 49s","accounts_in_message":[],"_revision_number":2},{"id":"64614471789dce5f5257e4dbb1419275f4f05293","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 07:48:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"54727cf60f03e847d9701cbecd42cd37b96aa3ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 07:49:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2386d5c119a94be799cdc53dc08279c1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34fea11812f344c98f5c25e246752e46 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62982a1ff4c7a28b20e55aaf8235d6d9448eac9c","revisions":{"81ad204b74d5270e56233f2a4c88cda15f2d62da":{"kind":"REWORK","_number":1,"created":"2023-11-20 18:40:50.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/78/901478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/901478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/901478/1"}}},"commit":{"parents":[{"commit":"368c45a692547f263bdfd6a92001dbe03cf8f75c","subject":"Remove agent veth_mtu configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/368c45a692547f263bdfd6a92001dbe03cf8f75c"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:39:31.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:39:31.000000000","tz":-300},"subject":"Remove vnic_type_blacklist deprecated name","message":"Remove vnic_type_blacklist deprecated name\n\nThis was replaced a number of releases ago by the more\ninclusive name \u0027vnic_type_prohibit_list\u0027, and is not used\nanywhere in the tree.\n\nChange-Id: Ifbe9f4cd0c291ab61e0adb93cfde866001086345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81ad204b74d5270e56233f2a4c88cda15f2d62da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81ad204b74d5270e56233f2a4c88cda15f2d62da"}]},"branch":"refs/heads/master"},"62982a1ff4c7a28b20e55aaf8235d6d9448eac9c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-28 10:18:19.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/901478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/901478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/901478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/901478/2"}}},"commit":{"parents":[{"commit":"0dcebba6769e0d7cbf1a0eed0d699c4654492753","subject":"Remove agent veth_mtu configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0dcebba6769e0d7cbf1a0eed0d699c4654492753"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 18:39:31.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-28 10:18:19.000000000","tz":0},"subject":"Remove vnic_type_blacklist deprecated name","message":"Remove vnic_type_blacklist deprecated name\n\nThis was replaced a number of releases ago by the more\ninclusive name \u0027vnic_type_prohibit_list\u0027, and is not used\nanywhere in the tree.\n\nChange-Id: Ifbe9f4cd0c291ab61e0adb93cfde866001086345\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62982a1ff4c7a28b20e55aaf8235d6d9448eac9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62982a1ff4c7a28b20e55aaf8235d6d9448eac9c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
