)]}'
{"id":"openstack%2Fneutron~901507","triplet_id":"openstack%2Fneutron~master~Iab8d3dada2e147da009e195700e64b072e5bfabb","project":"openstack/neutron","branch":"master","topic":"remove-network-update_port_status","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-04 18:37:56.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Iab8d3dada2e147da009e195700e64b072e5bfabb","subject":"Remove deprected and unused argument from update_port_status()","status":"MERGED","created":"2023-11-20 20:53:42.000000000","updated":"2023-12-04 23:26:44.000000000","submitted":"2023-12-04 23:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901507-remove-network-update_port_status","meta_rev_id":"44e796cd0357aa00cdb5e51b52e655f1950fbde8","_number":901507,"virtual_id_number":901507,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-04 23:25:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-11-28 06:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-11-27 07:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-27 08:10:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-11-28 06:46:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 21:40:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 22:52:07.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-20 23:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 07:54:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-27 08:10:18.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-11-28 06:46:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"f5489047b7da583e30925e4560b7399b907b7112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-20 20:53:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a25879f78c63a970de4ae788e3672acbdcace6ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 21:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/348ef35f7d074e87bdd16a6eb2292593\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ddc1f7b49894d329e5a8257fb583537 : FAILURE in 44m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4114cb53423f4b94bdf84c48adc61345 : FAILURE in 23m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/697d911b6d924b61acbd9a2735be81d9 : FAILURE in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e0397d68df94e65b899609508a4a2591eaf9f82","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-20 22:52:07.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/07/901507/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9781d6c46ffc83a8381ac728a8f6430b9875eca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 23:57:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4642aceabab740da88099f097d1fbc9a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d4f7d51fc4d244bb85a171f37986c949 : SUCCESS in 1h 16m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a5c20d06ec3f4ef4b103ae312536247a : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/aac0624d04df41159e5d3496ad2dc2ce : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/21ba572f3d58498d895a6279a922a260 : SUCCESS in 1h 39m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfe62a63b109425c98194a0fa7cbb292 : SUCCESS in 32m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd3ec463f0f4bc788cea27512e24879 : FAILURE in 30m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40037b6b350c497c8944f331b12080eb : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5586e671ab24499b96903f837430b26 : FAILURE in 27m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a1a41bee1ff434687f2b9fb94a7044e : FAILURE in 20m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072d0b94f3b94536be78d4f046eb2606 : SUCCESS in 10m 52s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/414618ad4adf4a7d85b9e942381e1bd8 : FAILURE in 20m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a5524e8343f94320b2d26d953d2cc424 : SUCCESS in 43m 56s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/12ddef06fa2d4e8e8acf8a3adc3999c0 : SUCCESS in 1h 04m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f30f90194e0b44be8f75f70b7835feef : SUCCESS in 58m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a6e09423b7b44c6a1f2f96fb1e07c09 : TIMED_OUT in 3h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e502eac02817400a8a428f5f924382c3 : SUCCESS in 52m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7007cc6623714f86b238c9c2e66d03b4 : SUCCESS in 1h 17m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9988b1becaa1435a9de7e057d3452b64 : SUCCESS in 1h 24m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f42e41ea37214e2f8ad84bbee3e1778e : SUCCESS in 1h 13m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a426d055fcaf4ee79be48128ec232d92 : SUCCESS in 1h 02m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bc109b4e7f4344208c0a76ef16277884 : SUCCESS in 1h 57m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/026845ec67bb476fb524b7340c5734a3 : SUCCESS in 1h 57m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e21ca87b995346faac7f85ef0e0a5e82 : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1e0234cfc4fff148dd6c2f5e18c1d4ee40d42053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 03:00:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"be4d54f63597f471c99d6f7c6a76a699c0cfdda7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-21 03:33:40.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/07/901507/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b5d4e31f19aa6a4c26bcf9dce63dc9e884e425b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 03:55:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f963cf97e2849639a5da4287f069e9f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06940fdfce4e465bb99b0c9cc5e1bb67 : SUCCESS in 20m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/948c2132d8ff451d87e0dd1da5bd2e2a : SUCCESS in 49m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06874fce5726478c9b50e3ff1087f928 : SUCCESS in 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8cb2915bb1481597e40aeebf238a78c78e2a5a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 05:09:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d7911d960a402990f4774025af5bb8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f31032125c184176b82ebf47e696769f : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/20261cc712694e449c0d0ea33674d869 : SUCCESS in 1h 24m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6bf01cebb9c14d4ca21411b48c8e0b86 : SUCCESS in 1h 36m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dc5c30602628426ba89a652cdfa50bd2 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb5037ee2d1848019ed305e9684d2907 : SUCCESS in 35m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a658b1cf3a74caabeffcc50a6a189a9 : SUCCESS in 43m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17aa302a76cf4782baf8f0cbc5659300 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c3b016ed38d48419a1ca59cbc4ac789 : SUCCESS in 26m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41a52ef335564935bc38be1dd51e1d04 : SUCCESS in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a970165c7d4aeabdb14e127447393c : SUCCESS in 12m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/44988aad8ab448a4bdba5d71874bf76f : SUCCESS in 31m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/caf2d070a75b4d439fd1834d2bdb0121 : SUCCESS in 1h 01m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d5302cb2cdbe430eb979adc8cb1b79f2 : SUCCESS in 49m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f398906381d474fbe0b03ce899cc5c1 : SUCCESS in 55m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b2e8697fb04424eaf9c569f77cf148e : SUCCESS in 1h 48m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb140ab5953b44c2a72a9cf20825b377 : SUCCESS in 1h 00m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90d875463abe4ece8363de51bf1f2ef0 : SUCCESS in 59m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0ff47d9d6ed24f3c96b9059608a9a9e6 : SUCCESS in 1h 27m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5a39e3ed817d456792b198d255264bfa : SUCCESS in 1h 35m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c2e194403ab24805ba38c77fa3954f05 : SUCCESS in 1h 06m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/181e57c1ccf348389820207d380efe6a : SUCCESS in 1h 28m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32fe5cd7c78e4948b3c7713c1c2a3094 : SUCCESS in 1h 33m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/78bd6623ba094f0494f7873761776b60 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1578b332a41f7035666cf910223f5e47171d7af5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-27 07:54:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e7588458d39cde146015cd7dc20b3251455ad4a2","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-11-27 08:10:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3e26ae4adc76007486c25e211c84d9779efbcc4f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-28 06:46:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"750308f5fde64e364875ba5e23dfb929801f46e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 06:47:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87e7a52a8b737b86f007a0c052223cb843224901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:08:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9232b1969dd4614bd284d17e46df9ab\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3d9dba3c37874489b254ab263d1df801 : SUCCESS in 2h 09m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78e5b9abe11c47919de3e24dd09d6575 : SUCCESS in 1h 58m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/05e9697a40aa4a35854ec4ba55815c0f : SUCCESS in 1h 46m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/49f836e92d004e50b494478b211e8b86 : SUCCESS in 1h 50m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8e8f30e66654e8d85f322f680ef3abc : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7454b6fee03a45baac3dc973d55edbdb : SUCCESS in 20m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a35c86706c54be99b5847f663d31066 : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e67d4916b84429bc83c560c219437a : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ffae97dce1fa416c9007b1d6613d96f5 : FAILURE in 38m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e7185f3d9714723806c17b267715834 : SUCCESS in 2h 14m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b7479fc87ee94d9e94ec05c3038425cc : SUCCESS in 1h 24m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28126809d6e74a18a08775bf96c84524 : SUCCESS in 1h 02m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/21ed4328ed6a4931aefbe72f0b7b7d34 : SUCCESS in 1h 07m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1755b8215c9c44c580a15f6354cf9de9 : SUCCESS in 1h 06m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/187d9b25cdb74a559d2f27ec40da6115 : SUCCESS in 1h 16m 57s","accounts_in_message":[],"_revision_number":2},{"id":"79d1c48f115d8556032c789e4f7bcf847709ec58","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-04 18:37:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81c4dd1303b3906bf797d61d7267f5b07ad0ad48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 19:29:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74f7c3b6cdab4071b56783f16e0ccd0a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8b1ee0ddf78468fa2f931f71232219d : SUCCESS in 44m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39b393e36573499590a53b01b54aafa3 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b4dcdff895bb41558f0adf007d70dffb : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1340c5decf7c2628d15a6b7853b71387d646bc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 21:03:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3d9fda52e50484ab8570bf0d9ae7d99\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f230c3e08a224c90bb9f2540ce34418b : SUCCESS in 1h 23m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/33d1bf7844b94035bf09eda4be4c25a5 : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/be16646ee0b34c95ae33ba4c10b9bf50 : SUCCESS in 1h 18m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c900d997f4be428887add0ef93f0b7b2 : SUCCESS in 1h 27m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1bdbd9b2c0444b1581c64966dc1f07f4 : SUCCESS in 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91537663c824479e962b8ff4c075d436 : SUCCESS in 35m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71222f78e77f4f858771394fe531bf00 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eb0342a2926465b8ed8db884e17e262 : SUCCESS in 30m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a82fccf9edc4aa28f07c27d974eb687 : SUCCESS in 26m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ae4e23ee19429ea6cd80ff3420799b : SUCCESS in 10m 13s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/09c91b68d226474d8f0127d7dd6b60a8 : SUCCESS in 34m 57s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c28a94ae7cbe445bab6422461da3ab96 : SUCCESS in 44m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/cc93830758d14bcb8139eb10f7f7c03e : SUCCESS in 1h 16m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15ee02a765f847228accb3a5c70eeb98 : SUCCESS in 1h 01m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5d6d5f18d924068943c208e0c4560d4 : SUCCESS in 2h 08m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6798e82be1104b93a2e60231971a6a5e : SUCCESS in 1h 05m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/628bc9f42fbe48229cf8bfed4188a8fa : SUCCESS in 1h 11m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e38adbd7c69244debfa2b55a3a5f3997 : SUCCESS in 1h 18m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/841a502ac78943149ac491cc427b845c : SUCCESS in 1h 28m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c53ee1524de347fdab09073e0cbefd14 : SUCCESS in 1h 05m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/535292879b4a41b1b6c1fac7d848be96 : SUCCESS in 1h 31m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/98ad0439b30c49acb00b40135391b4b6 : SUCCESS in 2h 19m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fe2dc877e05547bc84c6b4def73e9930 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":2},{"id":"e524894dcc6c6d0864e6924be366042f0dcdd93a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 21:04:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b398adf40cd4263d6c5140436fd2b6435ec6b6f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 23:25:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82e8c5d00efa4e07a2399cbd63a88e7c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/91eccbd2513f46ccaac8bf147a0ccdcc : SUCCESS in 1h 16m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fca9ac5c7a684e70982c28096882826b : SUCCESS in 1h 22m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07d4e16b599141bd8a9f11842eb23289 : SUCCESS in 1h 29m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0c7a46b6264f467a8218f17145a9789d : SUCCESS in 1h 25m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf0919d34a1455dbc150210a9fa35ef : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3715b92507d546b88e8cd4f65b155495 : SUCCESS in 22m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30bd6dce73ed43c88dcce8d8ce3d0482 : SUCCESS in 25m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37dcc67b5cb64d71a8aaa53a045b7ef3 : SUCCESS in 13m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d7feb16169848cc92c9a14b3a03e30e : SUCCESS in 1h 00m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/779b852eef2049d5b45dc82d9e72b7d4 : SUCCESS in 2h 20m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/85d3217324c64777b2e0c7457737ea13 : SUCCESS in 1h 29m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/07d4a77cc14c44c09e4ddbe1866177f0 : SUCCESS in 1h 01m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5348741be31046498dc52727bde6f2b0 : SUCCESS in 45m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ba11806887b64fab8816dbbe917a2537 : SUCCESS in 46m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c90b12c3f24f4d148a7d5bc855526a7a : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"65e50bac69231a0b90f58d01130fb10cbd182c8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 23:25:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"44e796cd0357aa00cdb5e51b52e655f1950fbde8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 23:26:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad3d5ac97834719a129abc9f6e566e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/621384b4644d4f6498cb58ed2bc1a835 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eddde3522b657c917482c7f409f3cd8fe0a0bf4e","revisions":{"8524a1683c392fdec43355d21cec0034aacfd831":{"kind":"REWORK","_number":1,"created":"2023-11-20 20:53:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/07/901507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/901507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/901507/1"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 20:52:17.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 20:52:17.000000000","tz":-300},"subject":"Remove deprected and unused argument from update_port_status()","message":"Remove deprected and unused argument from update_port_status()\n\nThe optional network argument was deprecated in\nupdate_port_status() in Pike. Let\u0027s remove it as there\nare no callers passing it anywhere.\n\nTrivialFix\n\nChange-Id: Iab8d3dada2e147da009e195700e64b072e5bfabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8524a1683c392fdec43355d21cec0034aacfd831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8524a1683c392fdec43355d21cec0034aacfd831"}]},"branch":"refs/heads/master"},"eddde3522b657c917482c7f409f3cd8fe0a0bf4e":{"kind":"REWORK","_number":2,"created":"2023-11-21 03:00:10.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/07/901507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/07/901507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/07/901507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/07/901507/2"}}},"commit":{"parents":[{"commit":"5c3f1cd02fe14b758db4cec6be1be8290c359f00","subject":"Merge \"Update OVN db sync code log messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c3f1cd02fe14b758db4cec6be1be8290c359f00"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-20 20:52:17.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-21 02:59:45.000000000","tz":-300},"subject":"Remove deprected and unused argument from update_port_status()","message":"Remove deprected and unused argument from update_port_status()\n\nThe optional network argument was deprecated in\nupdate_port_status() in Pike. Let\u0027s remove it as there\nare are only in-tree callers passing it anymore.\n\nTrivialFix\n\nChange-Id: Iab8d3dada2e147da009e195700e64b072e5bfabb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eddde3522b657c917482c7f409f3cd8fe0a0bf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eddde3522b657c917482c7f409f3cd8fe0a0bf4e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
