)]}'
{"id":"openstack%2Fneutron~901792","triplet_id":"openstack%2Fneutron~stable%2Fxena~I2cfd52a2657fad989e24e974fda470ecd960262b","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2040172-stable/2023.2-stable/2023.1-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-02-09 09:24:18.000000000","reason":"Change was abandoned"},"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2024-02-07 09:35:22.000000000","reason":"\u003cGERRIT_ACCOUNT_24245\u003e replied on the change","reason_account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}}},"hashtags":[],"change_id":"I2cfd52a2657fad989e24e974fda470ecd960262b","subject":"get_hosts_mapped_with_segments add filter agt_type","status":"ABANDONED","created":"2023-11-24 08:47:56.000000000","updated":"2024-02-09 09:24:18.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":901791,"cherry_pick_of_patch_set":1,"meta_rev_id":"50bd3a1aec5e4ef406eb5ff902ffe71c5440b5cd","_number":901792,"virtual_id_number":901792,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-18 16:46:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":-1,"date":"2024-02-06 15:44:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-23 09:47:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-24 09:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-24 11:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 10:23:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-11-28 14:42:22.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2024-01-16 12:53:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2024-01-23 09:47:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-02-06 15:44:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"6618f8728a86998a4441486fb5f752f3d21ff40c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-11-24 08:47:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"ea47b9e9b56f7776e22056bec4fdce4215fe63f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 09:35:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e059a0074dbc41ff94c800424ba4bb5a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76be6bc28da74aa2bfd8174d24bf0af9 : FAILURE in 17m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff8442f1d1534cce81cd1989ac992280 : FAILURE in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84a93d6c749819f890c79787b63663b0bf6c7174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-24 11:34:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce933d944b644fba9d211ff4eb63e78\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/7b155bd7c1b84b56a61ab6abeb8c7c3c : SUCCESS in 45m 43s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/bb521355f2514a7b93a2084b0d406e15 : SUCCESS in 2h 44m 29s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/59fdb41cf3434664ab555f4c20efeff1 : SUCCESS in 1h 55m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/582a271f9f4a40e3adfc877c10bd3006 : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/536f85015ee74ee8bec109f7bcd840c7 : SUCCESS in 1h 55m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8977295f803415b995154e542ebfa16 : FAILURE in 34m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f403948ae014360890abd5c1868b6e2 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/930816e81be940ab8a94b32f108686b6 : FAILURE in 31m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94660bc5384e4943a6e18a83b73b16f8 : FAILURE in 24m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf8d5271bbd440ed860879b950841696 : FAILURE in 22m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de76cef40fa24f49ab3a73a9b367ee8a : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/742184607617406187be002258ae6a33 : SUCCESS in 50m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/63fb82e8f6a243388b841933d9a6d343 : SUCCESS in 1h 24m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c08c056304d40d29dfa6e72be7a82f1 : SUCCESS in 1h 19m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee70d89ddafd450ba18ae4ddf4d4ebc3 : SUCCESS in 1h 04m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/08d83aaa08d14876945f99074ca034a1 : SUCCESS in 1h 13m 15s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e61ef6baae3647179675416ed5b965b8 : SUCCESS in 1h 07m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8947ebb1097442898c03ba480dde62bb : SUCCESS in 1h 12m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bda49c6a69048b6b0c1dd7886335228 : SUCCESS in 52m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c77a8f79a12442c28ee4aa2c3bbd4322 : SUCCESS in 1h 00m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a398a70806ea078edc6ee4eca63c8d0878ec0ccc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-27 10:23:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b9722ee88aadb3b2dfab9e18d0c16a41966643d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 10:43:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfe90ee160f54b14b57936cc2a6074ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f0f42efab124c8fb4d14d842eaf3a55 : FAILURE in 18m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/904c7701f4bc4ff2ac5be5475bdd4c2f : FAILURE in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e34b09e47bcb194cc6cac1973acbcca5283e610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 12:46:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48f9325848824283a604f7d2288cf9b1\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/85bbdf2f28494f318926b1925813df89 : SUCCESS in 1h 02m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/ec3bc1eecd8242eb8c70592b1ddd50d4 : SUCCESS in 2h 10m 28s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/76292f9e89d741e8912a8df1a003e1d6 : SUCCESS in 2h 17m 10s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/3ec43f0f4dfc4bf7a4a286457f85be31 : SUCCESS in 1h 40m 44s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/fe48e196e5444fcb8c001da87fece835 : SUCCESS in 2h 15m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff986d0ffdd849dfa81930998b43b106 : FAILURE in 24m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c9aa6c681b424d854879f8692afa90 : SUCCESS in 11m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ac8526eb7ad4187bf8ceed32c520849 : FAILURE in 33m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8065b9502e6446d09652f8252784ca1f : FAILURE in 27m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ffcc3627cb41e181709206963a14b5 : FAILURE in 22m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217ce4eaa57a4805aaea5ccaff484922 : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d30cd7f5317423fbdf0ac1d1f445ebb : SUCCESS in 55m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f48b35adb19b4cfcb3b5f8f9e8ce619c : SUCCESS in 1h 32m 22s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b646885930a442059e8d8228eb4321e1 : SUCCESS in 1h 05m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aee79490b12549dfadf159b2cf425797 : SUCCESS in 1h 05m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0990f980fee9479f8872a762c5e84612 : SUCCESS in 1h 10m 05s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f43e7c9680d1410c8c4bdb65fc2868d4 : SUCCESS in 1h 09m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/20dcd41002c546489c9e3c3a6fe18a41 : SUCCESS in 1h 18m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78ab4445e9b446c2b6f2a16bf26d1c16 : SUCCESS in 56m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8609e134f5ce4b59985184b569de8601 : SUCCESS in 1h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d070c77352d5c604dd72b35fa61ded621d9d8926","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-28 14:42:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71fc447ba4f1372123548aef3c4a005e333e94f8","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-01-10 14:45:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2da1234842ab0de251f3f5d42f7a40145cf9de09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 15:27:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df08ec97cf0a47c79dca80600e8fa51b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c35f15f85a7148cea95c60f65dfb657f : FAILURE in 38m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/649c71e2921a46ecbe68fef210c865b3 : FAILURE in 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"885ded6b8d6e4a65c6087118991cc615447bc814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 17:38:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84ca055f2f364d68b8872aaad9ec3805\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/8a5cf8cb8b11415d9b934246128d5a66 : SUCCESS in 38m 33s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/11fc1ac7f5fc49d2ab8b04df25b85b7f : SUCCESS in 2h 33m 05s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/0839548ff4f44c50bd21571e9ba66689 : SUCCESS in 1h 37m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/eff8a10ea39242dcb5bed0dda91789b0 : SUCCESS in 2h 12m 42s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/0ed222420c6b4b6693b371fac75ca3ae : SUCCESS in 1h 56m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4857b4685a2c467c97e2375f5c191f31 : FAILURE in 33m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1346768b0af4622b86030b900dfe02b : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b8db8f0a20443068c36f89ad8fe64b7 : FAILURE in 28m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9e22a6051a14443825f8a34d03931d3 : FAILURE in 28m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ec45f47770b495cabfc0449b87fde4a : FAILURE in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c30957e4d14bc0b03dd966f26fcaf9 : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/997591b1bd714233ad7c5e00aab0fffb : SUCCESS in 1h 39m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/df672c8d19db44f0abed63409454353c : SUCCESS in 1h 12m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5f16a73e54f244259e83d1a87bb7b31b : SUCCESS in 1h 00m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72e0d0d733114840ba9839fd318cfc0f : SUCCESS in 1h 05m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f3c79232c350406190433a7c0e6e68f1 : SUCCESS in 1h 10m 17s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/13a424085dd8470f893ec3a313fc5ed6 : SUCCESS in 1h 46m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c1b33285e4d641d9a49beaa2b188e9e6 : SUCCESS in 1h 18m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be8150a2460a4d35b2ace128dda295d8 : SUCCESS in 59m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cadbcb858a0f4c69870f809533d13d1d : SUCCESS in 1h 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f049fb84c30a04f1ecd838753d2a9cf2e3847b12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-01-15 14:23:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8acf7ccdbcc852188f301501483cba7cbdd6e0cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 14:42:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5fd69373c854dfc9689ff4226284823\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2c8b334b22b4cc38051fd071c7b2b39 : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ae2051dbdad4d039f84bf90b5a9df74 : SUCCESS in 18m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"416459f94e3a0769b59839f6c56b6735cbc4dbca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 17:35:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42a2f70e02bf40f58c31b799e07da2f4\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e737e58a77154c02acef8be44eb726a2 : SUCCESS in 57m 42s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/9b14802333834972a93d1aaa2b42faef : TIMED_OUT in 2h 50m 08s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/bae46426868d4ec182fe4cd069842838 : SUCCESS in 2h 14m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/726e84602b544a4fa6eac9ee1a055268 : SUCCESS in 2h 13m 12s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/e813bbf5a56a432e803906eb16a04f51 : SUCCESS in 2h 37m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a092a8c6d54bd7a3a254eca173a2aa : SUCCESS in 36m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cf8002ab5f42c090f3bab52c8959a3 : SUCCESS in 14m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45f7957dba6042f683f8623a277c7271 : SUCCESS in 30m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65b1636cebe844b8babc68a91eb6cef1 : SUCCESS in 24m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eba6c26387f04e36b14f4efbb55a0848 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec4b92b1ce264b01a37337eb1f7d1ff8 : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccb715d841504857bd3a204c3c9e77f8 : SUCCESS in 51m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/175b8b19df3b48218c83f32af9cadf6a : SUCCESS in 1h 11m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1a90ce2d878b4540bdaf5150d0a7c51c : SUCCESS in 1h 02m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8462313a915a440fbe29734037d23c75 : SUCCESS in 58m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5db8b003efc34d408d62ae33a1190a85 : SUCCESS in 57m 17s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6138b5c0e8b84c5cadfb143b5312f18b : SUCCESS in 1h 08m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ba8a605632a74e78b14ec0b8b8150fc9 : SUCCESS in 1h 11m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/035888055e574b859daf55fb77866b1c : SUCCESS in 56m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/50ddd76495754e97bae862364e351867 : SUCCESS in 1h 14m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"495719e830f606ac9c41262203e951307a0f9bac","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-01-16 08:01:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25ad7db3570f6c2f7e2024255c5a9857cb416e36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 08:19:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9105092638fb4dd59eefe60deac16042\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b5bfb9dab0a44ea9cd4bb7641a18f74 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ee946f66bde4f2a9a90703056d1a865 : SUCCESS in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f18d4ff4738dbe527f0fb4b6f0f2978ea88dd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:07:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e398e1b0b68042a9b770f7b166225dfc\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3bfc60e2b80d40daa8020c93d838206c : SUCCESS in 1h 01m 32s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/b714d62371eb4e8f80186eded4334bae : TIMED_OUT in 2h 49m 31s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/76559b248a6a4a35b1e8d3700f7229b7 : SUCCESS in 2h 29m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/47021ed0920649c0b0ff836d4075d673 : SUCCESS in 2h 18m 37s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/b2596874279f480c82e988d4fb7f5e22 : SUCCESS in 2h 12m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd5df99807a47fe85aaf55418c801fe : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1113275399844ef5b2d8c59132988ebf : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b47808d1e5945e1a2c8323cf350a2d4 : SUCCESS in 27m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45d0c062f71a4f87b2a5c1e5b70755ed : SUCCESS in 23m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ce1648c69df45b9bbbbf9fdd2eec217 : SUCCESS in 24m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32006add7c44225b4d99379d0cb4f96 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eeef2e2484f646798da0374916113fd0 : SUCCESS in 43m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/16e6eea952ee47dfa69bbdc271dcc06a : SUCCESS in 1h 15m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a1d760ffd37944768b982b1c3da5bbf3 : SUCCESS in 1h 03m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce891a2774dc416d9dcfb1bc6f1c4bcf : SUCCESS in 44m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1d686c28df549d99b11496fc5ec3629 : SUCCESS in 1h 05m 37s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e633a4c43ed47bca52c1ec62d46944f : SUCCESS in 55m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1018b60fa8974a4da93bd37df10c8035 : SUCCESS in 1h 20m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a11e47e6cc874809a2996a6f79692272 : SUCCESS in 1h 01m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f8d71af5ea942f4bb01ac21c815ae3b : SUCCESS in 1h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64be5b525589f9ee2e0ea4ae93071d2b60b91391","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-16 12:53:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"399d0d9ceec48b3c345662bcfca953d3fbe46dfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 13:11:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62452a786df49d68a663063861a552c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a59f6ec7b8524f758e04771cca1f2fd2 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bffc46a53a434b56b24be4679f5e15d7 : SUCCESS in 16m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47176778f4b90e0cb5dabc99640c90db721e2bb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 16:03:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98cae38017342d0ac23a0c0c0d9e8b2\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d3b2bff0a6f3431fb165c6044aa8713d : SUCCESS in 36m 13s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/478647f2871349c7b975779c9a9214b4 : SUCCESS in 2h 33m 43s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/fcf630736a594cfd873e13716474247b : TIMED_OUT in 2h 49m 28s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/b8043a94e01b490a866bcc2e545429a8 : SUCCESS in 1h 24m 44s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/6b128e540e0c4964be3f75f796e60d56 : SUCCESS in 2h 11m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c657fa2598d426aa5ab284bca862db8 : SUCCESS in 35m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e23c074a4d1147cd99aa47162148c1a5 : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c74bbdf1bba9484bac6e6bf5b1ee25ec : SUCCESS in 27m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26b881ecb8874b5583d3852f3abcf1b7 : SUCCESS in 33m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e33e6db0ccf24240adffd21d76382e80 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1d8a604cb241419fbc0aedba566bd9 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31f975c5ec594c3bb6ddaa765d1f3494 : SUCCESS in 48m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b0b7759c6d4c44efbfc4c4061a9948fd : SUCCESS in 1h 15m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/db87edb9ccd0432d828e35b90bd4d717 : SUCCESS in 57m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/582a7e68cad14de7b826e2b70d361249 : SUCCESS in 1h 16m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d5996eef7234e8c9d0b4f6486954762 : SUCCESS in 1h 00m 53s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2201ec6e8c541a484fd09271ffa4c49 : SUCCESS in 1h 03m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/915cb8d37385444a83bdeab0afd71367 : SUCCESS in 1h 11m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a50b9c4466054f17ac6201fb00711f00 : SUCCESS in 52m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e9e2629c5b6f49b09098b3e29ccad13a : SUCCESS in 1h 04m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7287fb5661e741dd2fe7ce5f4d3b162ace159d1","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-01-18 14:17:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b43f671e89f012aba2b1d2251abdc13bbb1c7c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:04:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da69f368f47b494fb8c1248836f52330\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbaaad5bb5894c3aadbf1fa281632eb1 : SUCCESS in 44m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f5889f2db8409caf19b42c45dc0068 : SUCCESS in 40m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c22f0e0a41c10acacc68b6b4ae1aba4dbb03244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 16:46:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2647538dbd47c08d6e6d948afb9f9b\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e6fe2727a5e84192a4c1e983159a9e76 : SUCCESS in 1h 11m 25s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/7b00fb81f07c43cb8179f65a0cf40646 : SUCCESS in 2h 21m 15s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/77015b2dc5b74707a0b2a4edb65c66ea : SUCCESS in 2h 00m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/7060d684cebf4955be0a4438217ba4ad : SUCCESS in 1h 41m 53s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/510d7b168a754bcbbf2930fc42b56e45 : SUCCESS in 2h 08m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d071f39c7d430881457ddfbf33132b : SUCCESS in 35m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c60fb0e33a4403394d060fd57b52ba4 : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7c9a0a576a84a64941c040acdd4e07b : SUCCESS in 25m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7386cbada6e4b93a1274933768516bd : SUCCESS in 25m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d33f9af7889c4a2fa13ef20e43e6ae1a : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4da54f479194ecfb6edc0bada1c47be : SUCCESS in 8m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06b289d6fc66457f8f93b46749e378b7 : SUCCESS in 38m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7661abbea809433bb250b45e3f3a52c2 : SUCCESS in 1h 27m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6d8f919d038c4d07afd792d10ba47fd6 : SUCCESS in 1h 31m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a63a748b071f4f98bff61a62d43939f0 : SUCCESS in 1h 08m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e37cbd08676045a29fc465dffbe9307c : SUCCESS in 1h 12m 02s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0fcc5a8917494dbbaf7e16503a79f7de : SUCCESS in 1h 07m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c0573a1c5de34742a5a86d816f6f90eb : SUCCESS in 56m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9723acbc2b246fe9d75b443d77861f4 : SUCCESS in 58m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d566a192a8524f83b9a8aa9f15f464ad : SUCCESS in 1h 05m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af8580a9690fe51ec98e567bfde665c5d11e052d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-23 09:47:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3df2c33da58384df9841c778c78babc4dfe9fd63","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-06 15:44:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d9acc4b8176e3b1e3c7d7b91e7718f4cad9e29d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-02-07 09:35:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50bd3a1aec5e4ef406eb5ff902ffe71c5440b5cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2024-02-09 09:24:18.000000000","message":"Abandoned\n\nAfter discussion with neutron core Rodolfo, dropping backporting to these older stable branches.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6481153f404bafd7b8003940dd2bd8845583db5","revisions":{"2673e2fc888646bce41b7b77a38f0fc59f69f28b":{"kind":"REWORK","_number":1,"created":"2023-11-24 08:47:56.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/92/901792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/901792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/901792/1"}}},"commit":{"parents":[{"commit":"86aa8a3a1031cbc6476399308a8f6344e9383652","subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86aa8a3a1031cbc6476399308a8f6344e9383652"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-10-24 07:38:33.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-11-24 08:47:56.000000000","tz":0},"subject":"get_hosts_mapped_with_segments add filter agt_type","message":"get_hosts_mapped_with_segments add filter agt_type\n\nExtend the get_hosts_mapped_with_segments method to add\noptional filters to include/exclude based on agent type.\nUses a joined query, when both include and exclude\nfiltering is used togheter the exclude filter is most\nsignificant.\n\nPartial-Bug: #2040172\n\nChange-Id: I2cfd52a2657fad989e24e974fda470ecd960262b\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n(cherry picked from commit 64b5787c3283b29cd36dde46ae298ae332f972d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2673e2fc888646bce41b7b77a38f0fc59f69f28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2673e2fc888646bce41b7b77a38f0fc59f69f28b"}]},"branch":"refs/heads/stable/xena"},"f6481153f404bafd7b8003940dd2bd8845583db5":{"kind":"REWORK","_number":2,"created":"2024-01-15 14:23:59.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/92/901792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/901792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/901792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/901792/2"}}},"commit":{"parents":[{"commit":"86aa8a3a1031cbc6476399308a8f6344e9383652","subject":"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86aa8a3a1031cbc6476399308a8f6344e9383652"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-10-24 07:38:33.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2024-01-15 14:23:12.000000000","tz":60},"subject":"get_hosts_mapped_with_segments add filter agt_type","message":"get_hosts_mapped_with_segments add filter agt_type\n\nExtend the get_hosts_mapped_with_segments method to add\noptional filters to include/exclude based on agent type.\nUses a joined query, when both include and exclude\nfiltering is used togheter the exclude filter is most\nsignificant.\n\nPartial-Bug: #2040172\n\nConflicts:\n  neutron/tests/unit/extensions/test_segment.py\n\nChange-Id: I2cfd52a2657fad989e24e974fda470ecd960262b\nSigned-off-by: Harald Jensås \u003chjensas@redhat.com\u003e\n(cherry picked from commit 64b5787c3283b29cd36dde46ae298ae332f972d5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6481153f404bafd7b8003940dd2bd8845583db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6481153f404bafd7b8003940dd2bd8845583db5"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
