)]}'
{"id":"openstack%2Fneutron~901849","triplet_id":"openstack%2Fneutron~stable%2Fzed~Ie0ec43566fa2b3e13b4917493336ce1519c9b6bb","project":"openstack/neutron","branch":"stable/zed","topic":"bug/1982569-stable/2023.2-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-11-29 12:17:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0ec43566fa2b3e13b4917493336ce1519c9b6bb","subject":"[DHCP agent] Fetch OVN Metadata port from plugin","status":"MERGED","created":"2023-11-27 05:27:13.000000000","updated":"2023-11-29 12:19:18.000000000","submitted":"2023-11-29 12:17:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901849-bug/1982569-stable/2023.2-stable/2023.1-stable/zed","cherry_pick_of_change":901848,"cherry_pick_of_patch_set":1,"meta_rev_id":"2e18583c432654581ab59444887b64ba8260f042","_number":901849,"virtual_id_number":901849,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-29 12:17:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-11-28 14:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-29 09:14:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-11-29 09:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 06:08:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-27 07:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 14:30:57.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-11-29 09:14:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"487d8e4290e4925af67d6be6a339336c7f235029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-27 05:27:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"7c0c8a4702201536de853f02b8700ae597be95de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 06:08:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6217128aee84b01b3d9f40c205da363\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a6c8aaab71f435ca1a6032881fad66c : SUCCESS in 38m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/747bd33cb65f4a49ba341c9e81cc8bed : SUCCESS in 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3236f462b4a479014f665f6a8565e3703c0ff053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 07:21:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a59564bffed476ebd5b322cfebcba6d\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/e4b18c8fdce74376a2e3029d80530e20 : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/ad927dc579274ff081cbd15338598bce : SUCCESS in 1h 46m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/5d77093576ed49f591bf930ba939525e : SUCCESS in 1h 20m 06s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a3b01d814f4b4be3ab3f11d8977e671f : SUCCESS in 1h 17m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee008a3a979943c9b4c880b0688a1b21 : SUCCESS in 33m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4754f66937b8497bb0af2ea3edd0085e : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ef1cce7b95f469a8dc3bac2f5dfb2c9 : SUCCESS in 17m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/701cb03f240f407d8fea90621bbc79ed : SUCCESS in 24m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f373e521913a4cbca2c9e74b56361504 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251ff3f5551348d99ab80d4f2132a575 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7869057b7dba49d4814780e41b3171cc : SUCCESS in 45m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a0840c697b846068ce52311c11c086a : SUCCESS in 1h 38m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bff4c4e0991a45bdafa70d4cc93b4d4b : SUCCESS in 1h 06m 35s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/91308d57108b4d498a05a38428125bd8 : SUCCESS in 45m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a55938294e3d4430bd9a34f5c63a3c50 : SUCCESS in 52m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e4534b208853408bb8e005c37d9b49b1 : SUCCESS in 1h 14m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d410f24de504b7fb002967bdab40d6e : SUCCESS in 1h 31m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ba92119a00e43759ab7873040b62f67 : SUCCESS in 1h 01m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f54406b4088b47d99f027ae0431f5c9b : SUCCESS in 55m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d79ccddf33e0ea8415038b880639555c53b3d82","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-11-28 14:30:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb5ce0963ea6d815eecf08b8ecd00a318e39f34b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-29 09:14:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95c3be62f49f5b062455f3f186c2e2fce1d71057","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 09:14:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b039331c4ca08a6d8eddb6844321a18d0b8b4156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:17:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f3e36c4b47b47d8a8e24ab0ceab84c9\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ee3b3369bf984c76b7c5c1663745a4e2 : SUCCESS in 1h 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f81a9b28eef4368904def1a0b09e4f0 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f56cb2ea1ae24603a98cadb1a972e8bc : SUCCESS in 22m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6276fb7965204f1ba36e9d5fd389f4e8 : SUCCESS in 22m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b45e4f033404ec2834839e56b964eb7 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16f3c2b556384d3c8e72242a3b0ab0b1 : SUCCESS in 46m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/85a3d0afc06240c6a0c608fef10d1715 : SUCCESS in 1h 39m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/48472e503f2c44ef9406c945a0807fb7 : SUCCESS in 1h 21m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf4e6298260c4b19bfe346241130eb34 : SUCCESS in 50m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f95135bf03e044568022d63354e90009 : SUCCESS in 1h 04m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af22cd6ae9b94c2d9750bce425d770d6 : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":1},{"id":"497c6d47897c88b3c71ae094deb8aa6020cbca71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:17:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2e18583c432654581ab59444887b64ba8260f042","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:19:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ebff6138a33474eaa20c89f5773dc6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce6763c8e754035b51ee9624d8be79f : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69e0dacd053454f38e724616138a519e7acbb203","revisions":{"69e0dacd053454f38e724616138a519e7acbb203":{"kind":"REWORK","_number":1,"created":"2023-11-27 05:27:13.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/49/901849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/901849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/901849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/901849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/901849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/901849/1"}}},"commit":{"parents":[{"commit":"3ab7d6e6b3c7328ce860144efba3f85b5538eb86","subject":"Merge \"Ensure ovn loadbalancer FIPs are centralized upon neutron restarts\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ab7d6e6b3c7328ce860144efba3f85b5538eb86"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-11-24 12:24:49.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-11-27 05:27:13.000000000","tz":0},"subject":"[DHCP agent] Fetch OVN Metadata port from plugin","message":"[DHCP agent] Fetch OVN Metadata port from plugin\n\nOVN metadata port from NetworkCache is being used and fixed_ips\nfor it were not available at the time network was added into\nthe cache. So let\u0027s re fetch it if it is available in the cache.\n\nRelated-Bug: #1982569\nRelated-Bug: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2213862\nChange-Id: Ie0ec43566fa2b3e13b4917493336ce1519c9b6bb\n(cherry picked from commit 0afa24d9af9425036b452ca3ffac685f2a30d47d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69e0dacd053454f38e724616138a519e7acbb203"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69e0dacd053454f38e724616138a519e7acbb203"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
