)]}'
{"id":"openstack%2Fneutron~902014","triplet_id":"openstack%2Fneutron~master~Idb10a86026a5554edb7d06087da001945420f0ce","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-13 05:55:00.000000000","reason":"Change was abandoned"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-12-13 05:55:00.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-13 05:46:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Idb10a86026a5554edb7d06087da001945420f0ce","subject":"Remove unused metadata options from ovs agent","status":"ABANDONED","created":"2023-11-28 00:56:47.000000000","updated":"2023-12-13 05:55:00.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00e9ae48f8f379bb540fb0108c56779bb8bba404","_number":902014,"virtual_id_number":902014,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-28 01:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-28 02:51:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 03:20:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-29 04:19:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-12-12 15:58:44.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-12-12 18:46:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"7f0f4aedd76c6e9d6c1eb9017cd7058112eafb89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 00:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b334b5e0c351f97ba37ac0f28f75cb765b3d729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 01:04:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eab6baa22e29b0c8a96be42020f781a70e15daae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 01:30:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb29e3ec4d764480a0e0ef31568c4ba8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3811ce36a46e40eb8512165bfed899e0 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b8e54853d96422b967ddffb076641b3 : SUCCESS in 19m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/77dd9345739f4c3f8739fd99915aae26 : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb1e0886ce89f23d8b2fe9c3137dde456d3fa808","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-28 02:51:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/14/902014/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"955f933c2a38463b203d86f8754c87213edd5b84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 03:20:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a40ced4c7d4c73870199c8ac02232e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6578ee17a9284ff8b5f0224a4caabe82 : SUCCESS in 1h 37m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e6e282b9ab7f4816832badce7fe7692b : SUCCESS in 2h 01m 19s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6a668494196a48b583f556e4a0f9ca6f : SUCCESS in 1h 33m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/575eb51d516e4cd292ffbe52b09bab48 : SUCCESS in 1h 55m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d475f71037c47cf877a813bb29f89f6 : SUCCESS in 43m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5797762385c5427b8af752404c272992 : SUCCESS in 32m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4495cb87294433db3cad84473ea9729 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f346c35c7f754dea975ec696d9a56668 : SUCCESS in 28m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0b24a8a6d2e440fb3763cd657d5b3ab : SUCCESS in 26m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e743d7356f241caa31e25b45dd3e3c0 : SUCCESS in 11m 14s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3d2b62bebabc464986ebd39e602a251f : SUCCESS in 25m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6bfc988347614fb0be835152a28509bb : SUCCESS in 48m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/5ac4db5741154a3b9b824db8793ab6c9 : SUCCESS in 1h 14m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e79980bee029415da6a46311beb1a174 : SUCCESS in 1h 01m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5ccd639287d7473b8b45519d538efe1a : SUCCESS in 2h 08m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83c7346babc44579a6f9bb7b586c1792 : SUCCESS in 45m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e53c21c089df4101893d2a39ebd8dd4d : SUCCESS in 56m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6096a8a26ad04d8890e547710f84d84b : SUCCESS in 1h 35m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/38b025cd74284fa384cb496e22c04031 : SUCCESS in 1h 24m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/95d56be2644f4254aab95ab7d86ac68b : SUCCESS in 1h 16m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/016855719ded465fab0af15fe24ead88 : SUCCESS in 1h 39m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a84da2f96954261bc53cf22a3544982 : SUCCESS in 1h 36m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3adc734d1765457b8772a57fc728305f : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":2},{"id":"7b5ec1f052f685def7fbef86aa79c6df1381f676","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-28 03:23:08.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/14/902014/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f87a2bc9d937924f2988789278928db99765d476","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 16:33:49.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"64b6e3b0961170f4450fa503f872e7400e91606e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 16:40:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"758c7ae2aaf8b505347d68b7e7914e72be1e6780","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 17:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc19e9d721b44c39f27ca6fd8268cb9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0381714504844cf8b067b08acd963779 : FAILURE in 20m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4c733ec96dd4b4a99a1353c274d6774 : FAILURE in 22m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0f49124e4744364a93462f22ff5ff95 : FAILURE in 24m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c80772ef8a11f4b0d9015f6a0f026f1725ab9af9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-28 17:12:58.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/14/902014/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"158ea2a66f91b3336c89a831f8e73389513d2f62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:48:55.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e121bcaa5d5416393e63beee4547682\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0ae7a5b32f7e4c7caf355b18dd4bf1b7 : SUCCESS in 1h 43m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df54561b3b234f2e8bf85d192d1612cc : SUCCESS in 1h 43m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0641c592846d417689b62878f06bedb0 : SUCCESS in 1h 23m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3ca87585b8740ef81c26d82371e5fd8 : SUCCESS in 1h 40m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/25c4d9e7fbe548a78afab7248057eed0 : SUCCESS in 31m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08069084ff4d4c60a3ca344cfaf59c13 : FAILURE in 32m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e0354bfa6ca4c3ca36e938001d97e7a : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1087e5b8d698430aa99aec23c632e85f : FAILURE in 25m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41721db48d424057b32e51d9bdd01922 : FAILURE in 24m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df17d995f5eb444ca2341a576101c412 : SUCCESS in 11m 34s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7ae6c9b651914686b4b1796691dfc9f4 : FAILURE in 31m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa66b1c7f182471d988784fbb6acf54b : SUCCESS in 57m 16s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da8cfc057b8347408a238c1bca3fd081 : SUCCESS in 58m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c1a44d71bd347df9a108329c91cbfc8 : SUCCESS in 51m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5626787f6f6549048330e8081456516c : FAILURE in 1h 32m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b03d35140f0044cd938ba54022fe6ea5 : SUCCESS in 1h 01m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/df9897e638104d2aab817a1713643f40 : SUCCESS in 55m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/94475208ebae42f6bf5531acd97da2b8 : SUCCESS in 1h 15m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/733bbef599bc45d09ca7490b2d1221c6 : SUCCESS in 1h 21m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/870b1809ed564563aa80d41f4169a093 : SUCCESS in 1h 01m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b78b0ba12c5a4b30ad13efbb3fccb28d : SUCCESS in 2h 02m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e8a7ced5b44345f097ccea000a1c53f6 : SUCCESS in 2h 03m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff273955eb5e4476bed8898eced3a7b1 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":3},{"id":"c684f945f3548b842fdaf983cefff65c7de62e92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-12-12 15:58:11.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"29e5e26b4845206883a773922fcfc3e90147a1e2","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-12-12 15:58:44.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f039835df9c15d2ebd0ad5f62768ec257d29483c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-12 16:29:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/14/902014/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"4deea2ae7eb08a9eefc4a7d69c233d6648a6a9d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 16:50:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e877fe3918c4b4fb9c40a5660e7a616\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b6827d3005d4d7a874cb0d042cf9bc2 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9d1d49d850e4e49b6ec128528e48d64 : FAILURE in 48m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0b0927adbe6343de97102de5a5f2c893 : FAILURE in 16m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8b869810926010b2e1b1eb2ea6a977fb07e92afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 18:22:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbab0c1a342f4e10a0023222cdeb1eb3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/63f7ee167b14458099e93a188ea794c2 : SUCCESS in 2h 03m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d5f8b6aea74949708654a98bcae8c6a6 : SUCCESS in 1h 56m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c5bd07ae88504bf2bd5616213026c291 : SUCCESS in 1h 30m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad7710de0b5e4c05b028e6f89a533225 : SUCCESS in 1h 24m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e5fb7a2f1667492f8d73c25398e1aa75 : SUCCESS in 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf5af9ee4d3f4f26b2f1bbce9cf47ce4 : FAILURE in 40m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374f861b828e4cd69c3a0d0a92b34e1a : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e8fdf835ce44db888a0e13f4b40d47e : FAILURE in 29m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d612e3edd2b4fa599c10e2d0315832e : FAILURE in 26m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f75c849f9d3b4e85a871ce7a79e84101 : SUCCESS in 12m 02s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fee5ae7246084a1d908b1ad210800971 : FAILURE in 28m 53s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1eda4ab7b7f444fd83bb5ff8015e04f3 : SUCCESS in 1h 06m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a13968b5433843dbaa83eda939c00778 : SUCCESS in 57m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/95b7f1f3456844868de2fb59baac3aa8 : SUCCESS in 44m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7f70977c34474766b12b602634654920 : FAILURE in 1h 24m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/05049c4f4e904a7b8610e9df9eb16ae2 : SUCCESS in 48m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b646be56926d4c15a1ea740d856fcb30 : SUCCESS in 1h 04m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f00cb9ddc683471b9fd72693a3c6d11a : SUCCESS in 1h 12m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7f148718989d4cfc90b68dd0e3250fec : SUCCESS in 1h 13m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fd668fbb002c4c94b85fe01f12875205 : SUCCESS in 50m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/54d6c0ac1cef4871a5e11e67fc31d749 : SUCCESS in 2h 08m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a001f01a24e44ff69a856726820e3958 : SUCCESS in 2h 16m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6751359d8cbc4cda8aedca98fe2fccdd : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":4},{"id":"d0d5a3c7a46130135b58fc8b24471ac4281ea088","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-12 18:46:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3bd15fb7bc790bdd5e5e21bb8e9a988fff69937","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:46:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ee8c575d913324bf87a698314bc8bfd58a868500","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:49:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dc7ae8266b2e097f4854e337329d69237bce2a04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:52:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d71c67a45ecbaa783e25acadb85b238ca25fbd61","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:52:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"db051a75d77133d3941bf766299458f68cdb61db","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:54:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"00e9ae48f8f379bb540fb0108c56779bb8bba404","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-13 05:55:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fb25cae40d82bf75ef4f1ff8396d8e311a558c9d","revisions":{"54a60f41f0aa75628037c0f53b15c4850bf5ad0e":{"kind":"REWORK","_number":1,"created":"2023-11-28 00:56:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/902014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/902014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/902014/1"}}},"commit":{"parents":[{"commit":"56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49","subject":"[DHCP agent] Fix route to OVN metadata port for non-isolated networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"subject":"Remove unused options from ovs agent config files","message":"Remove unused options from ovs agent config files\n\nSome options currently added to the sample ovs agent config files are\nnot loaded by ovs-agent. This removes these useless options to avoid\nconfusing operators.\n\nChange-Id: Idb10a86026a5554edb7d06087da001945420f0ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a60f41f0aa75628037c0f53b15c4850bf5ad0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/54a60f41f0aa75628037c0f53b15c4850bf5ad0e"}]},"branch":"refs/heads/master"},"69134b0346ce6cbfa5e7fbd02e768922d29b9f81":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-28 01:04:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/902014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/902014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/902014/2"}}},"commit":{"parents":[{"commit":"56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49","subject":"[DHCP agent] Fix route to OVN metadata port for non-isolated networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 01:04:48.000000000","tz":0},"subject":"Remove unused options from ovs agent config file","message":"Remove unused options from ovs agent config file\n\nSome options currently added to the sample ovs agent config file are\nnot loaded by ovs-agent. This removes these useless options to avoid\nconfusing operators.\n\nChange-Id: Idb10a86026a5554edb7d06087da001945420f0ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69134b0346ce6cbfa5e7fbd02e768922d29b9f81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/69134b0346ce6cbfa5e7fbd02e768922d29b9f81"}]},"branch":"refs/heads/master","description":"Edit commit message"},"53e93d0df4959aaa283398fbe8fa2acd59aa6729":{"kind":"REWORK","_number":3,"created":"2023-11-28 16:40:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/902014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/902014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/902014/3"}}},"commit":{"parents":[{"commit":"56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49","subject":"[DHCP agent] Fix route to OVN metadata port for non-isolated networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 16:39:16.000000000","tz":540},"subject":"Remove unused options from ovs agent","message":"Remove unused options from ovs agent\n\nThe dhcp options and metadata proxy options are not used by ovs agent.\nThis removes these useless options to avoid confusing operators.\n\nChange-Id: Idb10a86026a5554edb7d06087da001945420f0ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53e93d0df4959aaa283398fbe8fa2acd59aa6729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/53e93d0df4959aaa283398fbe8fa2acd59aa6729"}]},"branch":"refs/heads/master"},"12630ea47ca7735578d1c0b4d9066661a6b37a6a":{"kind":"REWORK","_number":4,"created":"2023-12-12 15:58:11.000000000","uploader":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"ref":"refs/changes/14/902014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/902014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/902014/4"}}},"commit":{"parents":[{"commit":"56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49","subject":"[DHCP agent] Fix route to OVN metadata port for non-isolated networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"committer":{"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","date":"2023-12-12 15:57:55.000000000","tz":0},"subject":"Remove unused options from ovs agent","message":"Remove unused options from ovs agent\n\nThe dhcp options and metadata proxy options are not used by ovs agent.\nThis removes these useless options to avoid confusing operators.\n\nChange-Id: Idb10a86026a5554edb7d06087da001945420f0ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12630ea47ca7735578d1c0b4d9066661a6b37a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12630ea47ca7735578d1c0b4d9066661a6b37a6a"}]},"branch":"refs/heads/master"},"fb25cae40d82bf75ef4f1ff8396d8e311a558c9d":{"kind":"REWORK","_number":5,"created":"2023-12-13 05:52:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/902014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/902014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/902014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/902014/5"}}},"commit":{"parents":[{"commit":"56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49","subject":"[DHCP agent] Fix route to OVN metadata port for non-isolated networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56172ed5aef07d278e0f9cd6bcd5164f3f0ffa49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-28 00:55:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-13 05:52:07.000000000","tz":540},"subject":"Remove unused metadata options from ovs agent","message":"Remove unused metadata options from ovs agent\n\nThe metadata proxy options are not used by ovs agent. This removes\nthese useless options to avoid confusing operators.\n\nChange-Id: Idb10a86026a5554edb7d06087da001945420f0ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb25cae40d82bf75ef4f1ff8396d8e311a558c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb25cae40d82bf75ef4f1ff8396d8e311a558c9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
