)]}'
{"id":"openstack%2Fneutron~902633","triplet_id":"openstack%2Fneutron~master~Ifa438d30599ad7e627c85c772ffae9ae9226f7ea","project":"openstack/neutron","branch":"master","topic":"fullstack-cleanup","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-12-12 09:46:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifa438d30599ad7e627c85c772ffae9ae9226f7ea","subject":"[Fullstack] Remove unnecessary tests","status":"MERGED","created":"2023-12-05 10:58:53.000000000","updated":"2023-12-12 09:47:36.000000000","submitted":"2023-12-12 09:46:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902633-fullstack-cleanup","meta_rev_id":"fc89e3f67bc9172d1d12c30cf74fb728e1318755","_number":902633,"virtual_id_number":902633,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-12-05 11:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-12 09:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-12-12 08:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-11 19:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-12-12 08:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-05 11:53:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-05 13:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-11 19:02:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-12-12 08:09:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d2ea47b2b8808f4dccdb74806e919ac69909f9ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-05 10:58:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c853c3f093bb3c95cb9fd5522bd857fc8b559620","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-05 11:53:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/33/902633/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2f371c910086a8247016149059350fb6c6ebef2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:43:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6aa8d814ff4c6e918529d17f88ca40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2a7283c53c4e579df159db84c67a9d : SUCCESS in 9m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10c73519dc24f7aa470ec6516674690 : SUCCESS in 12m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a27ec0cc80014f4c9c14ddd2a8377354 : SUCCESS in 2h 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"764dd627fb4975e3cf9074d97e418cd7b2a8945b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-11 19:02:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"325d85f753277bc6469ddc4c0b9c7d6aeea72e01","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-12 08:09:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"627bd805d72ee8d9a4cbeee0360ce8bd52c724c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 08:10:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6959b93f28e5546a81eebcb979851534573dc05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 09:46:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae569f211864f4485bd734d00f07297\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33603d3883f46f0839d82948d361088 : SUCCESS in 10m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c008449702414d9e2a0a7257a42315 : SUCCESS in 11m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3deaa4e73b6d4b8f96fbaa37614a2415 : SUCCESS in 1h 34m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bbc3b3a4b7a51d2485ef288a1b9304246e9a2329","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 09:46:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fc89e3f67bc9172d1d12c30cf74fb728e1318755","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 09:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6971732eeafb45939c39c86495c0f9da\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c827717c0f384d9f9c6826886d3c8a62 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1a0ceb86c64e6c312ad680b7051ccb931eb5fe3","revisions":{"c1a0ceb86c64e6c312ad680b7051ccb931eb5fe3":{"kind":"REWORK","_number":1,"created":"2023-12-05 10:58:53.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/902633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/902633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/902633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/902633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/902633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/902633/1"}}},"commit":{"parents":[{"commit":"729baeb4e7eba4fa5775e4620cc5ef72b91a9da4","subject":"Merge \"Remove deprected and unused argument from update_port_status()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/729baeb4e7eba4fa5775e4620cc5ef72b91a9da4"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-12-05 10:13:44.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-12-05 10:31:42.000000000","tz":60},"subject":"[Fullstack] Remove unnecessary tests","message":"[Fullstack] Remove unnecessary tests\n\nThis patch removes some of the fullstack test cases which aren\u0027t really\nneeded because they are either testing some use cases covered already by\nscenario tests or some other common tests.\n\nRemoved tests:\n\n* TestOvsConnectivitySameNetwork.test_connectivity - basic connectivity\n  test covered by many other test cases,\n* TestDhcpAgentNoHA.test_dhcp_assignment - basic test checking if\n  network was assigned to the DHCP agent - it\u0027s tested by many other\n  tests,\n* TestLegacyL3Agent.test_namespace_exists - test which only checks that\n  qrouter namespace was created by the L3 agent, not needed really,\n* TestLegacyL3Agent.test_east_west_traffic - covered already by many\n  scenario test cases,\n* TestLegacyL3Agent.test_north_south_traffic - covered already by many\n  scenario test cases,\n* TestBwLimitQoS.test_bw_limit_qos_policy_rule_lifecycle - covered\n  already by neutron-tempest-plugin scenario test,\n* TestQoSWithL2Population - trivial test which isn\u0027t needed really,\n* SecurityGroupRulesTest.test_security_group_rule_quota - already\n  covered by the neutron-tempest-plugin admin api test cases,\n* TestSubnet.test_create_subnet_ipv4 - already tested in many scenario\n  test cases,\n* TestSubnet.test_create_subnet_ipv6_slaac - already tested in tempest\n  scenario test case,\n* TestTrunkPlugin.test_trunk_lifecycle - already covered by the scenario\n  test from the neutron-tempest-plugin\n\nAdditionally this patch removes monkeypatching of the init_handler\nmethod from the neutron-ovs-agent. It was needed only due to the trunk\nports test and is not needed anymore.\n\nChange-Id: Ifa438d30599ad7e627c85c772ffae9ae9226f7ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1a0ceb86c64e6c312ad680b7051ccb931eb5fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1a0ceb86c64e6c312ad680b7051ccb931eb5fe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
