)]}'
{"id":"openstack%2Fneutron~902677","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I6da1b135c83ecfc41ec91e907ebf8500325a7a80","project":"openstack/neutron","branch":"stable/2023.2","topic":"check_haproxy_running-stable/2023.2","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-12-08 15:36:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6da1b135c83ecfc41ec91e907ebf8500325a7a80","subject":"Metadata: handle process exceptions","status":"MERGED","created":"2023-12-06 14:53:26.000000000","updated":"2023-12-08 15:38:29.000000000","submitted":"2023-12-08 15:36:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902677-check_haproxy_running-stable/2023.2","cherry_pick_of_change":890986,"cherry_pick_of_patch_set":23,"meta_rev_id":"9bcfe2696b75c7fe0fe5268b30c48134d2dc77ef","_number":902677,"virtual_id_number":902677,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-08 15:36:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-12-08 09:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-12-07 02:47:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-12-08 09:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-06 15:14:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-06 17:04:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 02:47:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-12-08 09:44:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"0d2568809e159b7d41fa230075bce56afa93d25f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 14:53:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cbcd418ea7bff008c5262098adca380e791a2bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:14:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac45dd6a93c142d69586b1456c80a094\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7313d0d9e9ef4393a1d1b1c0e6c016d2 : SUCCESS in 19m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56e76aa8c4654983898940779fd92484 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5459f4e31db345019f8feeeb64782769 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9cb9633ea5eb489583191156b8d122ed : SUCCESS in 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7295c1c9f2ef8fde9739853fd7285a20a102e613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:04:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b1a8f493274f40803c22ae99e2d7e3\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/42ca68decb4a4e9eb150c80ecdbf86ae : SUCCESS in 1h 12m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/937ae9fb2cdc4b69ba2e7ade507a7632 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/ba2cfbe8a12a441dbc4634aa347aad6a : SUCCESS in 1h 35m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95e5d535ad7c4620a8e990812664abe2 : SUCCESS in 32m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308467ce9f9542c296ffdf84544ab955 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00d8b33f56d946e18c7db13c5b1fd4a3 : SUCCESS in 22m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fec5bb26ca34a1dbc0934db4cfa6c5c : SUCCESS in 19m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57349a57198d4b5384f7d2bc396d9690 : SUCCESS in 23m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b06c219dfc84aa9b15ca5e4b32f3767 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ea00aecfdd45d784db9de106af683b : SUCCESS in 11m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d759447515ef4571a62a93992d876f88 : SUCCESS in 1h 04m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a4818005a984d75b2bec107b5e64f96 : SUCCESS in 2h 05m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d05b336d531e42a7b5d95c17fdd0f17f : SUCCESS in 52m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a1358febf6ea47079177313e6f965cc9 : SUCCESS in 1h 06m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc52a771a3384dbb931ae72944ba2502 : SUCCESS in 1h 31m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/81fd93758fb64c58b76fd9de4d254399 : SUCCESS in 1h 18m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fab28174643144638ec3fc467742c170 : SUCCESS in 49m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2383c1772c11450b82b6f5145a094675 : SUCCESS in 1h 27m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe4bb4c88e9947afa400e03731197647 : SUCCESS in 1h 33m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2602155e0e254db7bcbc878fd5520411 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c7454acbb741666e1f94fa84d507752edc2af722","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-07 02:47:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f75cfd0f5dd721c3a7c532f2f9f362e97db4c36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-08 09:44:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c80dfaf3a4da4e237bdadd203036a863221c516c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 09:45:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a78a810f017a63b9cd1ac43f419b45e0f17c115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:36:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182f04d01c094482bc9f60071407753c\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/729fe297f16040ab80eab5c2bf1a7422 : SUCCESS in 1h 48m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9dd802b9a143fd829cc95897ce861f : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e8aebf627d14a09817a7595b91492e9 : SUCCESS in 36m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4347381b60a34afda50d2a80eb16e890 : SUCCESS in 23m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14e9881f073d49a284a7c74143f87a41 : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a6e721d8d14c76b7e54fc856f09e5a : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82a662dc69d94b6c97d443e2dd5acc6a : SUCCESS in 1h 16m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe61e93d91d446c983502e5c9722b916 : SUCCESS in 2h 03m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1410a3d3f24642bc8775a1217f6b4c4d : SUCCESS in 1h 40m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2df9d49ba5f54f3da9e82f2bff297885 : SUCCESS in 50m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/37e35ff7b87e43ddbd2f0dbe4785aa1a : SUCCESS in 59m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c635b01f9e24dbc89087b64d0b57464 : SUCCESS in 1h 00m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/142d96290fda43b09a30a52574f95965 : SUCCESS in 1h 40m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5dee31d21003b7d88d5490a752c303ab3dcd2ad1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:36:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9bcfe2696b75c7fe0fe5268b30c48134d2dc77ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 15:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681bf42ae01d4f7a8fb291b48908c3c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79bbf25f5cad49d09e7ca9375c38a3e2 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14a830fb36dcbd7793cb766c68945660a622a37e","revisions":{"14a830fb36dcbd7793cb766c68945660a622a37e":{"kind":"REWORK","_number":1,"created":"2023-12-06 14:53:26.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/77/902677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/77/902677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/902677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/902677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/77/902677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/77/902677/1"}}},"commit":{"parents":[{"commit":"58f8d6797eb467fb2018de01f55f58055531ce41","subject":"Merge \"Remove obsolete PID files before start\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58f8d6797eb467fb2018de01f55f58055531ce41"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-06 14:53:26.000000000","tz":0},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14a830fb36dcbd7793cb766c68945660a622a37e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14a830fb36dcbd7793cb766c68945660a622a37e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
