)]}'
{"id":"openstack%2Fneutron~902841","triplet_id":"openstack%2Fneutron~master~Ic5d9608277dd5c4881b3e4b494e1864be0bed1b4","project":"openstack/neutron","branch":"master","topic":"bug/2045811","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-12 06:26:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"34980":{"account":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"last_update":"2024-04-06 00:26:55.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-04-06 00:26:55.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-04-06 00:26:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic5d9608277dd5c4881b3e4b494e1864be0bed1b4","subject":"Fix KeyError failure in _sync_subnet_dhcp_options()","status":"MERGED","created":"2023-12-06 21:42:13.000000000","updated":"2024-04-06 00:34:37.000000000","submitted":"2024-04-06 00:26:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902841-bug/2045811","meta_rev_id":"29021acddb6136ab2b1a86255a39e06f001c09c8","_number":902841,"virtual_id_number":902841,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":1,"date":"2023-12-11 23:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-06 00:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-12-12 06:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-05 22:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-12-12 15:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-05 22:45:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-06 22:03:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-06 22:14:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 00:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 11:11:52.000000000","updated_by":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"reviewer":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"state":"REVIEWER"},{"updated":"2023-12-11 11:10:48.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-12-12 15:27:28.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-01-11 22:40:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-04-05 22:45:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"62078bd372dbb872cc91c2410f6adafa2323f216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-06 21:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b55be0a8d6b0ef2ce3a1b2cc17b666fe5d02bb6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 22:03:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0876f7712bc94ec1bed207670ad3003e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9c983f5fcc84fbf944e54a872ccb35e : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/195ec15d5c6b48b78ca6167c648bc5f6 : SUCCESS in 20m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d794f758a014f3b98e4a39d9bf16d86 : SUCCESS in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7864283cf94e309ea5f366dc831f2df75dc97205","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-06 22:14:06.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/41/902841/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"57ddb5d73f8b1e848fb003ab2ed75db51dfd58f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 00:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b140e0fc3cd74c3685350133c2c57278\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e8cfeca365cd4c77a26788419fca8d74 : SUCCESS in 1h 36m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92bac7a41fa047d58f2536345329bdea : SUCCESS in 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b63e18715e204608ba2fd887a7e8ed18 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/941aad5dd89a473481f721e5f21ce2b3 : SUCCESS in 27m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05d8d418b59846d18db119eb80b76e86 : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf7907394e9468ebf0348faa6f6eef3 : SUCCESS in 12m 56s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fbd21dd0d9c2422f9a67416b2f24d7d9 : SUCCESS in 24m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b626019cd8694281821bb87503dbd1ee : SUCCESS in 48m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/547fba061a824be39be632e9d4116ac4 : SUCCESS in 1h 04m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c27d497bfff846299ec8368f3c1d50dc : SUCCESS in 52m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d0369c7931a54d1690021045c05f1e09 : SUCCESS in 2h 12m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9c2275e4090c48fbafdf5020d0f675ed : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bd1db8f84bacf0d88378a1f14dfc2225ca19fa85","author":{"_account_id":34980,"name":"Mustafa Kemal Gilor","email":"mustafa.gilor@canonical.com","username":"mustafakemalgilor"},"date":"2023-12-07 11:11:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d4142e563b7b1082508383b5480f3532cfeb802","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-11 11:10:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"265eacf0b5a4b1d9f9c589e4a892f0a73baa573b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-11 23:10:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6ebd9eb48b029dd2f512f2d63a24719578d4de8f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-11 23:10:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78c68d2da94b4b0d12eb5874e70bb7046e1620ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-11 23:29:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060824ad087f4d94829a2f5c06674c4c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/785f7fc4c28146e1831cd1b9fcc9fbab : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c75e326e687d4b4898d2a38ce68f2160 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d787cd698bd14679a5efc51f0f7b01ab : SUCCESS in 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ea10243dac2d88d25294f6b86bc8788c7baa6ca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-11 23:42:31.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/41/902841/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5596c48b030229be643f59b3174359e8205d0e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 00:47:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f4c63269a74f92a8f1ff412ac58c4a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b7ee712775c64a0e8e5e094a27711b18 : SUCCESS in 1h 35m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c30d0b2af0445cc952110b6bab55da3 : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cfe2a170efc41e2b5c548e8fbb4968b : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc03c05f986940f797235f19324b9362 : SUCCESS in 25m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5da6f93bd6124258bc8fb70486c146d3 : SUCCESS in 19m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19ba9ba56124806b38930a827c460ff : SUCCESS in 11m 26s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9696dc02e79c4213bf8af6269e136cfb : SUCCESS in 27m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/015ae8af77ec4aff9b4b0fad1898e247 : SUCCESS in 45m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c32fefd6cdf41309511d62cba63d22b : SUCCESS in 59m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b2bb2e7d0e8749aca425a6ecf28d56ac : SUCCESS in 1h 06m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/704b3dd7e8ba4bbca27e701ac5314592 : SUCCESS in 1h 20m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0955514cbd5b4eba94c1a9a8095c29fc : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3d841a465827348d67d5e328566352c4c40d955d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-12 06:26:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"106a59a5359f931a3b1678a01d17b29e259c8997","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-12-12 15:27:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c6a1cdcd97aaa47ef9969085a6413fb5560bf223","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2024-04-05 22:45:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a853b46a0e5472498aa479541d3eeaf27b6341eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 22:45:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"429664bc26b66f64d82e64579082f91bd93f89cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 00:26:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32022054ddff43409e7b14d61ab5bfc5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e9be9245b1c347c6b9d124fca0e5e047 : SUCCESS in 1h 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13982d1aa40a494e8fc54f9d157d6b2b : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bc1414ae86c4a87b0930ace66d8b32d : SUCCESS in 18m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79c543d50e5f4c5b8abdf04120037a41 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b43de9ae3f840b3947b9760503f6489 : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eeb9669e6e7c4d8ead6f9becd0a83f3c : SUCCESS in 39m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/352fb0fe6ad34c229844d6df0d854619 : SUCCESS in 43m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c1097bb2d05c4c5ca9dcf317956ce258 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f179a15285cf898b870b62aafdf510d5a9e9c50c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 00:26:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29021acddb6136ab2b1a86255a39e06f001c09c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-06 00:34:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9a26e05e0f14432b76f2c0a9f5a2b6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0842e2c8f7384a85b8a89bf57e0d6bb0 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e4323e1f209ea1c63fe7af5275ea2b96f52b8740","revisions":{"0da20b30830a4e4c00661d354429595159878e60":{"kind":"REWORK","_number":1,"created":"2023-12-06 21:42:13.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/41/902841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/902841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/902841/1"}}},"commit":{"parents":[{"commit":"58dcd30dbba67464f6fd7880ce7aee543156af65","subject":"[Fullstack] Double check that agent is dead when it should be dead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58dcd30dbba67464f6fd7880ce7aee543156af65"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-06 21:37:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-06 21:38:54.000000000","tz":-300},"subject":"Fix KeyError failure in _sync_subnet_dhcp_options()","message":"Fix KeyError failure in _sync_subnet_dhcp_options()\n\nIf the netron-ovn-db-sync-util is run while neutron-server\nis active (which is not recommended), it can randomly fail\nif there are active API calls in flight to create networks\nand/or subnets.\n\nSkip the subnet and log a warning if detected.\n\nCloses-bug: #2045811\nChange-Id: Ic5d9608277dd5c4881b3e4b494e1864be0bed1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0da20b30830a4e4c00661d354429595159878e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0da20b30830a4e4c00661d354429595159878e60"}]},"branch":"refs/heads/master"},"e4323e1f209ea1c63fe7af5275ea2b96f52b8740":{"kind":"REWORK","_number":2,"created":"2023-12-11 23:10:19.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/41/902841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/902841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/902841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/902841/2"}}},"commit":{"parents":[{"commit":"58dcd30dbba67464f6fd7880ce7aee543156af65","subject":"[Fullstack] Double check that agent is dead when it should be dead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/58dcd30dbba67464f6fd7880ce7aee543156af65"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-06 21:37:24.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-11 23:10:06.000000000","tz":-300},"subject":"Fix KeyError failure in _sync_subnet_dhcp_options()","message":"Fix KeyError failure in _sync_subnet_dhcp_options()\n\nIf the netron-ovn-db-sync-util is run while neutron-server\nis active (which is not recommended), it can randomly fail\nif there are active API calls in flight to create networks\nand/or subnets.\n\nSkip the subnet and log a warning if detected.\n\nCloses-bug: #2045811\nChange-Id: Ic5d9608277dd5c4881b3e4b494e1864be0bed1b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4323e1f209ea1c63fe7af5275ea2b96f52b8740"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4323e1f209ea1c63fe7af5275ea2b96f52b8740"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
