)]}'
{"id":"openstack%2Fneutron~903237","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I3b7c5d73d2b0d20fb06527ade30af8939b249d75","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1973347","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-19 18:55:44.000000000","reason":"Change was abandoned"},"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2023-12-19 18:55:41.000000000","reason":"\u003cGERRIT_ACCOUNT_19298\u003e replied on the change","reason_account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}}},"hashtags":[],"change_id":"I3b7c5d73d2b0d20fb06527ade30af8939b249d75","subject":"Move ``PortBindingChassisEvent`` checks to ``match_fn``","status":"ABANDONED","created":"2023-12-08 21:04:38.000000000","updated":"2023-12-19 18:55:44.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"217fcfbcd49f3843389bde89c2c1ca4a2d4b882b","_number":903237,"virtual_id_number":903237,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-12-08 23:12:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2023-12-12 11:51:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-08 21:04:38.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-12-08 23:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-12 11:51:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"d3afc4845bba224ae349d9ac6a8fd02bde213ac2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-12-08 21:04:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fda9fddae486ff7d5f47262038447de05f3d3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 23:12:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98152e3e2f294d03b52c8119db3868eb\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/aef3727b3c23498ba2dde8c792688c04 : SUCCESS in 50m 11s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/6aa907c29c854de2b4f277b13694dc2f : SUCCESS in 2h 06m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25c9f4267d174359b5557ac32585d054 : SUCCESS in 34m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0953ed1cada467084ba06b794434dcb : FAILURE in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ff364bf6904c04b05946c2b2393b24 : SUCCESS in 30m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/139ba8b075b047fea75f68678c84607a : SUCCESS in 20m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f65344933724dcf833e48840a5ef927 : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d54efa731ae04dcc84e30ae318663df0 : SUCCESS in 1h 21m 44s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/3a6aedf797fb4564b70dca1559bb925c : SUCCESS in 1h 14m 38s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/c13d41fd02b5462b8c9d8de58c74dd8b : SUCCESS in 59m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bdf45b962e2e408ab410593789321b33 : SUCCESS in 45m 11s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/871df9caae744d799bdaad753e26da09 : SUCCESS in 1h 07m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/961826d81f6a42348e59732e512ad91f : FAILURE in 28m 46s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e57e676c216145269fee6149997f5d61 : SUCCESS in 1h 20m 22s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/665123fc003f4eeb95f57965ae2b756e : SUCCESS in 1h 15m 44s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6fd58e3ee9574270ac10a2c900218244 : FAILURE in 30m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovsdb_monitor.py","accounts_in_message":[],"_revision_number":1},{"id":"df2567bd7ddbc5df5bc7d43a6c02d5b53a65e39d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-12 11:50:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d95be6beef7d9619b78e49ed156a73a50cab9fb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-12 11:51:03.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"a67d997ffcf6b78850729ba591af2043b4286875","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-12-19 18:55:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"217fcfbcd49f3843389bde89c2c1ca4a2d4b882b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-12-19 18:55:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"613337e8ae43a90f03939351852bb6c97ddc39fb","revisions":{"613337e8ae43a90f03939351852bb6c97ddc39fb":{"kind":"REWORK","_number":1,"created":"2023-12-08 21:04:38.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/37/903237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/903237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/903237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/903237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/903237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/903237/1"}}},"commit":{"parents":[{"commit":"64da695c7912dd6acbbd116f85eb124952b1d858","subject":"Skip \"PortBindingChassisEvent\" if revision number changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64da695c7912dd6acbbd116f85eb124952b1d858"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 22:02:04.000000000","tz":0},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2023-12-08 21:03:42.000000000","tz":-420},"subject":"Move ``PortBindingChassisEvent`` checks to ``match_fn``","message":"Move ``PortBindingChassisEvent`` checks to ``match_fn``\n\nMoved the event checks to ``match_fn`` method, that is the correct\nplace to execute them, before the ``run`` method is called.\n\nThis is a follow-up of\nhttps://review.opendev.org/c/openstack/neutron/+/842147.\n\nRelated-Bug: #1973347\nChange-Id: I3b7c5d73d2b0d20fb06527ade30af8939b249d75\n(cherry picked from commit 6890204765c5de1a91284b9b0b6bf0565673f53f)\n(cherry picked from commit c49a3fba3a67f1fd0aea1646dbf743e7b7793062)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/613337e8ae43a90f03939351852bb6c97ddc39fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/613337e8ae43a90f03939351852bb6c97ddc39fb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
