)]}'
{"id":"openstack%2Fneutron~903510","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e","project":"openstack/neutron","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-18 15:08:04.000000000","reason":"Change was abandoned"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-01-05 23:59:33.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-01-05 22:16:52.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-01-05 23:59:33.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-12-18 15:08:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e","subject":"Enhanced external port scheduling","status":"MERGED","created":"2023-12-12 14:48:22.000000000","updated":"2024-01-05 23:59:33.000000000","submitted":"2024-01-05 23:59:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903510","meta_rev_id":"507a8340fcc3f1b2bc0ea8095bfc541446b0ff5c","_number":903510,"virtual_id_number":903510,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 23:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-12-20 08:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2024-01-05 22:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2024-01-05 22:16:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-12 14:51:02.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-12-12 15:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-12 16:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-13 10:31:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-12-20 10:55:45.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-01-04 11:41:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"f10229d88a524e48917bade82fd9915df3df2442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-12 14:48:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cacd525bc5999e3ad5a2f835abe6f8bfcd00687","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-12 14:48:54.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc4db854a88b7dc79463652cedecc3f522d79a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 15:09:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e651748df847468ea0979a3b980eee1f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/328582c7d4e042b597f9cba1d4a39824 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03dc863944a2480bb54e4362cbf48c72 : SUCCESS in 18m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb6ff46c802ebe7f8830fc60c9cdb174603ead4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 16:45:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd07647a82974007b483457253fde06d\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/6c0eba7555844604a11e80e2065c2902 : SUCCESS in 1h 48m 31s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/985d398d4f594d45bdf1ecaf4dbdce4e : SUCCESS in 36m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2815a2b5e5424d59b33361b82945aff9 : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b20a021e8674f59a2fda6fbbdf10d21 : FAILURE in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be0c1324953c43f6b2e5175ae9b36270 : SUCCESS in 24m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e257148d9d9a4167a5aff6a474620153 : SUCCESS in 23m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3dabd0139ee4028b846addd813cd280 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ddfc9f41ee34f1c938b6029156ffbb2 : SUCCESS in 12m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7c4c7c6d1f63421c8df11ee0980d9c08 : SUCCESS in 20m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0be0aef8707948e59464e91cb9b43e23 : FAILURE in 23m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/73411bcbf1b84fb3bb718d6ab57d3444 : SUCCESS in 1h 55m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2e7eff838eb6423799c587c2ba0c0ec7 : SUCCESS in 1h 03m 34s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b5993a4a4f074d24ad14cff57324d393 : SUCCESS in 58m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62837b8a6eca44f780b28916cd1808bd : SUCCESS in 1h 44m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00abe8bc62db4219aa05925b75390679 : SUCCESS in 59m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/50d30a4c63bb4568a0864a911cd770bd : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c979d80b2e375807937e50773b26002527ea6202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-13 10:13:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fbefa89dc80fa0cdb3be20187bd06268c89ee223","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-13 10:13:35.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"bff47a0ab6f44724b85e150333e26c759c2bb27f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 10:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e8418f9eb0e492bbcd8fccd6621f62f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c5e04d7bb614a1ebe1a26bc22fc6837 : SUCCESS in 17m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8966bba5b9d444fc84c70c08929d9516 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47cfb228513d3faf957193882158352299db58e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 12:35:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5b9effc9e004660ba95794ae16d9d87\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/e4ec8fed81e94e368ba77583fd48941e : SUCCESS in 2h 01m 51s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/20e3f7a229ae41b698cc54c8d4f34c87 : SUCCESS in 30m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b60ca35a2dc47b5ab544c677264eba2 : SUCCESS in 31m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f1b3518695e48f0bbe46d708e124ac8 : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ded37d8e5184e9bbe3e34ead6778d7a : SUCCESS in 25m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8676d2f6379a475bac30e70b65ea3d56 : SUCCESS in 20m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa945db11131443e894ce25d353582ce : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7d8c0a434e24598948ac21c4d8756d3 : SUCCESS in 18m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/51871cff2a5242819543e29ea52d91f1 : SUCCESS in 20m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15ba5492e8624cc088ea81cef4b9e363 : FAILURE in 51m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3b0c5c5b7a843d3ba4faf8073047cd1 : SUCCESS in 1h 43m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/409763820bc84637aa9aff7be21fdf6f : SUCCESS in 1h 24m 22s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/967828f288dd49859130f0e6e10a540b : SUCCESS in 1h 11m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/008ef95ba74c434488f755b0b2151070 : SUCCESS in 2h 11m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/32e97968e09a4107bd9fa2de148eb0d4 : SUCCESS in 59m 18s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2e9866a8762b4a088298412baefc3184 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ed51dd31570d889166a95d65b088e4d2c8249c1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-14 09:31:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"68d4ce5214092765b679214dea73e539e37d1f3c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-14 09:32:11.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"350546b1f64d94a8a5b27b9efcdd2a4dcb85ae9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 10:31:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f260751abea34c6a93fac500c7ece51d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99da922e592f48a4b1ef97b0c39c1ab7 : SUCCESS in 37m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39db34cce7da4d828ccd1fa467369547 : SUCCESS in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec6c9fed9dd4ccfebe15f8a0b23b6e88ef05bb9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:44:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe090e82b774f65820041ce68f555cd\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d36e38c129464d27aa46a3bb7bf0925f : SUCCESS in 1h 24m 53s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/01a97c07e3b541e0a48d726c6a168970 : SUCCESS in 44m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d8409baa814231a9844790ba0ad9e9 : SUCCESS in 34m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8b1eea277b44f9bf5bcbf45e0a8299 : SUCCESS in 13m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8072f8e2f0a425387b96e70f12caaff : SUCCESS in 22m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9079cab5ab94ae5b79e709744eb0b71 : SUCCESS in 20m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53edcef9610a4376a931c4101cfc553f : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6378d93d361d41eda3bffbb45dc70837 : SUCCESS in 10m 58s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a2cb804f76a44d1c89e207f702491693 : SUCCESS in 26m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54ebe03989b04d8fb72ba73ab1e48bca : SUCCESS in 57m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a592655e65245958e88617f61b48619 : SUCCESS in 1h 45m 13s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/975c324e86704fed85c4f46aad389edb : SUCCESS in 1h 14m 15s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6a2d925feb76469cb7fab7e0b56303ef : SUCCESS in 52m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/231eaaf064c9468d8c732de44a25a56d : SUCCESS in 1h 26m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bfc39769e2774c6c9afde4107ebd5574 : SUCCESS in 1h 15m 08s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3206c4da1b284fbdbf069cdae5f6e638 : SUCCESS in 1h 08m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e06ce87b7d47591a3750a5b8b5f87a3c50a3316e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-15 13:24:41.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"2bd20a1bd14e983e93434aaa0aa9d3b787de95ef","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-15 13:41:14.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"0316a7c42657d4635532c43409a0f056c56b12c1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-18 15:08:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"9867e1a277afcd006091ccd8543d817d53354cfa","tag":"autogenerated:gerrit:restore","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-18 15:10:38.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"0cb0acd4a955cb024d10f469412a1513ad01291f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-18 15:10:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b0fecb3f7160bdcfaa6237b29c8667640cd68e49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-18 15:44:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2fc06aa8076ec3fbfe3239b1ec808381449e2110","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 16:01:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678945f3e7564ac590e1848dfece0a59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61ef03241c8f4668b6f99d152f29c144 : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/588a712e85fa4265b2c69e4197d0ebc5 : SUCCESS in 15m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"68c014f1eb7c03ba4cc5ee7bf3f6b6efdbc6ce2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 17:37:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18a66b2c812d43aa85125b9c13d67961\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9d3466285a9f4acfa1c7d9cb1f5f76b4 : SUCCESS in 1h 49m 59s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/7b549ff2d0da4ca3884b2423169aefc9 : SUCCESS in 29m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebb5c613c5494a1789d3cfbdf497e171 : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585558dbeeeb4c74a31aa39b74960892 : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa56cf0608234a369f1a4c13d7b6f81f : SUCCESS in 25m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78dbcaed89da4f709735cd1481a55454 : SUCCESS in 23m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1416776a5d92420286ee98c7c1489c3e : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45dfcee817fd448a8de4f3cec97688ca : SUCCESS in 16m 02s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9b49e1f7903848e49bc1e26137dba509 : SUCCESS in 21m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c45bac96469b453db65d47a13b823332 : FAILURE in 22m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/545cda527b984e03a65c06ecde8b9e65 : SUCCESS in 1h 23m 41s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a0c9c1faa77a4ce0b8b757a31198d1ae : SUCCESS in 1h 04m 20s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dec5a1bb9d3445458abbec0fd606c72f : SUCCESS in 1h 06m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/200b6feac9ba4b9ab2287579984080c8 : SUCCESS in 1h 32m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/39206994b9a84ff3830957298d62cea6 : SUCCESS in 1h 03m 48s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/21ecca7cab6d43ea931e100c00d82350 : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":5},{"id":"81b83bba888b363605b48f3a4fa15688deb2141f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-19 09:41:17.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8fbceab0a277f69899df4731bafd2af321f7ca71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 10:21:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f9c2e96168d472792dbf5a3d9518413\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53008e70688b41888d031e7fcd96cc64 : SUCCESS in 37m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f2f7eb5d6ac4b6583b5a7b65bf1d4da : SUCCESS in 16m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7630dbed932b3962af6e911d3b39b3cb935f1cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 11:45:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9a9c4ce13d84335a10e3eebb7cd5f8e\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/0b4662830c9343868d6b31afc1ccd140 : SUCCESS in 1h 49m 32s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/c38fadeae6744d36ba76992303e04111 : SUCCESS in 36m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96afe693b90648d591f44b2069c24a1a : SUCCESS in 48m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de2d608383a4014992afd1270095049 : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e3a25e5f7f4d1eb57c1c0db6bf7d89 : SUCCESS in 26m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bc206af3a27405cb6b4f7e8a3638354 : SUCCESS in 26m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd8e964de1a4342b543663d6669ccc0 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01f825986a9c435ab415fffacfbc3c58 : SUCCESS in 12m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/43a0f24cbd5b46fc819f2ade920e49cd : SUCCESS in 20m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1614f236da3e4a1a974e3425d264c58e : FAILURE in 49m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd13e81966ae4536a97a875d1d067b08 : SUCCESS in 1h 46m 05s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2524f5f51d904d74b076978fe21f0760 : SUCCESS in 1h 04m 23s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/00a4f58dfba54bf8b0bce3404b531981 : SUCCESS in 54m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/665808def08e43e48414024af2910277 : SUCCESS in 1h 58m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5226d5047c3c4a41be82cd31af84dffa : SUCCESS in 1h 34m 36s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8616b7062d2b4272a84ab1129bf6d825 : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":6},{"id":"dfe634f51516d7b956629abd2de1d6dc650a8ff0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-19 11:50:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8e13ab089ae7a6bef8c524bb1458a89844c228d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 12:07:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6000906c028d4b4a9d9e9840033e25dc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbde3767397442ae9107aa52d489f15f : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ab231a6950d4db1aa29c6ece4aa0151 : SUCCESS in 16m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6241e76dc170b758893d5958d11a2a081e3f0db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 13:40:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/754efb74f3604754b5d763d42aec23bc\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/1044c2f0a8e04c88843c003f859f20ce : SUCCESS in 1h 22m 04s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/6fe4996d149a4333a1aadfa749284742 : SUCCESS in 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e097af7f216c47b3828d7e40f4c4b925 : SUCCESS in 34m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f37e3e18b6414fad871a5a71d166db : SUCCESS in 13m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b13378950d5456884c24aa352a8dbf4 : SUCCESS in 23m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e872e19fcc9649fca43578f09655d418 : SUCCESS in 25m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a0ad6d34144cad9467f61c27f7f47b : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27dbd7052590412c9d9134510ee34c98 : SUCCESS in 22m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6f9012baddb74ccebddd9dcd9902275a : SUCCESS in 20m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/78e9ae59188841d5b91e669fb619fd57 : SUCCESS in 1h 03m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef69a4bf7ea9490ea89ea4c3daea3efe : SUCCESS in 1h 44m 36s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1abc062f945c4a3b88e12d4d38537f24 : SUCCESS in 1h 07m 13s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e879ce8e8280479aa5ae25a55504b359 : FAILURE in 54m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/af3f8fce93d8444191fec88b710b8de4 : SUCCESS in 1h 23m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ce01be5b3dc940b99d63ca3787996518 : SUCCESS in 1h 04m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/abd494a70b4b467ab5845328390f6c6e : SUCCESS in 51m 55s","accounts_in_message":[],"_revision_number":6},{"id":"6c7cb618b32a2c789713258b8a5381c47c454e30","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-19 13:42:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a37fe4b78efdb88ea4c5338a48d6975f67f046ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 14:05:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4860f5c0ac342388b5a983e8f630f5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56de4efac51442a2a1632327ca6d54ec : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6207354e0f2c418c8189dfb732fc272b : SUCCESS in 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c27c2c29453f63590ab2bccfdbb002878a524db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-19 15:10:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"41de8a497ff243d95aa7368b72da5e83d9e90bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:44:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbd16a707e14ae3aaec78d87e8cc6e3\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/88b9e7aeb5f04195a678b50305f583fd : SUCCESS in 1h 29m 31s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/ae4740f66d1e442ebdbb6454740e6bad : SUCCESS in 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d31eb360e3894e8cb8af2aea454b7e75 : SUCCESS in 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d41ea03098f4c3e9a9aa35f0a261552 : SUCCESS in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d9b4b8fc0e045f2aa46eed5224074a4 : SUCCESS in 24m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fcd0d160dcd45efafaaebfaada2ba7a : SUCCESS in 34m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058c4ae4d6bd487c815bddd21463438a : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fd953aa88e4a0889833bb31e54b6b4 : SUCCESS in 18m 32s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/44541bfc4abb4608a3d38819c09c7f4d : SUCCESS in 22m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84cceddbffb942b080a07a026a064d19 : SUCCESS in 55m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c9e82766dba4147a0234f541e4dd2b6 : SUCCESS in 1h 50m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b0d2e5536b3a4ab08b396ca133aeb51d : SUCCESS in 1h 19m 08s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/538e180f90f841778baba5a9299f3fc2 : SUCCESS in 57m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5087a517f3c642adb717f4f6bd0f59de : SUCCESS in 1h 16m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20bf8eac83cd4fc087341addb29902bd : SUCCESS in 1h 06m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c0f0b46ded974b93b2ead80c997cdeab : SUCCESS in 58m 19s","accounts_in_message":[],"_revision_number":6},{"id":"7c35b6b4e673a4f76c251df054f14ec09814988c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:52:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51676aa87c504cd7b2b97536962ee7a1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/417f9728e16d46b3a770cd2eed28fc14 : SUCCESS in 39m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc989b4db3964bfaab3274e87f4bb031 : SUCCESS in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e6e1d5992dc392551e79f4d9768e3d7f435960e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-20 08:05:01.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"fdbc8313eefbb07ead283de496bb16b6589f212b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-05 22:16:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"17f0858e7c292ac1ceec6d4cf8143574197c4d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 22:17:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4a37382145198ce0f5476ba69232f459f9aa4ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 23:59:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdbfc882ee074ab2b515a944be89d906\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/8aa3a44b878d48bf815e85c3397f2256 : SUCCESS in 1h 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb53a24d83b740bb855ead041931ec11 : SUCCESS in 11m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8ec9629ee8a44878e75eeaf0e605f8a : SUCCESS in 17m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0166d21695945ffbfe2d323dc2bff29 : SUCCESS in 19m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec7b7b6c4d3412b9cac29e3e8bb05b0 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a505d3bfbdc4e2b9dfbf83046f2a96b : SUCCESS in 11m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f019fa862eec41c2b832c654dabe4a69 : SUCCESS in 39m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a30a26303c5e4ed8b4a80311d894a7b6 : SUCCESS in 1h 28m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/498beb0e27394fbbb7a715ccae73adc1 : SUCCESS in 44m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a8d7e9225e1944c6a97bd79e9e6e4c6b : SUCCESS in 1h 09m 27s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fa34fd657f6047878cb14145e3008ffc : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":6},{"id":"507a8340fcc3f1b2bc0ea8095bfc541446b0ff5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 23:59:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1dbe4248bbac161b2fbdc3bcd8443873f9473ba9","revisions":{"fb6a069be384db3609ad8e790d07a8cfda3fcbf8":{"kind":"REWORK","_number":1,"created":"2023-12-12 14:48:22.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/1"}}},"commit":{"parents":[{"commit":"1460352432fe79545528461297a810f19087a643","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1460352432fe79545528461297a810f19087a643"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-12 14:44:54.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/common/ovn/test_utils.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nNOTE this backport is a heavily modified to avoid the backport of a\nprevious patch that also changed the sync_ha_chassis_group() method and\nbumped the required version of ovsdbapp. This backport does not require\nthe newer version of ovsdbapp.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/872023\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb6a069be384db3609ad8e790d07a8cfda3fcbf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb6a069be384db3609ad8e790d07a8cfda3fcbf8"}]},"branch":"refs/heads/stable/2023.1"},"f49e4ae1ec62cf933992877fb34376ec1b0a9526":{"kind":"REWORK","_number":2,"created":"2023-12-13 10:13:01.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/2"}}},"commit":{"parents":[{"commit":"98a3eae816be6d33665d8cd48d359ae28a07a86a","subject":"Merge \"[OVN] Add a log message after the \"post_fork_initialize\" method\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98a3eae816be6d33665d8cd48d359ae28a07a86a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-13 10:12:58.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/common/ovn/test_utils.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nNOTE this backport is a heavily modified to avoid the backport of a\nprevious patch that also changed the sync_ha_chassis_group() method and\nbumped the required version of ovsdbapp. This backport does not require\nthe newer version of ovsdbapp.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/872023\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f49e4ae1ec62cf933992877fb34376ec1b0a9526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f49e4ae1ec62cf933992877fb34376ec1b0a9526"}]},"branch":"refs/heads/stable/2023.1"},"8a0f65c7835d2383e516c89f610b1ec7d7de8099":{"kind":"REWORK","_number":3,"created":"2023-12-14 09:31:41.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/3"}}},"commit":{"parents":[{"commit":"98a3eae816be6d33665d8cd48d359ae28a07a86a","subject":"Merge \"[OVN] Add a log message after the \"post_fork_initialize\" method\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98a3eae816be6d33665d8cd48d359ae28a07a86a"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-14 09:31:21.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/functional/common/ovn/test_utils.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nNOTE this backport is a heavily modified to avoid the backport of a\nprevious patch that also changed the sync_ha_chassis_group() method and\nbumped the required version of ovsdbapp. This backport does not require\nthe newer version of ovsdbapp.\n\n[0] https://review.opendev.org/c/openstack/neutron/+/872023\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a0f65c7835d2383e516c89f610b1ec7d7de8099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a0f65c7835d2383e516c89f610b1ec7d7de8099"}]},"branch":"refs/heads/stable/2023.1"},"85a72f5014281a87c52b639b7f6f1d8802c176ab":{"kind":"REWORK","_number":4,"created":"2023-12-18 15:10:58.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/4"}}},"commit":{"parents":[{"commit":"4fa0487c77ac6a1ce6266bf0b094c0525ed4750e","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4fa0487c77ac6a1ce6266bf0b094c0525ed4750e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-18 15:07:44.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85a72f5014281a87c52b639b7f6f1d8802c176ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85a72f5014281a87c52b639b7f6f1d8802c176ab"}]},"branch":"refs/heads/stable/2023.1"},"82a31daba22105737fe25b0a484d19cb9258b3c9":{"kind":"REWORK","_number":5,"created":"2023-12-18 15:44:48.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/5"}}},"commit":{"parents":[{"commit":"4fa0487c77ac6a1ce6266bf0b094c0525ed4750e","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4fa0487c77ac6a1ce6266bf0b094c0525ed4750e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-18 15:44:28.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82a31daba22105737fe25b0a484d19cb9258b3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82a31daba22105737fe25b0a484d19cb9258b3c9"}]},"branch":"refs/heads/stable/2023.1"},"1dbe4248bbac161b2fbdc3bcd8443873f9473ba9":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2023-12-19 09:41:17.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/10/903510/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/10/903510/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/10/903510/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/10/903510/6"}}},"commit":{"parents":[{"commit":"62051588314b66ba884e5df1cf2e6f39dfd50506","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62051588314b66ba884e5df1cf2e6f39dfd50506"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-19 09:40:49.000000000","tz":0},"subject":"Enhanced external port scheduling","message":"Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nConflicts:\n  neutron/common/ovn/utils.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 770914f98df9abda2b5b8ca2896ff05618ea0181)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dbe4248bbac161b2fbdc3bcd8443873f9473ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dbe4248bbac161b2fbdc3bcd8443873f9473ba9"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
