)]}'
{"id":"openstack%2Fneutron~903938","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I8f164c12db7c4c1aa54ba07feede3586e6a19381","project":"openstack/neutron","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2024-01-05 23:59:25.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-01-05 23:59:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f164c12db7c4c1aa54ba07feede3586e6a19381","subject":"Reintroduce agent bridge resync test","status":"MERGED","created":"2023-12-19 09:41:17.000000000","updated":"2024-01-05 23:59:25.000000000","submitted":"2024-01-05 23:59:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903938","meta_rev_id":"33599b2c8fddf782a6b1d36f2741c93309c11e3b","_number":903938,"virtual_id_number":903938,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 23:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-12-19 15:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-02 15:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-02 15:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-19 09:41:17.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2023-12-19 10:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-19 15:08:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-12-20 10:55:31.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-01-02 15:17:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"362cd381b3af97581099b1abfebfd3b238fb86d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-12-19 09:41:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"319299e6a739a02b35a49900d676298aeb5dc4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 10:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d542880fd88b4c19ba4eaf240f46c937\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2547f17d326547e4a1360859cd7ab0b4 : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102a704b2e8e46e59ca21d8b0229f129 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bff3e0b588754b09b5bbdcaec64249f6 : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"4c0419220a0075c4836943a44e8909f88e242ebb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-19 15:08:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06fcda7f2a801e849392ca647bd805dfc79bdb4e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-02 15:17:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9979d1c5deb57993aa6a9b5840ecdf47359e6110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 22:09:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"852084ddd1a08ab1a345013bfb06e5c8e37dc2e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 23:59:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5164fd7bf57f486b87e49ffd8ef615f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff2a3994458545d39f76931c6ecebb59 : SUCCESS in 11m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9085065d471f46f88ac30b82d7f69378 : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/509f0fce05fc4a5fa7fb6ebba9c164f2 : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"33599b2c8fddf782a6b1d36f2741c93309c11e3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 23:59:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d69d8688b88cbf4344f30a2bb5cf91a08bc562ab","revisions":{"d69d8688b88cbf4344f30a2bb5cf91a08bc562ab":{"kind":"REWORK","_number":1,"created":"2023-12-19 09:41:17.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/38/903938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/903938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/903938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/903938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/903938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/903938/1"}}},"commit":{"parents":[{"commit":"cc642431f0509c3ddf8d0755964d2317beb81be8","subject":"Fix the common/ovn functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc642431f0509c3ddf8d0755964d2317beb81be8"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-02-28 02:58:53.000000000","tz":-360},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-12-19 09:40:40.000000000","tz":0},"subject":"Reintroduce agent bridge resync test","message":"Reintroduce agent bridge resync test\n\nIt appears we have changed the test_agent_resync_on_non_existing_bridgetest\nmain purpose in change 709544[1]. The test is no longer testing on what is\ndescribing it will test, i.e. resync on bridge change. Testing resync on\na bridge change is still important. I feel like what the test is testing\nis already covered in _test_agent_events set of tests.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/709544\n\nChange-Id: I8f164c12db7c4c1aa54ba07feede3586e6a19381\n(cherry picked from commit 75e83602247360cf10b7b9aa4e53a0135b491cae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d69d8688b88cbf4344f30a2bb5cf91a08bc562ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d69d8688b88cbf4344f30a2bb5cf91a08bc562ab"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
