)]}'
{"id":"openstack%2Fneutron~904833","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I442587e7ef55917f4ea873e190bf8afbc0e911e1","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1961740","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-17 10:07:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I442587e7ef55917f4ea873e190bf8afbc0e911e1","subject":"Add sleep before checking if ovs port is in the namespace","status":"ABANDONED","created":"2024-01-05 11:31:34.000000000","updated":"2024-01-17 10:07:47.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1221e5da03f4e103bc53af85d165c0ad80dd0ffe","_number":904833,"virtual_id_number":904833,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-09 17:38:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-05 11:34:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-01-08 15:14:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2024-01-09 15:13:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-05 11:31:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-01-05 11:34:03.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-01-05 12:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-05 14:08:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-08 15:14:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9c49b424016b2bfb414290c30da9c2b527ef5dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-05 11:31:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f45af070259e63272349fdf8baf7fdc6f305afe","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-05 11:34:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fdf82ce4b0c8ea344f53f10232d3a431a93c7ab2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 12:14:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ee9a91613b4627a6814d6b9bd3ef87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7d8a0d294d64b6ca763df734b2d8e9a : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/319c60bc9a744353accfd22ca532da5f : SUCCESS in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1dd9042f12fe17a3247ce3d3488afb833b097e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 14:08:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87cc28c74ed4caeacb3ddcc3cea8623\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/dec1bb7ed11d401bbe54d6866961d498 : SUCCESS in 2h 30m 14s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/4a81df59aada407f80ff6e83b6eb76dd : SUCCESS in 2h 26m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/60679712ee6d47c89c6abbb59e6c0477 : SUCCESS in 1h 53m 07s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/7500cfbc6e2d48d2aa3267f626a6b7dc : SUCCESS in 2h 29m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a6ae8f9a3394706b3ed65788c77b057 : SUCCESS in 41m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2116f7bf384347b59bf8353d19e7c81c : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f75bddb447c245fe9736181dc661beff : SUCCESS in 29m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c14e6ebbe814e6f9e675fcda7e0d49d : SUCCESS in 25m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a305644a5cd948af8b14f1df348776ba : SUCCESS in 24m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e78c6ed953480593d28295f010de8a : SUCCESS in 9m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc28eb32529a40bb900ac26c9ea2e4b7 : SUCCESS in 57m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/233f245bd78c48b8873ae9cb6b248036 : SUCCESS in 1h 26m 45s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4b737247314e47bf8964b4e6dcae6989 : SUCCESS in 1h 07m 29s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9c7a22e30474ad3bb50f01cb80ff224 : SUCCESS in 1h 06m 41s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/458f18c29ac1472a908e64b3d7601392 : SUCCESS in 1h 05m 29s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2f25f6ff0a8341b9915db30f9def6226 : SUCCESS in 1h 11m 51s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/281f972b30834a4ebe2858c18a03d7eb : SUCCESS in 1h 18m 09s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45cf3aae9764494e83675eebec61ab81 : SUCCESS in 1h 38m 20s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a4c9367409d49fab88053b43e515273 : SUCCESS in 49m 08s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ba440fab60d44d0b8998d7871346522 : SUCCESS in 1h 06m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9c132bcbc9544a8ba6215b5798f91d44 : SUCCESS in 37m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bd623b672fbe428e932aadc0d766321c : SUCCESS in 1h 18m 16s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/0319715eb5a843238c2baacc298d5241 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8a27c2d7c8c47429a8440e6168e7b0821c95e4d6","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-08 15:14:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f29e20fad2e25632b8d3be4fe8fdc1077306ea85","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-09 14:48:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69bdfb33eef72172c01df57f0685e9bcf0d2629f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 15:12:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c06287a90dae4499a25223d6056d180e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd2134ff5ae74e20a135e117c131c9af : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2eb0e87e1724b4c9745c3a4fd3fdbeb : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebce34af491aa4287eda31b2b0e6a44c53f67471","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-09 15:13:34.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"5ec3856eba8da21fce3c7654e192f86a5dbd3322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 17:38:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/292a2e773e61424da6862d08c2b43b4f\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/1cd091c251fa479885a18d278f944844 : SUCCESS in 2h 21m 31s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/f62839d7b8a14aa8b8956d0eb790689f : SUCCESS in 1h 48m 57s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/a70fcb82e995420388898b583470f814 : SUCCESS in 1h 33m 44s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/58520becea0f497798a386752ff17e23 : SUCCESS in 1h 37m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74bcd03d84c54f548a797c905b14c151 : SUCCESS in 40m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f28638656f49e39aa3386dbd68058f : SUCCESS in 14m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa6cb0b6cce49f7823c407a5a894790 : SUCCESS in 28m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccf436906dde4c1ca3df92b81535cfcc : SUCCESS in 22m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a3bf9e835554e9f80fcfb8cc7a16a59 : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80a048a58024bab85408582f705ee3a : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42a30c14536f45aabf8fc26547110aa4 : SUCCESS in 54m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3975765dda7c4d70a2338441b6e57662 : TIMED_OUT in 2h 33m 06s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/693856c1650445fbbb85925be5322823 : SUCCESS in 1h 24m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ffdeb0f2ce6447e8cb5b8edc4dfde3d : SUCCESS in 1h 13m 10s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3cf80420e5aa4259bec8515975e12c5f : SUCCESS in 1h 23m 45s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/27f62ca5b29a4cb2a845f2160cd6f579 : SUCCESS in 1h 05m 54s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/904311455692487c9b7f904c0662136b : SUCCESS in 1h 08m 38s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/04b3296e8def4de5a7fcc0457374bad1 : SUCCESS in 1h 24m 35s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ad61f2323fa4a89bb049aa5d494062a : SUCCESS in 1h 01m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4326a3e5a0574e659324bfc513d96597 : SUCCESS in 1h 11m 56s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/60e9060b94f64fa5bb942ef290f8b2a6 : SUCCESS in 34m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c544079a4b4f45a6aa2e60bf9718dc12 : SUCCESS in 1h 31m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/b5e26d44cd3542e0b2973d6e360b9f17 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1221e5da03f4e103bc53af85d165c0ad80dd0ffe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-17 10:07:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d78b3a1b5f196190aa8141e36bdca78fccc038b5","revisions":{"d78b3a1b5f196190aa8141e36bdca78fccc038b5":{"kind":"REWORK","_number":1,"created":"2024-01-05 11:31:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/33/904833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/904833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/904833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/904833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/904833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/904833/1"}}},"commit":{"parents":[{"commit":"6329c41447b324f8b016c912da7b4a5cb217257c","subject":"Merge \"Metadata: handle process exceptions\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6329c41447b324f8b016c912da7b4a5cb217257c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-02-28 17:27:29.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-01-15 11:30:40.000000000","tz":0},"subject":"Add sleep before checking if ovs port is in the namespace","message":"Add sleep before checking if ovs port is in the namespace\n\nWhen network device which is ovs internal port is moved to the namespace\nit may happend sometimes that it will have \"shy port syndrome\" [1].\nEven though there is wait for device to be in namespace in the set_netns\nmethod it may happend that device is in namespace during this check but\nit dissapears for short time later and that causes failures e.g. in\nfunctional tests like described in [2].\nTo avoid that, this patch proposed simple (and ugly) sleep for 1 second\nbefore checking if port really exists in the namespace. If it will be\n\"shy\" port it should already flap during that 1 second.\n\n[1] https://bugs.launchpad.net/neutron/+bug/1618987\n[2] https://bugs.launchpad.net/neutron/+bug/1961740\n\nRelated-Bug: #1961740\nRelated-Bug: #1998337\nChange-Id: I442587e7ef55917f4ea873e190bf8afbc0e911e1\n(cherry picked from commit 2af5fd889b3286dcec21e2bc89f287a0e4129d0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d78b3a1b5f196190aa8141e36bdca78fccc038b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d78b3a1b5f196190aa8141e36bdca78fccc038b5"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
