)]}'
{"id":"openstack%2Fneutron~904872","triplet_id":"openstack%2Fneutron~master~Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a","project":"openstack/neutron","branch":"master","topic":"super_rpc_2","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-17 18:35:00.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-17 16:33:06.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-01-17 18:35:00.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2024-01-17 13:55:56.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a","subject":"Make get_ports RPC method common for the DHCP and Metadata agent","status":"MERGED","created":"2024-01-05 18:53:54.000000000","updated":"2024-01-17 18:36:41.000000000","submitted":"2024-01-17 18:35:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"904872-super_rpc_2","meta_rev_id":"79c3880a6c7f6f137897dfbf5f1422e6afba9351","_number":904872,"virtual_id_number":904872,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-16 21:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 18:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 16:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 13:55:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-17 16:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-05 18:55:25.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-05 18:55:25.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2024-01-05 19:27:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-05 19:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-05 21:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 18:57:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-01-17 16:33:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"8dbcffdf2f693b925be5c6440ac3c13d0de845d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-05 18:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7275ba046bdfe1a5ff79009632f6a402cd64f6af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-05 19:27:30.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f1aebfd1f5afd659352f6c87b245398818ccd8e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 19:33:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb17ff7a21654c18afd6d8f7a51b000a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d6a87bc0d924d1fa1305fc744a6cb3d : FAILURE in 19m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e93cc328c9144cf7ae42b696c6db81fd : FAILURE in 37m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5fecc5d6a21449ffb1e26ebbc314127e : FAILURE in 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a463ea7a0fb51c6a85d6c2943546cc8fb027fc54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 21:27:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17e2ac659a6142a6a5ad3efa37afc18b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/99fc38734d044530a475d12f736a4f99 : SUCCESS in 1h 28m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4133b36ef3f5470eba0a521e9e480f07 : SUCCESS in 1h 30m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/4ac41477850147b2b5953493b5e1e638 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/373b1b4eeaf64383b46f9f2b1045ffe1 : SUCCESS in 1h 50m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/078b923c3b4843cab78be018a483a32e : FAILURE in 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d039906a764ddab968a22f2550e6b9 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b88a1f38d784f0cb16b8ded2ed8b9ea : FAILURE in 25m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a710a485edb45ec8c9d9c96aa347fd5 : FAILURE in 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070cbb0d290440388831b851c2d3fad4 : SUCCESS in 11m 25s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7c5abc88ff4846ce94472ddfaa4e455d : FAILURE in 26m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/eb4614882cb64b25b3a583b2f3280a44 : SUCCESS in 59m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/7b66f978a30d497c8619ebbbbb3e006b : SUCCESS in 1h 14m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71f182a0283e456a9f7c3f840365a588 : SUCCESS in 46m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aad3305f6535452a8af91946e781b163 : SUCCESS in 1h 50m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9f9a0146f8c41979849ca3a66289069 : SUCCESS in 43m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/80330a411a544930845c30158da1aab0 : SUCCESS in 38m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d55f9a3a5de4f45b675924810ddc28f : SUCCESS in 1h 28m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c72b1e8d26542ddb142e4e357133987 : SUCCESS in 1h 19m 40s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fedaefd8241f45f6bc2abdfff1c46f52 : SUCCESS in 38m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/636787db39df44d2a26f7c43fc263c82 : SUCCESS in 2h 09m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7d456c4508594633abe6ef7bfd0f7279 : SUCCESS in 1h 37m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/051abe126b114c16830e508c5c72204f : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7991e1da52d7acbd6c27dae21274e94ed10e520d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-08 19:16:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"885cf2a5b9b38c8f77c41959f713efbe90ed7051","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-10 18:54:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63b4d5ec9ffb0e67435a8bff5f3de0f7526dff77","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-11 13:53:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd9aa5a3ce6c3d09b6566ae1ac10b5eb4055f0f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-11 18:26:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4483c79d871fe2cd93bb47476f4e3d03e2395504","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-11 18:57:44.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3ad42178b5ccf1132e44c1626bb2e8c00fb7df85","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-11 18:58:06.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c64ba1b0e07800f0d23457ea05428442f88de50c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 19:15:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/009322a065ae4ae984df451b292db509\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7d6cca0780646f589d0d78a4094ae38 : SUCCESS in 34m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ff305d68bb9f4c0b8699d903eda7c690 : SUCCESS in 45m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38a7e984271744ebb752eddc094b78ea : SUCCESS in 40m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c15e508b8764fe7d02aee1fc6df7519fe938e34a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 20:40:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5b14b7f17847e8a174f50cbd195ca2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2d20471838424807b540a2435c577979 : SUCCESS in 1h 20m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/22b7762fc57545fc94ea42efe8ac6aba : SUCCESS in 1h 24m 38s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/554d00be3de84504b65a22a60a6cce7e : SUCCESS in 1h 42m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8ad5bde6d6924bbe86ea89482a294c9a : SUCCESS in 1h 22m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/005b424cc4104a4b961a117cfb397f23 : SUCCESS in 37m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfebe262f7244cf83f138bee106e4d5 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01148102206e486db296fa8b4811da3f : SUCCESS in 32m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f212142f2c948288880cc0bbcec12a5 : SUCCESS in 28m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd19d53798241a9bd818ad1419b061f : SUCCESS in 11m 21s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b12be2162c6c4ad4b36680e04255d7fd : SUCCESS in 29m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/dcc88990bc37443781d9f76bc4a7bb29 : SUCCESS in 1h 03m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2ea6506df15d434dac1b932adc99c2e1 : SUCCESS in 1h 49m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd5eea317c2b480dabb858750c54f5f7 : SUCCESS in 1h 01m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ee62a43203a4e21bc6e519b4d53d1d7 : SUCCESS in 1h 52m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23cb499cc1a1428ca0e4f5732cf6032c : SUCCESS in 54m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/635770af3718456c870ec8a49ecc0fc9 : SUCCESS in 1h 02m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7e93090213484927bcaa846866dae826 : SUCCESS in 1h 34m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1fe8f225a9b547c9b76aa35d3162d51e : SUCCESS in 1h 28m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f8bb2b17372b4755a0acee49c69e7cf5 : SUCCESS in 52m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/48a85da2d632452caee5730a1f9c977f : SUCCESS in 1h 34m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/26ff6df1699b477e90560dc3ba31f6c4 : SUCCESS in 1h 17m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e3d5e7bd534e4e5ca5cda1bea94c26f1 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":2},{"id":"70966bfd233b11bce5ecda812c89d03868121ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-11 22:00:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"76094cc105f46701a29c202be2bb80d62f3fc1ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 22:36:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc1d16cef78456eaaed552c673cac95\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2ee3d6083df499cbe8b56f391f5da8a : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2179ade659e2433b8004ce120e769efb : SUCCESS in 33m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f92d5d99349544ef97960d9059b9f303 : SUCCESS in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8090091931b087d1c8816a951ef88171fa48f07d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-11 22:58:11.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ea4e48f6dc95b0db4876d15b728f6a33c48a32fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 00:09:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f27cadafc964c7ba333f97de94aa19c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a23290975784ce3a787e4f8960460f7 : SUCCESS in 1h 40m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3fda2d9feceb41d4acc93a11cec052a7 : SUCCESS in 1h 42m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8cf622a2264b4945879ec0d9ddc34571 : SUCCESS in 1h 48m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7b3ae2971c8f45e7b7601a2a4f6d2f14 : SUCCESS in 1h 51m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb2110ee76464d6a9232e46480332b47 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1933cc89a40418c990650c762b8c252 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d153b4dbc3c4e6685a9f4158c2aa03a : SUCCESS in 23m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/681e09c069c24468a57e2026cca37d50 : SUCCESS in 23m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1ae2d686a44ababee8b017baa34be0 : SUCCESS in 12m 41s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/36d7f29e94404affabb57a75b243d1d5 : SUCCESS in 29m 46s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b2b65e69759e4270b4a5775840261360 : SUCCESS in 53m 03s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3977626900bd48bbaee8a63c0934110c : SUCCESS in 1h 05m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f2fa9cb2f8c4a86b9d102a4ca3e7f22 : SUCCESS in 45m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e20ad689b0b94ba78119d5745a92f075 : SUCCESS in 1h 26m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a2824788c8f488e925ea6817e35bf8f : SUCCESS in 1h 03m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ccc82adbcdc4bb5963febba3428be4c : SUCCESS in 59m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dbc039fe29e646e2823c33bc8ea0adf1 : SUCCESS in 1h 30m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39d274ff67b24981bd743227a57cc0d5 : SUCCESS in 1h 13m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53d9259fa50a476c853b3b797662358f : SUCCESS in 1h 10m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/739cc02e296e45909d5da58bf8ed6da1 : SUCCESS in 2h 02m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8bcfd1c733e5453cbdb7abcd9c28df85 : SUCCESS in 1h 55m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d901c02c06af4db182e71b563fe7e907 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":3},{"id":"75bf44402efecc839bcfa778488924d90d5485f5","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-15 02:28:46.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2b8407c96331ad3e1c39577f7787ec637cd0fca6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-15 10:39:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"737e4fb8a6a786b49dfef3c261f4ab139b1d2626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-15 21:28:36.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2407b0d7ba55889a3cc19df92d4d29fdb12794df","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-15 21:29:16.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6abaa8fa2b361126adc3418b8d5f922749f27cda","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-15 22:02:41.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"0a3b8342c66d987680f5c1e13c403353ea978621","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 22:11:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e165001328543e0908d8de53e25ecaa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3fb68919bcb47fcbe74e5409acd6dd3 : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea9e091f48ba4263b125a6af7d60e600 : SUCCESS in 40m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7da0dc679cde4c74a72f2b053215fae7 : SUCCESS in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"661de5d3631a54d81f83f01490a8de8e7db1f9b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 23:55:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9f41e0a681946f3b55d072a0d179e93\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9fc0cb6ec7434efbad3703d765161602 : SUCCESS in 1h 20m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0eaa4a53369944b384edaf3f938c4d4c : SUCCESS in 2h 18m 29s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b8641f7023584feab0036991632ecc8b : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/09fcb3d9437f4e2da8b2ffca672c97f9 : SUCCESS in 1h 55m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d49156fc0a47402d89b24cbb0a4bad63 : SUCCESS in 38m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ccb91f14a449359b0728bbe8e7f37d : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27e84e645cee4b6c82b40fc773be0df5 : SUCCESS in 27m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ebabb41bee848f6b1313f7eb0522c94 : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b00c6f49e144a7ea8c299186a0653d8 : SUCCESS in 11m 06s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6a259ca546ef4ad498b892e35e5bb3b1 : SUCCESS in 33m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0118b784d4624c6face4ab1cb33e2eaa : SUCCESS in 58m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a8db273f80314aa5a20e4906faed4a55 : SUCCESS in 1h 14m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54993c172dd540698fd5ae5c96483a81 : SUCCESS in 1h 02m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/acbc3b78abd8480cbe5de2df3c6e301f : SUCCESS in 1h 42m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97bb8b263acb4496b37cbe24414c4575 : SUCCESS in 1h 02m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5785e300e3704da3ad3cf550563c18fd : SUCCESS in 1h 03m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be960d04f1354adc8128a6dbb64f32e0 : SUCCESS in 1h 31m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ec4914e031143648f9a01af5a3f433c : SUCCESS in 1h 18m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cbae5e45562547b4a4ff7d3868726b39 : SUCCESS in 1h 05m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8a9890da44fc4165bc143eac299051c3 : SUCCESS in 1h 15m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fdde6bc0e465472e9f6a0f38cf1ae0e3 : SUCCESS in 2h 10m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8613aa6a31f04b15a86acca299af8b55 : SUCCESS in 1h 21m 37s","accounts_in_message":[],"_revision_number":4},{"id":"5a0dbbdbb262f70faf76824fdb570539396d878e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-16 14:22:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f585c3038b2638f282f73e7940b3abfd48952172","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-01-16 14:49:29.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1a34983832d9e81592d6155eebf34864173348a7","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-01-16 14:51:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b29d2148f4b88959ad2024773980da4596b2034f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-16 20:18:27.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"38df4a47b9db7a2f4cec90df737faa8398722f9f","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-16 20:18:46.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4d585de02fa62121125b8d00bad155bc4d235112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-16 20:30:38.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"22743454e53c82fbf36def4087fdc855486390ec","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-16 20:31:16.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf7879584a4a74c4df3fb6d8687f82a1b407a0bb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-16 20:53:43.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"ca1895523214251749a2c21f3366727d4f5f0923","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 20:57:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14dba4b43e3428d8218522b80595817\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41031a21431f45f4b8a3a2dedbc225ea : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03123dbf6de8471daccb062979b1fc6e : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e46e4581b26c4decb1d6541cebb4afa4 : SUCCESS in 17m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bfe2d109c709e3d8fa51f37d45f604043ac8e8e0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-16 21:27:48.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/72/904872/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"98a84f5d0d5d5d50ef97496f8fe51acc4d366cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 22:35:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efe2ad57686e417f8e84b07aa7c5be5d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4bd7fea6ed6145d8a32e817d281a787e : SUCCESS in 1h 23m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/21fb700d0eab4b36bf34a41212eeabfa : SUCCESS in 2h 02m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/02dc2c18294e4f4a91d300970a8dd3bd : SUCCESS in 1h 19m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f4aa521f8d94f158a89982b09dbeb65 : SUCCESS in 1h 25m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52decb9ade63432bb3e3505a1a1b9473 : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de2d379a8ea4d61bde2b997971a09bb : SUCCESS in 11m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65a983ffaf8943768aab38be62dd7fc6 : SUCCESS in 26m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/678a3ef775b2402092f1971152c73745 : SUCCESS in 24m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88d9579b1fe41f2b250aa3d8a6d2649 : SUCCESS in 10m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8d924c37369549aa96d27ffded5f6d4b : SUCCESS in 26m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d284c6f038ab4cd6b190aa937de1a9e7 : SUCCESS in 51m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1cfe0d72f11348c48cb2e75be6ccc563 : SUCCESS in 1h 14m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96c4e6a900ee4ff0a84ca0d392fd1165 : SUCCESS in 51m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/600c3b59bf9842cd86132c3cebbb451a : SUCCESS in 1h 22m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28aa168e8c524486bd4f204fc8f41774 : SUCCESS in 1h 03m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/959d830bd0894fc9a81e0d07686024cd : SUCCESS in 1h 06m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ebcd55cbd7e741a2a1d5935801d24c88 : SUCCESS in 1h 24m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f2b3a378346b487bacbd6b67af9039dc : SUCCESS in 1h 24m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cdbe682b23ae488983cb91f456d2a2c4 : SUCCESS in 1h 02m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d700c906ccf84adea91e05d5b12981fa : SUCCESS in 1h 33m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9f85fd7bb3d04ad1a4deb6ad7aa071b5 : SUCCESS in 1h 37m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cccbdb2e966c459f850c7c1121915543 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e40f98a2ddb5e59065a3967ac2dad57863b9f906","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-01-17 13:55:56.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5eee90d7705fdc09801cd4c56a9f2e602d5d9576","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-17 16:33:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"39de70d0b8f18ae94c98c975773ba5f2d646dd66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 16:33:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e851bab962cd7bac0cc3065776bb6f75c688435e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:34:59.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6314d91f74c4c25ac8af1daa8c14f7b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1e44e54613064ad3974b4cdb9a062c10 : SUCCESS in 1h 21m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72f12a908f544c2eaf16ff6183d8f5ec : SUCCESS in 1h 48m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/677194c9b0594b57bd21b6f1babc668c : SUCCESS in 1h 23m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/727198d56e094e80b336769877589daa : SUCCESS in 1h 30m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f267704fe14340a862600d47442fea : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/506a820fd3b24e6cb01bc1603011336c : SUCCESS in 25m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faeb28d92a17474d8c59647881daee26 : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2bae49b80547078b6e253855054ba5 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7151c8a27f744410b9dc4056b7344fbe : SUCCESS in 42m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/354dd4b155fe46e99f5d77fe70e1be90 : SUCCESS in 1h 38m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea8370f3493f48df975671d700c0041b : SUCCESS in 1h 31m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8b4a60bd6c14a078ef9b1643448114f : SUCCESS in 1h 06m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7e4a82281dfb41b5b8432e254c8ca5d2 : SUCCESS in 1h 10m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c3fc7dc5147a4d05aa90cbdb58e25352 : SUCCESS in 1h 04m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/987869daa3424125870ec89bd222c883 : SUCCESS in 1h 28m 50s","accounts_in_message":[],"_revision_number":6},{"id":"02b62df7bc0a09b0edd3672e845fe6be0673c4f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:35:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"79c3880a6c7f6f137897dfbf5f1422e6afba9351","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:36:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64d72fbd458d4d0680e2ced838080b3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0767177aa7c424da204b5e8c6dfd522 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"637e7a50077f215dbaa25ebde6f6ca792ef4955b","revisions":{"88cd8152ccb04c03d3e7e702734a76ee4cca7f2a":{"kind":"REWORK","_number":1,"created":"2024-01-05 18:53:54.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/1"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:53:37.000000000","tz":-300},"subject":"[WIP] Make get_ports RPC method conmon for the DHCP and Metadata agent This WIP patch is the initial implementation on the suggestion from this patch[1]","message":"[WIP] Make get_ports RPC method conmon for the DHCP and Metadata agent\nThis WIP patch is the initial implementation on the suggestion\nfrom this patch[1]\n\nTODO:\nunit tests\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is not change needed to the API, however it does\ngo against how we historically setup the RPC layer between server and client.\n\nTODO:\nunit tests\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88cd8152ccb04c03d3e7e702734a76ee4cca7f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/88cd8152ccb04c03d3e7e702734a76ee4cca7f2a"}]},"branch":"refs/heads/master"},"289b947ca9fbb3131c0d1d61ae9570b593358546":{"kind":"REWORK","_number":2,"created":"2024-01-11 18:26:58.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/2"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-11 18:23:03.000000000","tz":-300},"subject":"Make get_ports RPC method conmon for the DHCP and Metadata agent","message":"Make get_ports RPC method conmon for the DHCP and Metadata agent\n\nThis patch is the initial implementation on the suggestion\nfrom this patch[1].\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is not change needed to the API, however it does\ngo against how we historically setup the RPC layer between server and client.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/289b947ca9fbb3131c0d1d61ae9570b593358546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/289b947ca9fbb3131c0d1d61ae9570b593358546"}]},"branch":"refs/heads/master"},"0239cfdeca85c437b7b8cae78693e9ecc8cb0f59":{"kind":"REWORK","_number":3,"created":"2024-01-11 22:00:28.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/3"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-11 21:59:15.000000000","tz":-300},"subject":"Make get_ports RPC method common for the DHCP and Metadata agent","message":"Make get_ports RPC method common for the DHCP and Metadata agent\n\nThis patch is the initial implementation on the suggestion\nfrom this patch[1].\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is no change needed to the API, however it does\ngo against how we historically setup the RPC layer between a server and client.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0239cfdeca85c437b7b8cae78693e9ecc8cb0f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0239cfdeca85c437b7b8cae78693e9ecc8cb0f59"}]},"branch":"refs/heads/master"},"1d42872a149ea1b518c739dac404f0df6afdb84d":{"kind":"REWORK","_number":4,"created":"2024-01-15 21:28:36.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/4"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-15 21:28:26.000000000","tz":-300},"subject":"Make get_ports RPC method common for the DHCP and Metadata agent","message":"Make get_ports RPC method common for the DHCP and Metadata agent\n\nThis patch is the initial implementation on the suggestion\nfrom this patch[1].\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is no change needed to the API, however it does\ngo against how we historically setup the RPC layer between a server and client.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d42872a149ea1b518c739dac404f0df6afdb84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d42872a149ea1b518c739dac404f0df6afdb84d"}]},"branch":"refs/heads/master"},"198a3d1d0b7dfd0fee45841b22d005de15bca9d7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-01-16 20:18:27.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/5"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-16 20:18:27.000000000","tz":0},"subject":"Make get_ports RPC method common for the DHCP and Metadata agent","message":"Make get_ports RPC method common for the DHCP and Metadata agent\n\nThis patch is the initial implementation on the suggestion\nfrom this patch[1].\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is no change needed to the API, however it does\ngo against how we historically setup the RPC layer between a server and client.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nRelated-Bug: #1982569\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/198a3d1d0b7dfd0fee45841b22d005de15bca9d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/198a3d1d0b7dfd0fee45841b22d005de15bca9d7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"637e7a50077f215dbaa25ebde6f6ca792ef4955b":{"kind":"NO_CHANGE","_number":6,"created":"2024-01-16 20:30:38.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/72/904872/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/904872/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/904872/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/904872/6"}}},"commit":{"parents":[{"commit":"0c251cce604477e07f145cb78980e31e6b9a787c","subject":"Merge \"Handle creation of Port_Binding with chassis set\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0c251cce604477e07f145cb78980e31e6b9a787c"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-05 18:42:12.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-16 20:30:24.000000000","tz":-300},"subject":"Make get_ports RPC method common for the DHCP and Metadata agent","message":"Make get_ports RPC method common for the DHCP and Metadata agent\n\nThis patch is the initial implementation on the suggestion\nfrom this patch[1].\n\nThe DHCP agent can query the existing `get_ports` RPC method because\nthis method is already exposed in the MetadataRpcCallback(server side)\nwhich runs under the same topic(PLUGIN) and namespace(None). The benefit\nhere is that there is no change needed to the API, however it does\ngo against how we historically setup the RPC layer between a server and client.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/903572/comments/3d4e0453_4b4d2ab6\n\nRelated-Bug: #1982569\nChange-Id: Icd7c55d2a5103bdbd90907b1dbfb9ccfe34c020a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/637e7a50077f215dbaa25ebde6f6ca792ef4955b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/637e7a50077f215dbaa25ebde6f6ca792ef4955b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
