)]}'
{"id":"openstack%2Fneutron~905417","triplet_id":"openstack%2Fneutron~master~Iaa1aa0022e207cceb26ed45474bc6edfbbec4faa","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-17 08:30:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa1aa0022e207cceb26ed45474bc6edfbbec4faa","subject":"Cleanup setup.py and requirements","status":"MERGED","created":"2024-01-12 08:45:40.000000000","updated":"2024-01-17 08:32:03.000000000","submitted":"2024-01-17 08:30:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905417","meta_rev_id":"1a77dfd80c6095a7043855a76186f692dd545b0d","_number":905417,"virtual_id_number":905417,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-17 08:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2024-01-17 06:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-01-12 15:32:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-01-17 06:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 09:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-12 11:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-12 15:32:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-17 06:09:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"a5c5133897a7d1578acdfa5fa051d38f508e3014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-12 08:45:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de15a12ce35e872e3dd870917991f3fa98b379b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 09:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59cc30bac097409a9dbf294ae450afab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f42c95bff3224d2c9029ba73b6e0560b : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/944c0c49f6e04c3f9e83250eda8d94ce : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/93ca84c4f7b9482a96bfa7856d4c5622 : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74f78414d1823bdb3902c10a103ae0504d423723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 11:02:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c366f3a8b844cd382eb2b74e4d51be3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63b4ec894bee4c40a77b1422a84bdbcb : SUCCESS in 3m 08s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1dd6dc5847294632b337c2abf59a4063 : SUCCESS in 1h 35m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e36b123b624e40e1a6710d2051eac464 : SUCCESS in 2h 14m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3331d4c743c04b3bbfe747031330f716 : SUCCESS in 1h 40m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/78d47e115c904bea806de9b58e892151 : SUCCESS in 1h 55m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a7ca6c291d5437ea0b6ed6691850320 : SUCCESS in 31m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0745bbfaa03b4d57afc6d01edaa62a5a : SUCCESS in 31m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda3da7f669d479293bf9a10c84449c2 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1ec6ec839864f1aa83851f39aba882b : SUCCESS in 26m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fae8e789798b49179551bc6b4fab2964 : SUCCESS in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad01a280dd314c58b83014e12cb1181d : SUCCESS in 11m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14d4f42cda844c87852b1a6b5c572151 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de02f7017e9a462d98ee363886de8a7d : SUCCESS in 10m 30s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c7282d7d64bf4615bdd664ea616d5a95 : SUCCESS in 28m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ce8a40d205e840f5b36c0337b7fb0f64 : SUCCESS in 1h 00m 53s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f5a1b51a84f4467c8081134a758c40f9 : SUCCESS in 57m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/48db67c872bf4ff4b33b240b04c8abf5 : SUCCESS in 49m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/de503545ca044cf0a9cc896550f41db7 : SUCCESS in 1h 39m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3daa3554f9a4f378e0d34f63d652177 : SUCCESS in 1h 01m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b447c608cecd4f04a18032f1917b44f2 : SUCCESS in 1h 02m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/55d801c607114be4a0a780608236af9c : SUCCESS in 1h 20m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8ea4d514cec490a8c6f51cc68ded097 : SUCCESS in 1h 26m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/08290255554448ddad5bd5cf13d1c257 : SUCCESS in 1h 14m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/16b9f41dea2146e2915e8982b3f8f9ff : SUCCESS in 2h 08m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/913d63869d3b4c1583680de376b8b280 : SUCCESS in 1h 57m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/49751b56f24b4aed88c836638e087d0f : SUCCESS in 1h 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eeeeda0e8a89d04ea2f2f164e0b6177925163c3e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-12 15:32:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c640c2c41b45b08c214f2b1184f76229f3ec2bd7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-01-17 06:09:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"113a4dd969bc6d18040dc29ef7584b1d03afa53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 06:10:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f627b6edfdb7fef0fc2e20635e01b6cfa31e354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 08:30:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81321f5a919400bafbac3d0785cdcf2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2916f0992061448ba3849555031380ad : SUCCESS in 2m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0892af9351f42f69c32bddaa1ad11b6 : SUCCESS in 1h 41m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a000b2190142432e8c0c623cb72136b1 : SUCCESS in 2h 19m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/267ca6b8bfdd4d79aa7154560b3867ac : SUCCESS in 2h 16m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/eb4980a699dc4dc2aaf7d8747df9ae6d : SUCCESS in 1h 42m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc39e380892049d2ad2be400e722c487 : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc64db0dd3234ecbab605beabf20831f : SUCCESS in 29m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3dffbb0ae444384944382066defb8d8 : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1161e717e9d147998b5180ddac834bea : SUCCESS in 10m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/efa9eba8e8df4caba5f35e397699f1af : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18cb6211b6a84eeba949f006a8a372f8 : SUCCESS in 14m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f32ca11a643242839a018bf0f93bedd3 : SUCCESS in 1h 00m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce2c7e7fd9ba421aaf8c24b842e6faf8 : SUCCESS in 1h 46m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f3e75436f36542f08889fde5acc92425 : SUCCESS in 1h 44m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23de06159b63489a9d9cea6108833e7e : SUCCESS in 1h 01m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0462f5617977435dbec58c2e33fbdcaf : SUCCESS in 1h 02m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/637145ca6c8c4455a2bb064650db5f1d : SUCCESS in 1h 09m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/df72a41d2e754b078fe77fefda708a4d : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"31df545d0e1d438a6d577aa87ddf119d347682d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 08:30:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a77dfd80c6095a7043855a76186f692dd545b0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 08:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42fda7cc9c5544a5929bec7a6fca15e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a38125d473411ba3970ecb799e6474 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7231f81fb43440a090b9ec764f56d3f7 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ac18a8ee1f15eb77ac15cedbb4b4765ad2241f0","revisions":{"8ac18a8ee1f15eb77ac15cedbb4b4765ad2241f0":{"kind":"REWORK","_number":1,"created":"2024-01-12 08:45:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/905417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/17/905417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/905417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/905417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/17/905417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/17/905417/1"}}},"commit":{"parents":[{"commit":"a89b5e57721118e16df0af0c3bd71197bb2a08df","subject":"Merge \"[FT] Add more information to ``IpRouteCommandTestCase`` if fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a89b5e57721118e16df0af0c3bd71197bb2a08df"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 08:44:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-12 08:44:09.000000000","tz":540},"subject":"Cleanup setup.py and requirements","message":"Cleanup setup.py and requirements\n\n- Python 2 is no longer supported\n- setup.py is no longer managed by the global tooling\n- Recent pip does not require appropriate order in requirement files\n\nChange-Id: Iaa1aa0022e207cceb26ed45474bc6edfbbec4faa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ac18a8ee1f15eb77ac15cedbb4b4765ad2241f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ac18a8ee1f15eb77ac15cedbb4b4765ad2241f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
