)]}'
{"id":"openstack%2Fneutron~905528","triplet_id":"openstack%2Fneutron~stable%2F2023.2~I7c43812fed5903f98a2e491076c24a8d926a59b4","project":"openstack/neutron","branch":"stable/2023.2","topic":"fix_netns_deletion-stable/2023.2","attention_set":{},"removed_from_attention_set":{"29074":{"account":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"last_update":"2024-01-16 17:39:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c43812fed5903f98a2e491076c24a8d926a59b4","subject":"fix netns deletion of broken namespaces","status":"MERGED","created":"2024-01-15 11:18:42.000000000","updated":"2024-01-16 17:40:30.000000000","submitted":"2024-01-16 17:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905528-fix_netns_deletion-stable/2023.2","cherry_pick_of_change":896251,"cherry_pick_of_patch_set":9,"meta_rev_id":"4f4fc96613e6373a04ef9f653b2c86fa76057f07","_number":905528,"virtual_id_number":905528,"owner":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 17:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-15 11:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-16 15:03:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-16 15:03:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 11:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-15 11:58:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-15 13:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-16 15:03:51.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"4dbd848f90cf1c8a5153fedb6dedc195c625de5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"date":"2024-01-15 11:18:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d8618cdae096110466c70dd458a5f60e1034ec9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 11:54:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3276894b68bb4bd7a40c9cad4f5edae5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c89a578ed59949b584780ef08117fa79 : SUCCESS in 34m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a685c7aba1ad4474b4350c27ecee9af0 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2efb221529540f1a59d1f24494d89b2 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7d3d6a23d5e4d6eb85bec2152f46716 : SUCCESS in 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9348140db844f713f6b1f4ce3d658659da42d2c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-15 11:58:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"793527f64e0a8ff4a00249454b92ca31ac264530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 13:45:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95e97930f2ba46e3b46fadd55c28787e\n\n- neutron-tempest-plugin-openvswitch-2023-2 https://zuul.opendev.org/t/openstack/build/6fedc72083794887aeaf42d403b57238 : SUCCESS in 2h 10m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-2 https://zuul.opendev.org/t/openstack/build/d731abceb1f24168be7a493921937f4e : SUCCESS in 2h 04m 14s\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/81d6f7690ddb43d58eb9dbe1206cb991 : SUCCESS in 2h 09m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed3f3efbfc224bc9934572c2c0cfc402 : SUCCESS in 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88929401aac4ecfbe584d6e0a6cb5e5 : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2807eb0117e348f38b879d783d5602e3 : SUCCESS in 23m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99dfa28b1ce843d4961e4e288964f0a4 : SUCCESS in 26m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9226c80abb8476abc4626ce327889e4 : SUCCESS in 23m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0894c667b454eb3902b0c4ab76fab0a : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a90a8a21f0c4d8091d1e326178db501 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcbacc5e84e8410fbfce8e1c09c48bcf : SUCCESS in 57m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dad6ed6b6a274f0d9d6b348d5462925f : SUCCESS in 2h 00m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4c3450cf3034c8ea7926efb68701908 : SUCCESS in 1h 03m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/866a25754c424932aba1d36cbca1dfd2 : SUCCESS in 1h 00m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd3ed2b0f1b0422b93f17e84bceb30a4 : SUCCESS in 1h 40m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7c1afafc57a149d191ed9bb0d368e9e9 : SUCCESS in 1h 21m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d6e745600fe94581a8aecbf9e905adf2 : SUCCESS in 1h 06m 09s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c683e81b49784e38848205621a9cb16a : SUCCESS in 2h 06m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dcc005a6b25d481b99dffa561da75b41 : SUCCESS in 1h 47m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b23eacb4a56e40bf9a7252ec1be988e9 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5c55ef52d5d2a3fb4948ce200d03b42b514e2e25","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-16 15:03:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"144c5465150ed60c234a05b1737c1f9b106e1260","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:04:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f770b4f8a300d55fc92d6264f7913068ef1337b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:39:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37fec13031724407b7e8bbe849538963\n\n- neutron-tempest-plugin-ovn-2023-2 https://zuul.opendev.org/t/openstack/build/75783aea47f8404ebc5e6e7e99a43d0a : SUCCESS in 2h 07m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916054e95b674b36ad503531c7d481b2 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4b3b2d9dec948d8b53cbd1797e106f8 : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc4965fccbc44f3a9b74633030cb1735 : SUCCESS in 31m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27daa62039224b07b10f94277807ccbf : SUCCESS in 33m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f685dcfdf3243069f59d214ce29063c : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d004987a1d24825b8bc07e9045ef105 : SUCCESS in 1h 02m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca14c03305294b04983cd9772cda5ef7 : SUCCESS in 2h 03m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ec2dbe3355514b46ad98449e363ad804 : SUCCESS in 1h 37m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31b7a0e0e77d47ed975045311d4de958 : SUCCESS in 1h 07m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e773e2f7dd746e4859d936e5443dcab : SUCCESS in 55m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/040211e7ed274cb385cf43e81b8cfa3d : SUCCESS in 1h 10m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52b31951d8cb4c0fb245e2635fdb7abe : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d5d4911e66abe586cd319999395a5ba6d5d9bf63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:39:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f4fc96613e6373a04ef9f653b2c86fa76057f07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2f91f03a69492ca9b7c5510a960ce5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fbbd7377a64bf68cdb9909bb2c2470 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f07cc43964e9dea7a48bf8564944d1e05f4e22a8","revisions":{"f07cc43964e9dea7a48bf8564944d1e05f4e22a8":{"kind":"REWORK","_number":1,"created":"2024-01-15 11:18:42.000000000","uploader":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"ref":"refs/changes/28/905528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/905528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/905528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/905528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/905528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/905528/1"}}},"commit":{"parents":[{"commit":"e9cf2fd6cca8a3d5c06bcb073cb310cd61208b41","subject":"Handle creation of Port_Binding with chassis set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e9cf2fd6cca8a3d5c06bcb073cb310cd61208b41"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-09-22 14:25:10.000000000","tz":120},"committer":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2024-01-15 11:18:42.000000000","tz":0},"subject":"fix netns deletion of broken namespaces","message":"fix netns deletion of broken namespaces\n\nnormal network namespaces are bind-mounted to files under\n/var/run/netns. If a process deleting a network namespace gets killed\nduring that operation there is the chance that the bind mount to the\nnetns has been removed, but the file under /var/run/netns still exists.\n\nWhen the neutron-ovn-metadata-agent tries to clean up such network\nnamespaces it first tires to validate that the network namespace is\nempty. For the cases described above this fails, as this network\nnamespace no longer really exists, but is just a stray file laying\naround.\n\nTo fix this we treat network namespaces where we get an `OSError` with\nerrno 22 (Invalid Argument) as empty. The calls to pyroute2 to delete\nthe namespace will then clean up the file.\n\nAdditionally we add a guard to teardown_datapath to continue even if\nthis fails. failing to remove a datapath is not critical and leaves in\nthe worst case a process and a network namespace running, however\npreviously it would have also prevented the creation of new datapaths\nwhich is critical for VM startup.\n\nCloses-Bug: #2037102\nChange-Id: I7c43812fed5903f98a2e491076c24a8d926a59b4\n(cherry picked from commit 566fea3fed837b0130023303c770aade391d3d61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f07cc43964e9dea7a48bf8564944d1e05f4e22a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f07cc43964e9dea7a48bf8564944d1e05f4e22a8"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
