)]}'
{"id":"openstack%2Fneutron~905680","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I6302e9a373cf93e706cec10f87c3beaf632a0391","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2036877","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-24 14:34:56.000000000","reason":"Change was abandoned"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-01-24 14:34:56.000000000","reason":"Change was abandoned"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-01-24 14:34:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6302e9a373cf93e706cec10f87c3beaf632a0391","subject":"Correctly validate subnet arguments when using a subnetpool","status":"ABANDONED","created":"2024-01-16 11:00:59.000000000","updated":"2024-01-24 14:34:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3cf7ff62c6ee526561ea4bc9e0033002a5a216e0","_number":905680,"virtual_id_number":905680,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-16 12:55:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 11:00:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2024-01-16 11:19:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-01-16 11:24:35.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-01-16 11:43:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-01-16 12:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94199af70c8f06afaece09fb25343782f0d08631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-16 11:00:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7eae4eba5a4cd38547d612409a3764f6ecf5cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 11:19:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/530fa4799e4d4289ae7c72c660c93c39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f51beef1e9c24afeac44d0791b5c2ec6 : SUCCESS in 16m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c119a47d715d4f06a362118ad56e8ff7 : SUCCESS in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc3b04fbae8a7e9786ef9f2a360caeb2ce160d83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:55:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4b4c0fd8c424c93ab4056db6e06c207\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/3110a14303474e9ab4763ce45e624812 : SUCCESS in 47m 22s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/2dbeef9503f8417a9c2f5b7f49f5ed45 : SUCCESS in 1h 33m 39s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/7818c613adb24269a012188b135eaed9 : SUCCESS in 1h 08m 04s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/a5ff3aec93d547ebbd0effac61a00451 : SUCCESS in 1h 17m 09s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/22bfa7d7f9124f66adb03619e01f11c2 : SUCCESS in 1h 07m 25s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/eaf2e254eb2a4a0bb52fb464b7690429 : SUCCESS in 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/393700b292904d44aaf2bef81e944a5a : SUCCESS in 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a529290a81b4cef95a84e59e54666b1 : SUCCESS in 13m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b2f4e155364ebbb9dcff8607e4bb63 : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252f03402d724a1fa7a8176e11205994 : SUCCESS in 27m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff61171b4fe341e78896f3c090bbb957 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd4f0e6d99dc486ba1274f037c523dbb : SUCCESS in 51m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/00f17ffa8fa44ffaae033f3fb0bc203c : FAILURE in 1h 37m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bcc7187c8a9c4a69a06f66e4fc2f7634 : SUCCESS in 1h 06m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b976536190dd474b9cd56e404434209e : SUCCESS in 51m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/946df617452b49e7accb4592e3884379 : SUCCESS in 57m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6dd20a7e0c5d4abe8fc00817d4c1fe20 : SUCCESS in 1h 09m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/732afa405d0340258ecd719e9f44f1f0 : SUCCESS in 1h 18m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6aaab01117c45a6ac506ceb7ae58683 : SUCCESS in 51m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4c000af96f864d6680551abfd0c8d1b7 : SUCCESS in 1h 06m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"171bb86d77db981993542edb9a2d61273b2751f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-16 14:52:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef1d93a408091ca3cee8e3fd4aa1675e2b038e49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:22:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aac570103aab4a23999e5ae100ea4ddd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffbd443e70134f66b8183eb77392a887 : SUCCESS in 19m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af08eb378cfd4642aecc601b4aef5ff7 : SUCCESS in 27m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f62256838935711284737f88f4e2c09507fd4453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 17:03:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efdea06a3630424e958db18684058eae\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/198fa1e3f22141eb9a0ac3437471fab0 : SUCCESS in 1h 04m 34s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/ea428156caa54468a5e43c62eb183363 : SUCCESS in 1h 32m 07s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/ab6e3e8dcdc1438b8c5cee81dbdecbb7 : SUCCESS in 1h 30m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/60d7465b05324757a76e6b2c8714d3f9 : SUCCESS in 1h 45m 02s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/6063d1e965984e94ba57345f049da3cf : SUCCESS in 1h 10m 44s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/100eb3d6208741b7b537520bf8bbba58 : SUCCESS in 43m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f4ff66ac68e44e795ca0136b695bbe3 : SUCCESS in 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a13aa1631c483597679d2566da09ee : SUCCESS in 13m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c475b708a0f146e5809a4e581ec2bb17 : SUCCESS in 29m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/963dde99de6643b9992324d28b5ef233 : SUCCESS in 27m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487ab10b8ae74f67a5974c97440b5fdc : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/045004a8bd4e424d9b4517a9d5ac179d : SUCCESS in 52m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c8e985e75e0448ab8b53f48df7e46c1 : FAILURE in 1h 47m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3fdabc6c6edb46c8a21bb1c794b8d33f : SUCCESS in 1h 07m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bc05d2c9dd74848823068cc1f82303a : SUCCESS in 1h 01m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3a35204bed704af49f2d6174010d31c1 : SUCCESS in 1h 20m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2a94eadc790c4b079bc06509fcea03af : SUCCESS in 1h 16m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ee1e58cd4f2f4767aa27413a79517694 : SUCCESS in 1h 19m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e8a2f13ba284ce1b13f57db876f14c8 : SUCCESS in 1h 19m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc4861704edc4ed9bb3dc97b9f3e6ddb : SUCCESS in 1h 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cf7ff62c6ee526561ea4bc9e0033002a5a216e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-24 14:34:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96ee1f080c2ede0283bd1ab98882359f6e748526","revisions":{"96ee1f080c2ede0283bd1ab98882359f6e748526":{"kind":"REWORK","_number":1,"created":"2024-01-16 11:00:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/905680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/80/905680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/905680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/905680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/80/905680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/80/905680/1"}}},"commit":{"parents":[{"commit":"fe802625081a9a2f02c874c08e0e269687b13122","subject":"Merge \"Add sleep before checking if ovs port is in the namespace\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe802625081a9a2f02c874c08e0e269687b13122"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-06 20:03:50.000000000","tz":-300},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-01-15 16:54:54.000000000","tz":0},"subject":"Correctly validate subnet arguments when using a subnetpool","message":"Correctly validate subnet arguments when using a subnetpool\n\nWhen creating a subnet using a subnetpool, we were\nfailing to validate all the passed API arguments in\nthe dictionary, leading to a case where you could\nspecify an invalid DNS nameserver. For example,\nusing an IPv4 nameserver on an IPv6 subnet. This\ncould cause daemons the l3-agent starts, like radvd,\nto fail to start correctly, leading to a loss of\nconnectivity.\n\nSpecifying a subnet by cidr without a subnetpool\ndid already correctly fail with an IP version\nmismatch error, this is just an edge case that\nwas never tested.\n\nSince _validate_subnet() was called in so many places\nit was moved to a common location and is only not\ncalled for IPv6 prefix-delegation subnets.\n\nCloses-bug: #2036877\nChange-Id: I6302e9a373cf93e706cec10f87c3beaf632a0391\n(cherry picked from commit 2f91d330dae0127be1adb98f3d6d1fd49745f25d)\n(cherry picked from commit 74a3262a8a96a870213ef4b52667ed2179246f9e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96ee1f080c2ede0283bd1ab98882359f6e748526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96ee1f080c2ede0283bd1ab98882359f6e748526"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
